)]}'
{"id":"openstack%2Frequirements~863199","triplet_id":"openstack%2Frequirements~stable%2Fzed~I1ebf6bf72883556fcc3cb22e591d7cbd3926c236","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"I1ebf6bf72883556fcc3cb22e591d7cbd3926c236","subject":"update constraint for python-manilaclient to new release 4.1.1","status":"MERGED","created":"2022-11-02 08:09:37.000000000","updated":"2022-11-08 01:34:45.000000000","submitted":"2022-11-08 01:34:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863199-new-release","meta_rev_id":"db2e13e33740d0e8c7ae36df33f854018ac0b30b","_number":863199,"virtual_id_number":863199,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 01:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2022-11-08 00:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-11-02 12:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2022-11-08 00:08:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 08:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-02 09:51:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 12:46:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-07 20:38:50.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2022-11-08 00:08:39.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"8ef1d1e8ee6970d24ecf578b3592f87ffe5c0bfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-02 08:09:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2eb955ba570156e996a05596cd64d214bf91fddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 08:20:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58541f11c8974299bc58fb7f634f85b4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/893af896329845a18df7896c489bb5b3 : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"222b98a8fc62a05b8245d9997f32df94e1564e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 09:51:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/834bf0b17aeb4ad1bec20c6b57b07f6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91eb8298617b490d999b1f863a821d8f : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3d43297efd84db48f67f6751a825459 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6129d855a5b43e88e128f43aef1468b : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4303817c05454f2f9bd39dc3b8d3b6f5 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b42204b26c84407694eae408b15bf325 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/805b89f3875e4c19981c6b18843fc9c3 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b433ac0b1e2543ac994e4d706ed44aa3 : SUCCESS in 6m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2ea6ed96ba984815aca7771555e08fb8 : SUCCESS in 10m 22s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/46df8901de8a47f2949a822df39b9481 : SUCCESS in 5m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bebde3768a0d4fb9b1bfd8dd231f45ed : SUCCESS in 4m 11s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/87bd9886d3714c6f836c11ab89edf262 : SUCCESS in 5m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/edd92daf9d3c44cfb6d04a2f111fe264 : SUCCESS in 8m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/eb76101a8fbe46cd8d82cbf4f85bbf79 : SUCCESS in 4m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e45289489ce84d46ba90d03aa0f0d761 : SUCCESS in 4m 54s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2dcc41d238be443baa11ab7b2c3f28e6 : SUCCESS in 7m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/141b60542eed48a7b04203beddfec8af : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/891a6ec3cf6f446283c0d60b6a967cc0 : FAILURE in 6m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7493d1e46295430eb0dc911129f9ce43 : SUCCESS in 5m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f43c5c280a96413786f0a5bd60073ff9 : SUCCESS in 10m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/609f578351a2412f9eeb0a511991eaec : SUCCESS in 3m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8309f1cf903441fd9c6ab6201a1d2213 : SUCCESS in 8m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0341f6c0518e4883b610f060a7ebf7bc : SUCCESS in 4m 10s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/9fc690c29d764f1fad1bd66e433023f1 : SUCCESS in 8m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/12ad4d6110c04c8c9a1c64ae42482eae : SUCCESS in 21m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2afa3e2665ff42c38f903a12316db495 : SUCCESS in 10m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8c5bbc2283ce4b98a17743a7bcfacaa9 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f2ba0ef980e455288db7d803867415b : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07d7ddef73e94b11b80a88e35cce991a : SUCCESS in 17m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/89176246ff7b47e888d3ec5e0f14cfd3 : SUCCESS in 4m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d0f01925d1fa4d8b80d3f41ea71c796e : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65b1a779812a4eb391afd270e4d4d903 : SUCCESS in 6m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5e3106e46b2149e39d0889d35d0479ec : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7370c5320d2b4de8970582b7714d7f5f : SUCCESS in 35m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2e19fcfbc9a435b97c6602fc655bf53 : SUCCESS in 1h 40m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1829fc2c11a00b41c64202a63c7c0c555a6fe4c4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-02 12:46:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"827aac08c29cc303feb79ec3766b77dc3cbc9397","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 12:59:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344327a320e14514856666dae5f78ae8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d34259c491e47d2840d704e2b4f00d7 : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"72b5c59e2ae440f6696f1c4b5ce1a3bccbbdf3d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 14:05:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb3483b1e04d45e1bf0e2192b0fedaa5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769528c452e24312ad886384b1b0ff25 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff215af2c11c4e808db7e2f06dce9131 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f51cd6726ee746d3a5d3d91a19deb9d5 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ee095cca418456f9c93cbebb5730217 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d77e595a15e4c1a8f5b3f028f9623f6 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93a66bf50541405fbc6e47d4ce077893 : SUCCESS in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd564175ebbd4b84ab85933e64687efd : SUCCESS in 9m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69773537822f4eacb381c86d8ce89041 : SUCCESS in 12m 22s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a0ac08e7466842c8b3807f1048a4261b : SUCCESS in 6m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/74cf5e9c61d5437a830d908f01e45259 : SUCCESS in 5m 08s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/b199d78ebf26411b8d70474db327c076 : SUCCESS in 5m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/af6b00a13f6b4a96aee13b4cc00a3a1b : SUCCESS in 10m 03s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7f1128836a3a46f388bc91becc199282 : SUCCESS in 4m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/64341e4c014e469b9cc9bbaa3e198acf : SUCCESS in 5m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/15e64a0dbb0a4033a41b3455bc0d36d0 : SUCCESS in 6m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b9d4b410d9bd4adaa091af820933c552 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ba20448068140edb6b4ddb6aab49840 : FAILURE in 6m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7ce24485090c4bbe9de6654f7d322401 : SUCCESS in 8m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2db43d1da8b342bbb191f22c40a2ed74 : SUCCESS in 10m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ce19bb6139fd44d3876ca42eba648a38 : SUCCESS in 3m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cd76ae54713e4596b2ecfb845a2a6876 : SUCCESS in 9m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5bdf53385e324c788e154d9323d5229f : SUCCESS in 3m 50s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/29c5111af99a447cb68dcfca739e8b4d : SUCCESS in 11m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f288493d03914a6ca3f0a87edf65cab4 : SUCCESS in 22m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/94ae9dd0d9da4673a8429c77b3d3b419 : SUCCESS in 11m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fa2933b7c49544569f147b50625c3ccd : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09473e18e91e4473bd54d52537603fb0 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c11340e128e405396d2a94e9f05f3d1 : SUCCESS in 22m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2b566ff02519495dabd7020fcf430013 : SUCCESS in 5m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a735961d3eb14d36adf3b49262dff1eb : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/60d43c5c16864d16a3a45fdcd2907264 : SUCCESS in 6m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/97655317880e460bba02ecf975b7cfbc : SUCCESS in 9m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cee9148ec86467a8a63de1bd4616efb : SUCCESS in 47m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fb27e048be847b3abdab91274304839 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1f397cfadb190243fd5299ae586a01ed8c811265","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-03 10:45:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"743b3338e5e0fd702876465168c0ddc08bcaee98","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 20:38:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3d81902dc0865ea20cbdd034288228b55e5e73b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:54:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e3df00d1046467dbb2a4fe88b06132a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7af4a0fe10214fe0996dc1f6d9e695ad : FAILURE in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1b6694a0bd2fc0bd2ec5d68f1dbebf2edf5618df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:10:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab1326a1a6f4f32941a361787bf1bd7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee754dce04a4414951f670d3a199ebe : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70862a9279e543ebac10853728d9f81f : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9010b445d7c24ee0b7c7cf93fd2709ea : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b9b1ad134ad4870b2eca1adc37c912c : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/959d36d9cd174eff915643e51c1b4d97 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4c8398413e54055b2913fd8915cccc5 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d15cf044a984b3da23d9431516ab10b : SUCCESS in 6m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a3d5cd30e27d433f9102e2656dca2647 : SUCCESS in 12m 13s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/4704b0d9e87846bca999b9c385a0065a : SUCCESS in 8m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1cde1fd636dd41df9441e85898fbea09 : SUCCESS in 5m 43s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c2c0da95d4184e1aad390cf1b2fa89fe : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/100e74a97edb46eaa69d30a80b3306fa : SUCCESS in 9m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/88d66b80135a46be9457ea67d2cb45b9 : SUCCESS in 4m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1f81dee7209e4d7bbcb04cdc2335829e : SUCCESS in 4m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7bb07a9e75f2488eb3d9bc0d726704c5 : SUCCESS in 7m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/36caffe122934123a6e9f51176668c70 : SUCCESS in 8m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cbfd7cbdcc941dd91854c0aea3431e8 : SUCCESS in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/59c77110c67e4653b62da9eb486b09ce : SUCCESS in 7m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b62f741f652641ea8409583ced5761fa : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3bb08905c99742b4a0657b4a28e74f25 : SUCCESS in 3m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4b3472cc52cc4b6ab22f9800f04a2d8d : SUCCESS in 10m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/432ca2a4b5b746f587b8a9976fdaf362 : SUCCESS in 5m 07s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/5bf1f28b068c4451bec9db7f2918f967 : SUCCESS in 9m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d7aff9e534534c079032d848e72dd6f4 : SUCCESS in 22m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2c666583586444a88e4fc173604e77f3 : SUCCESS in 11m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1c77f0aa5697489fa2e23d770065b465 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7ead1b46d02416b96c7d2b77f2f76b7 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7996dfbf69714f758e5e70ec410b086d : SUCCESS in 20m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/85d58e28920b4573aa842b84a880cce4 : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9a585dceed5f484d9bdeb8aff07c0e0a : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c0505990a73474f8a4ab7ce1dd4d1a1 : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0ef3d51a15624846ba28a13828bfeaa0 : SUCCESS in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c74a3b8abbf4ee28a6e7610ff9810d9 : SUCCESS in 45m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14f82e95ceb44a72a311754a631cd58f : SUCCESS in 1h 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"89c76d63f4ddd37408739a12ec84ec562cf36e6e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-08 00:08:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3843253c86ddd54489572dc03f4e1564487d0253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 00:09:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5d3455ced031355d97d366f370b493f082a03ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 01:34:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e02ac1519864b99b2e32d71ebda6ad9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd9274c42e541bf8083028bf777f2b8 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f77264d115564c69ad14b000c14b9dd8 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2c2ad030a844e1fa0234565485f693d : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff9ae36633db42a49f423a0df0dddd0d : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dffd4bdeb27f448ca84f297da18fd7c4 : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/648e88971d1e4074be3252e9e0f002a0 : SUCCESS in 8m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/411562f925214989a6ea78efffbb9dfa : SUCCESS in 10m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e232da27c052450caa9db2141fb5bf8d : SUCCESS in 9m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c123fc2a47de4b5a9c4f620885184bce : SUCCESS in 4m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/10b97a9af6514eeda3a560249e546004 : SUCCESS in 4m 58s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c9c438bc283847f99f947ccc6b048055 : SUCCESS in 7m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3d6fbec6e7ee47979e1f3afc362fb998 : SUCCESS in 7m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d0a6a91db914cf7ae900cbf5d13b849 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/30b45ebe96aa4c8d994d0c379e3016c8 : SUCCESS in 7m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ece0695261844bc3a2a90cb969a14610 : SUCCESS in 10m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2ddafcb32c924aaf89d9f85190dd0c9b : SUCCESS in 3m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/810c881ea8394545821f582450de7c64 : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3db73456d4924d9db6fa206458efd835 : SUCCESS in 4m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8d0d5be6a9ab4e988bff9025a11a9510 : SUCCESS in 26m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a6b5f499cf3b49dcbc69f230c9519fb9 : SUCCESS in 11m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1974c80eb1214d2fa681c82bb32e33fe : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c339ea80f2e441fd9df5c63fed9ed282 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82b6513035bc429cae8d50098ea95638 : SUCCESS in 18m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/390dd43fdf70436b955106373bba78df : SUCCESS in 6m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/310b4830bb164e768e172ddde8729d4d : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d157cecc75314422872526cee2dc8456 : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d6a42caa28a54c03afe3f38cc5b84667 : SUCCESS in 8m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3602cac3a6d942abaa4902fc6893febe : SUCCESS in 47m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/893e44e6688f456c88e305cf713ed9e3 : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"db2e13e33740d0e8c7ae36df33f854018ac0b30b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 01:34:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6989b43d6eec953f6c7087b00862e3696f3d82bd","revisions":{"6989b43d6eec953f6c7087b00862e3696f3d82bd":{"kind":"REWORK","_number":1,"created":"2022-11-02 08:09:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/863199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/863199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/863199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/863199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/863199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/863199/1"}}},"commit":{"parents":[{"commit":"7ff7b9e5638814d8318bc73138850779adb9bb24","subject":"Merge \"update constraint for oslo.cache to new release 3.2.0\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ff7b9e5638814d8318bc73138850779adb9bb24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 08:09:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 08:09:36.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.1.1","message":"update constraint for python-manilaclient to new release 4.1.1\n\nmeta: version: 4.1.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Commit: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: Ie9f228ce04dd07dba6405041d8d37d7683deaeff\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1ebf6bf72883556fcc3cb22e591d7cbd3926c236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6989b43d6eec953f6c7087b00862e3696f3d82bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6989b43d6eec953f6c7087b00862e3696f3d82bd"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
