)]}'
{"id":"openstack%2Frequirements~863407","triplet_id":"openstack%2Frequirements~master~I88e20e222992464ac225353cfa27ced5a11c4177","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I88e20e222992464ac225353cfa27ced5a11c4177","subject":"update constraint for os-net-config to new release 16.0.0","status":"MERGED","created":"2022-11-02 16:29:57.000000000","updated":"2022-11-07 23:47:32.000000000","submitted":"2022-11-07 23:46:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863407-new-release","meta_rev_id":"adfb75a8a52dbd27b7075532e777aa201aeb185b","_number":863407,"virtual_id_number":863407,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 23:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2022-11-07 20:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-11-03 10:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2022-11-07 20:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 16:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-02 18:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-03 10:43:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-07 20:40:43.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"a12488bad7d7d1ce328e38959f08790c12dbdd95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-02 16:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bf9244500aa2184bba8b3988f15a467a63f93a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 16:55:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef6faef04b643519e0e4272dae73a83\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/06fb535eb05c4c0cb9cfb4341a5e6bc5 : FAILURE in 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"85ab65bca297160e50cc48d1087ee134f75d1867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 18:02:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d52b3144c79a459dbf7df033b92eee1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bca68ea7dee4d7c81f6e142c02deee3 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ebf529ff15146d494dc802c8729c8cd : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff46fa9a8f8241efaf55660298b7399b : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15ba3e6eb36543ffba63856f580f87ab : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5800357ca214e83bb3be5fe6300bbcb : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a418c3e375c4423bb0f38d34e5d024bc : SUCCESS in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/527f0078c62f4b64bdb8624bfd752fd0 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1f5566f9f19b4afc892cf44d0244e99a : SUCCESS in 11m 34s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5cf4964de0cf4044bb42b5f384f5b57e : SUCCESS in 8m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c16d9c4d95d449099fcbb3c000f3a822 : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5915ffc58e404dc88d2e89a17ca12bde : SUCCESS in 5m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/19ecfedab23d42ce8d7916670fe64f90 : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/33e87e67c5b24d50a390860e66dc1651 : SUCCESS in 4m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9be7d65c964e43cab6b62af482a8ac45 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/013f6d72ba8b4acaa6e7e0e425d2e6d2 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4cbaa8df1d8d4b1d8e5dce7cfaa68146 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0caf305bf3a4540aa005fe129274cf7 : FAILURE in 7m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9d0cc3f6cccb40e4af290eb5d41e3ae8 : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c83419db7cc14d70aa47ac61c41154cd : SUCCESS in 12m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c9d884375c9248be90e04c95c9be6ab2 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d30a6503f6b3471cbb814371dc085e75 : SUCCESS in 11m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/81076169fb884d2396011b6b37fecc99 : SUCCESS in 4m 59s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c6cb938d805f4a159242fe2e116c9647 : SUCCESS in 9m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fef776163f9e4006b2c547071989efd9 : SUCCESS in 23m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b3cb4dc73f31480db84f9bc263250304 : SUCCESS in 13m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6ef4d9f155224af9b41fd3757a624b19 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5104fada8bf446ab912ee89fc4a0f07 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9c8b69eb9644a4d90b6ed325fb32804 : SUCCESS in 19m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d2d55d0b9bf647a29dcda85c605bc1ed : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fd4a40b05c8445138b8edfaa9429f92d : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d21af3a8216f41e19e679eb929315f4a : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5e31bf42be374fd7a235ec86d833cd62 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78eca3da6c084d2989d849b8b3edbd13 : SUCCESS in 46m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5031336106664395a641945a9cc95a86 : SUCCESS in 1h 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"05d6483c9fe74e31b15ae76410ae9f498a3a0ca3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-03 10:43:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5572d7368976678ef634f2a0d849d5d884cb83a5","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 20:40:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"632bfc612df02183b675f5a1b3046fabd2b65a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:05:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a0a6bbbe30d4c0698ad9729cb2a70f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5a3cf472f74b0c9a18dbb1267e68f4 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba1b9e549bba4bfe935669dc9442ba9b : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/943405fbec554c41a82a041d7c26b663 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84e63c8b0fc94bda96afd592f10f7cf9 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/123ada67de44473281d3c381b4402b8f : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5c019769add4d82bbc1e6cf832d14ed : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/285a7b0522a54715b68cfc8031c6a963 : SUCCESS in 10m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/291c67bcbc2c485684f65bb979bfc9ba : SUCCESS in 11m 03s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/92a02f4555484d87abb4d7d80761b745 : SUCCESS in 8m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/65736d375caf41aa8afefe785e639e1d : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e927cb5ca31948f5aae31e52f5948bf6 : SUCCESS in 5m 57s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1e77a39f65834ddaae189ed8bc5ccd7e : SUCCESS in 10m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5fd9fbfd6c1c49c3abf8417d72baed26 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1e16daba976847e0bfa114b098190ca4 : SUCCESS in 5m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c8582d4ffab14f7d997f2513b19b24ad : SUCCESS in 7m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/29b9302f6ea34c83befa6e0383ea5a18 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7479fd8b5df4ebe85235d77b004c2bb : SUCCESS in 7m 13s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ab8b9cde2d9341d5a781636fafe1e17b : SUCCESS in 8m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e63473e7b1a242ccbc8fb98bed011e00 : SUCCESS in 11m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/73732196a93c40cfad8d4a3b9e581f21 : SUCCESS in 4m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4e79e53ab54543bfaafc0b13cb0d7539 : SUCCESS in 9m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/148d3240425a4fd6abdb2b97c481b0e3 : SUCCESS in 5m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9b5b6246f88d4709a9a82bb6fe427f59 : SUCCESS in 8m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fd531a53d1cc46599d8b7c849b81e785 : SUCCESS in 23m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e8a107b7a9944e6faccc253e6561b137 : SUCCESS in 12m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/456dd5f307d04dacab20a89187e0eda8 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46a2e92d26384033a924b91f16bc64aa : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f8d374cef6bb4315b45b81c8ee095ff5 : SUCCESS in 23m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/31da2d992f644554ad2141713ebe1f72 : SUCCESS in 6m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/719a211b2dc942d6aad6122ccd10528d : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf45a12122a8485386167518d0894c6e : SUCCESS in 6m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/331c59ce345042deb8699c5103fc5e22 : SUCCESS in 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acfc76d3fd784fea8dbcbdb1b7c522c2 : SUCCESS in 54m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbd05889f8614ad8ad0a622dc05ecd84 : SUCCESS in 1h 23m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cbbd5ec2c23cadc6085dcefb6a5def56aea69720","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:06:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb19c10d5ba4310a72e40eb79cd6c5a950e934d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:46:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/424c4256b31a43e781c67e15ac8af5b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edddaf5145f4414ba9c84d41b51f72c2 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc08f77615849749386b47e356f87d3 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0ddb978a7994ff9abb5749decc74efd : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0ab7db850d144aa958a8ffac85bdf0e : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/749411d9d9bf414fb91ad4ce50a8d455 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be3d16a7d8224e149721d0635794fbd5 : SUCCESS in 4m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3f592f436c3b47d2aa673e994cfb993f : SUCCESS in 8m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/277fd522183a4e68bc050ca97e3bfeba : SUCCESS in 11m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8b397699e37241ef98d441fb28baaa6e : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e4c4324514504c2ba828c372664e1abe : SUCCESS in 5m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0897758734a1469f9c259c02e975140a : SUCCESS in 6m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/15a26e7f6c644502b02b1a1bf52685ca : SUCCESS in 7m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/20320f50de7146d19b30bdf775452b9d : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/674a633fd0d1469091808fdc4450307a : SUCCESS in 8m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8c8f44a95a6d4976a88edbaf16466964 : SUCCESS in 9m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f86c3d75d0fc43efac5cc9e8c9166da6 : SUCCESS in 11m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2c23ebe437af480ab5da2f23f1bd63fe : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fdafaa2282a74254af89207eadf3e526 : SUCCESS in 10m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0b883c0250164120b3cdb074c66b0e3d : SUCCESS in 4m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cad62df4e8da4b52bf1eac1a98b0f151 : SUCCESS in 24m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/84582dbff1f74d6f898ac7ac255229ab : SUCCESS in 12m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1ca346dd31e84e2ea14353d449ed3cf7 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c054d97378674594acbfab768189ee95 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06e2fab64b64459e947aa4ec16f76d81 : SUCCESS in 22m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/72bda081ec9b4e20917266d8c5f5d10b : SUCCESS in 5m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d84bd535447c4d1193faefb7606e1d9b : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b4c745cca7346c18c9f3460cc6434dd : SUCCESS in 6m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/961e6ae0404747678791ae9c82677ec7 : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d4f00ba242643a9a0134ed9d29fd6fe : SUCCESS in 48m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b181fea1145b42d098c3e16a1ba8fc36 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f2c3ddfdcdb8a036616215af1e2de80c4acddc05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:46:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"adfb75a8a52dbd27b7075532e777aa201aeb185b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:47:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ea2093d986b41168685e3d68837b175\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2de6673f80422caac73b5317fc882a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a08c466163e9ffc1dc046d51929e0dab0b61c749","revisions":{"a08c466163e9ffc1dc046d51929e0dab0b61c749":{"kind":"REWORK","_number":1,"created":"2022-11-02 16:29:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/863407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/863407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/863407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/863407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/863407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/863407/1"}}},"commit":{"parents":[{"commit":"00e8f93f7558babfb2460f6f58c2bd44b71109fa","subject":"Deprecate openstack-ansible rsyslog roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00e8f93f7558babfb2460f6f58c2bd44b71109fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 16:29:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 16:29:55.000000000","tz":0},"subject":"update constraint for os-net-config to new release 16.0.0","message":"update constraint for os-net-config to new release 16.0.0\n\nmeta: version: 16.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Commit: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Change-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\nmeta: release:Code-Review+1: Takashi Kajinami \u003ctkajinam@redhat.com\u003e\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+1: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nmeta: release:Code-Review+1: mbu \u003cmat.bultel@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Ronelle Landy \u003crlandy@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I88e20e222992464ac225353cfa27ced5a11c4177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a08c466163e9ffc1dc046d51929e0dab0b61c749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a08c466163e9ffc1dc046d51929e0dab0b61c749"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
