)]}'
{"id":"openstack%2Frequirements~863415","triplet_id":"openstack%2Frequirements~master~I2bc4b2e1e7f05e415564207ba4ca51626a091a39","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2bc4b2e1e7f05e415564207ba4ca51626a091a39","subject":"update constraint for tripleo-common to new release 17.0.0","status":"MERGED","created":"2022-11-02 16:38:33.000000000","updated":"2022-11-08 00:12:52.000000000","submitted":"2022-11-08 00:11:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863415-new-release","meta_rev_id":"df302b7f66cac5da4c4a834bdc8814227c5f6653","_number":863415,"virtual_id_number":863415,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 00:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2022-11-07 20:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-11-03 10:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2022-11-07 20:40:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 16:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-02 18:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-03 10:44:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-07 20:40:13.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"756f57fe767e0964a0671f1a8a5ddf350de79ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-02 16:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d11ff70924617016561a4bcab2cf936ba00733ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 16:58:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/953e6e8ac9ef4af99731cd3d46144df9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/26b9cfb0b53642f69f0eab64ff4f986f : FAILURE in 13m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d7df9acf7d6b3a166a07ee2409d825b3cf488964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 18:02:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c7a9f0a64364bd3ac898d0245693e22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f90400b4aa4088a773affcbd2d82d7 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d19968a7c04089861af3151eeb61f3 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fd7ae0da9754b7683f4303e980e4ba4 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d8bf1ce33c14d8b9ea331c98c5c72d3 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c40c915a7ad4ca6abbd452ba4d53c8e : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db36472b755144ebaeacd92d665029e7 : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0a66b9cf1f66422e81fe156bfe021eb8 : FAILURE in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/626261227eef4da7bae4044fedf64c05 : FAILURE in 4m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/387e2a4a01fa4bf99d69c2ad572ff2fd : SUCCESS in 9m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5adbcf13fd344948b217402e8ef46938 : SUCCESS in 6m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/84fd652f30fa4041b6042728acc8dce3 : SUCCESS in 5m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8a25a7b699be406caea8717cf5258a7b : SUCCESS in 9m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/092cc0ce3a8542979300fea018219f10 : SUCCESS in 5m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/94ef8e61e3414f258468bce94e2b2d95 : SUCCESS in 5m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b8e1b0d74fe4485cb9f4eaa7c6daf8e7 : SUCCESS in 7m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/709a9cf41e7a4ec283098e4eb9faaf60 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81cf178bc17e4542bc22af4bddfe45c8 : FAILURE in 5m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/22bf5cc98e454415b7bfd9d38bf22749 : SUCCESS in 7m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/be4a68929fcd4e0db97cdfefe7fd30e5 : SUCCESS in 10m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/267e0ff8b1b349f9aaa455a102d0892f : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/18525598e79440338240c6db27033808 : SUCCESS in 10m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6c9f21228e30494eb73ed4f98bdca2b2 : SUCCESS in 4m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/00119f69d9324b7bb73c93d9be94f792 : SUCCESS in 8m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/55b2ec30ad6f4d30af76a1f8478c5ca5 : SUCCESS in 22m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/08bab06ca6fe4afab4229ead112210fc : SUCCESS in 12m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/835374fc134b4161941c2392049d15f9 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77a246e870d74ba19782d47c425e46bb : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a8b0ed3a4724bb0ac6a6343671e20dd : SUCCESS in 19m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4c0d37022979403a944e5dc4286ced5f : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6ebdbd41fbe34deb98fd7d152b4dcecb : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e691ab31891d41ba9033ab6ee41f9dfb : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9dfb8164c24640fea5974ca5dde5f48e : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77f2ef0b2a4c45789f2785e6e1e358e1 : SUCCESS in 46m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5685f9e34a47470aa606ab8480d8bf7f : SUCCESS in 1h 18m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f4e5068ceada0796fd8d55e36785b4993786f6fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-03 10:44:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4d9903d7b803d78af9ade2969d109a06a62f322","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 20:40:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa8c3f1145ecf1236127d12bb69693d81656f16e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:36:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7722f87b8f9c41a486eb22a13e182daf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3ca556a4db46a49bec04c262db6c87 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7cb7efca5b41d09106f1d93ddf0749 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdf40fa385004393aaea759a0cec8e9f : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89258342b5d642118cc229b9b3bc7ac5 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b13bb7aa4c574236a62e88607c45c4c6 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f17e119f2443433cb8e3ca3d73c83030 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e118740e5ae419b83e552fd3af699e5 : SUCCESS in 9m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a68a4d3bcee54ea1817b58f57845e7e2 : SUCCESS in 12m 49s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/417443dbfbb94712b60c8ddb3d480b2e : SUCCESS in 11m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/577c4e012a344807adb5eb560c547a71 : SUCCESS in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9234eaf967eb41fd8c1082f9a94da181 : SUCCESS in 4m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d3567a9f60b84529918c5dcf8412657e : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b9c213aa88204d3d8bccead6d809823c : SUCCESS in 5m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9ca119dd59c447fb85443367463d9557 : SUCCESS in 4m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/775be44b10274f4b97033f8f9f1aa1c4 : SUCCESS in 8m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7c873e1ae3ec4e03a7aea461e9fffbdd : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b81463ed74d948c28a8565577a80e3c7 : SUCCESS in 7m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dafafbe761864f2cb8447d21316cfe3d : SUCCESS in 8m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/da4dcfa3de2e440d90c92daec3e70a0f : SUCCESS in 10m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/976be641e44a4be298c2a0657eec6dfa : SUCCESS in 4m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a123ee7595644ecebd11606a256d9540 : SUCCESS in 8m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/17d7bad442d342f79179d7748cde6ef8 : SUCCESS in 5m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/240c64e2a16741fd8370142132da5481 : SUCCESS in 8m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bd4efd3c486346f0ad72504e21cae00b : SUCCESS in 22m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0ffe1f3d100c4b25a264536d94e9a15b : SUCCESS in 12m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f248f6124ce0487f88875d3324dab03a : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70ce394fcfcd49b4a2e923b92a71efb0 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/983499b6692e459492432c3ffa893456 : SUCCESS in 19m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c21de6504dee4331841463db00a6c3ef : SUCCESS in 4m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/87d89370b5b24caabf34e9192106e9cc : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/652aefdeae094547bb9001aba472d8be : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a31525c8e33e4d608d034511802186c6 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/314b9062abc041deb004cf298556f7a3 : SUCCESS in 54m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1a108a6694f4312a33602e1a94aef97 : SUCCESS in 1h 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cf7ce948265eb34c7ee6bc6208d4231c97e3cb56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:37:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"595fe0552a48093b83299166dd6836c67e3ee1b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 00:11:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78423b5a702e4e4cb5cd210f0dd47de6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ebbbb11953947ef8e497d083cfa856c : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a849888730d3467989879c3cb350a1af : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9630dd2bd974644be6566a679a3a7ca : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/807faba3d5b3429eb69be203d2325718 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2291d020761484ea4248df97c0c31fc : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b046a51b7ef4cd4b33323947b39af49 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86104d31c1874032a2eeb3b2ab26446b : SUCCESS in 9m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3783218a29bf4f06ac4dfea4e7cb8503 : SUCCESS in 12m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/025f11398f6e4323afff3c9613b6164f : SUCCESS in 10m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e532ff52a25140b2a0215a64e6f4c427 : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/751d77d81b72471fb513e849d52f1bb8 : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/831cf281cefe42928b387f7ab1b3d0e8 : SUCCESS in 8m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0b60e3f032bb48b6a62c308d95388f9e : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/89b98efbabaf4dba865a51dcf34f6497 : SUCCESS in 7m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/828b5652301e482e914ee165c113978a : SUCCESS in 10m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e0d3c4f3cc354f32b0aa0286202f58b6 : SUCCESS in 10m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e20822a71bf64cdfb99a6c0edce49116 : SUCCESS in 4m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/aba6f747c5c54a0d8a5003c6ecf3dfab : SUCCESS in 8m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/aaf28a3cd8194cf397d63ec318a547ce : SUCCESS in 5m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/433f7e83258d43e689019818848423f2 : SUCCESS in 32m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d0a5bdef39ca471189ae4ba676244fc1 : SUCCESS in 13m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6b4cde66ee124c89b9a9b2151400deba : SUCCESS in 7m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3599d9eb03894915a5b630faaef039fd : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95d2a33ef10b48d397538f5a2d3442b2 : SUCCESS in 22m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6715549504d24609a7aa94146e6cfb6e : SUCCESS in 4m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8b2684c5b80148d59cd32a4a003f58cc : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0b8d5a42e9c4fa19d4d190949cc36d3 : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/124868ddb9c74fba8fd3d0d5b5bcd55f : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9dda29201724903895943db331af2a8 : SUCCESS in 54m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57ebe444071240c8b289f7b1e690a201 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e129bcdc480c129e5524200dd7dfecf74567210d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 00:11:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"df302b7f66cac5da4c4a834bdc8814227c5f6653","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 00:12:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce841b270ea54cc98ba583827f80aecc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938e5d6349bc48c19ecdeb06e5ca9956 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c995e3cbc229b4b6caf4f116c0cad97bd37b3e4","revisions":{"2c995e3cbc229b4b6caf4f116c0cad97bd37b3e4":{"kind":"REWORK","_number":1,"created":"2022-11-02 16:38:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/863415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/863415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/863415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/863415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/863415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/863415/1"}}},"commit":{"parents":[{"commit":"00e8f93f7558babfb2460f6f58c2bd44b71109fa","subject":"Deprecate openstack-ansible rsyslog roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00e8f93f7558babfb2460f6f58c2bd44b71109fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 16:38:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 16:38:32.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 17.0.0","message":"update constraint for tripleo-common to new release 17.0.0\n\nmeta: version: 17.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Commit: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Change-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\nmeta: release:Code-Review+1: Takashi Kajinami \u003ctkajinam@redhat.com\u003e\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+1: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nmeta: release:Code-Review+1: mbu \u003cmat.bultel@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Ronelle Landy \u003crlandy@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2bc4b2e1e7f05e415564207ba4ca51626a091a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c995e3cbc229b4b6caf4f116c0cad97bd37b3e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c995e3cbc229b4b6caf4f116c0cad97bd37b3e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
