)]}'
{"id":"openstack%2Frequirements~863417","triplet_id":"openstack%2Frequirements~master~Icc3ca2c341cd4e828f2940a57ef84c3a97f82557","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icc3ca2c341cd4e828f2940a57ef84c3a97f82557","subject":"update constraint for validations-libs to new release 1.8.0","status":"MERGED","created":"2022-11-02 16:47:18.000000000","updated":"2022-11-07 23:48:12.000000000","submitted":"2022-11-07 23:46:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863417-new-release","meta_rev_id":"a4f618cd0d81ecdccbf5b84daaec84252f1ec9f7","_number":863417,"virtual_id_number":863417,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 23:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2022-11-07 20:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-11-03 10:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2022-11-07 20:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 17:08:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-02 18:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-03 10:40:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-07 20:39:51.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"13787774f887a0c6da02badc68a4092344e95aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-02 16:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbb381912a7d1f1016f5f5b68ef8f5ae3a887f8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:08:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/057f6a2ccf554447addb4c3356ccaef3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8f2b4ff417b54bf1b77f1123e94e2758 : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bf66f1db28b1fa5aaf016e9ed3882e48da78aed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 18:47:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1866fb42ac7f4510addc53a291eb2847\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5539f48b18143ab924d039e04c64b2c : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37860c75d3b411ebd9750a060246e9d : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e43740626ada427e9d2ea31152af27de : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67bb58c4341a47499197e1aaf920ae79 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0eea672f7304798bdd805f80d1fd734 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bbcc8c244d5438287a9f262252e4037 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb3c910509f7429eaec45cf3c02b2324 : SUCCESS in 11m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/858f414dbe30415bb1e6050f781a10dc : SUCCESS in 15m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/be4ee3ed65564c5aaab23c6710d43660 : SUCCESS in 7m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e61b3aa7b64b4fc48dc2bfc831d5815b : SUCCESS in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5dd9d3aaec474266ab8828db8d37e5e7 : SUCCESS in 5m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c1144466af964a23b1ea678ea5f166ef : SUCCESS in 10m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6b990656e7694cf9995804efca715c89 : SUCCESS in 4m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/11e769664e574f5ebf2e7e9345b1c8eb : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/163e33bbc8bd497fa726450456d9ca47 : SUCCESS in 6m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d465f1b976e94b8fbfe6c94306613976 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d2fcc7e36f944a1a4eb774a8967598f : FAILURE in 5m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a95d3d112d9a4a559c7d1f289b3cea8b : SUCCESS in 7m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f52dd3aed7464eefb2504b3d559b68cf : SUCCESS in 12m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/58430a98dae744088614c25aa27b977a : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/065611b314b847b094b9eec2fe277eae : SUCCESS in 8m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5a8261a4a1854b15b35428daf3c45e78 : SUCCESS in 6m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/30e2fb951cff4831a70ffed6aa80b6b1 : SUCCESS in 9m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6b0002f05e1d4b7f94ff6a4db33585af : SUCCESS in 32m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a5c1970ee1764f7592c99e1b85fcd382 : SUCCESS in 11m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a9640ac943d24ddd8a73c9ad15a98d81 : SUCCESS in 7m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67638c22cc804cb291ab3eec0accaf5b : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efeed59c09f640189a07389f63bf6e4e : SUCCESS in 23m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0d48e09645704e1db149ce5cc9d37df7 : SUCCESS in 4m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c405d5b370294303a33e1ab6d94300c0 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/276962bf3c904bca82d321b6b290caeb : SUCCESS in 8m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cc6bae831f434a2a91e038f38891326e : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b401c391bb041139597658e5d5ed1f1 : SUCCESS in 33m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cddae6c640c744e485ee8197e29e3003 : SUCCESS in 1h 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7c2ab11cfc1e33ceb603a8e3ad691d56bdf3c106","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-03 10:40:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eda7b748fe320230f1cba19cbe778b2244c4e3a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-03 10:44:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c7807ec70c46d6f8a27e4f0b77a73d6c05ca7a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 10:54:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f80b1aaf254f43f686b101770645d706\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/069ff8e55a4e4d468ba2888cb7dd9d11 : FAILURE in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b0a657bc52d5f781d36ec5a75b5e7ddd951a5351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:59:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76347f15304540d78f5a115d43b3b43b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15fa76c5276a470c99f15dd9520f4a9c : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a8b2bb5ef14fe791ec9cb74f42582b : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f84b9411ef34506bec46227e395e210 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b7c5d4e82fd48a68df793852a262fa6 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbd46e054c7c4b9d90c2fe10fdc66c97 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88873ba7f51148a88b119d2608ed65af : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b371e7f200924c28b83a93f1435db403 : SUCCESS in 9m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/46c8518709fd4dd2a808c6fafbea8e19 : SUCCESS in 13m 13s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/09215cd0a126420eaf3e5b0c7aeadf22 : SUCCESS in 6m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2f471bd98e954ab7ad9ef65e13f2302d : SUCCESS in 5m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4a3f8f2db2574e4c96bcd0a63dcf2106 : SUCCESS in 6m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1ba1a4e995154d70a956fbe7eb923c4a : SUCCESS in 10m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/23e6cf208edf4da8b45954667041b360 : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ed5e9291400d446a8e3dae8fcc8f1ee9 : SUCCESS in 4m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6505931678e0418b87eae13a76dd9098 : SUCCESS in 7m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/08eb2a317d544c1cbb9f0964a949dff4 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d1af3f52bf114493a19d08cb8997685f : FAILURE in 6m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a27622d67fb84e8f9acc47873565725f : SUCCESS in 7m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2328f2dd361047b4a1ac9c2972fcd231 : SUCCESS in 10m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/52cd29afb0a442b995e3a5c7e62bde31 : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/624d0cc7da5e46538825413398f77bf7 : SUCCESS in 8m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0feae8e546c84dc587efedb3b75d2e04 : SUCCESS in 4m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c04a2125efd74c058d5eeb0dcd03c7e6 : SUCCESS in 9m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ae91ec592c314a1b893904d624775b0c : SUCCESS in 27m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b1a4def505ce481885475c287a375f83 : SUCCESS in 13m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3ee04295ed194b4290cee8d8c58e9006 : SUCCESS in 4m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ce72ba166d30402ea8208e88bd205241 : SUCCESS in 2m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff36a995f3664f9794cf930674ebedbf : SUCCESS in 20m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8fba16e4df4642d3949a2cf6cfb1e25e : SUCCESS in 4m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c1cdc71b30cc4b57a75cba70fa389689 : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/393e4be6ca08445997f7b55ef7146b19 : SUCCESS in 5m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/03c8bc9ca0c045508957c64b999fa6a6 : SUCCESS in 8m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8b34e8bb87544578fcc9c585d417d7d : SUCCESS in 37m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc2097d2a89643e58d83b595adaa2e75 : SUCCESS in 1h 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2695fc6591577ee0356dcec9933e2982f5f90664","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 20:39:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc9c56bee5ebc4a4345e70021fdc863cb3d059b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:13:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216df80ab08c4703be460f7edeaa589b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af945633b97741df9df6157a6be5852c : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b8df4a2aaf47b7a4a508d451f9ab12 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dc0b928ba884b0d93d494a211eb416a : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36d78ff3a80847ddab87be87570033c1 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14cb8540a89740ffb8edae659169e773 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b5a8462ef564c2f87d90d512259b350 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f70971ff8134a07917180222bf16ebb : SUCCESS in 9m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c7defddfc4e14bfda45121b082a46ecd : SUCCESS in 11m 46s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e1de000690064eb6bf294d8727e09a67 : SUCCESS in 7m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cb12b32849b34b24b6c1762b5aeee78d : SUCCESS in 5m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/000dae52c49d49929255bcfe94678f14 : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f66a956290a14ec99e0910c59608479a : SUCCESS in 10m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7c01564c65b043cc807552ba2296c2e6 : SUCCESS in 5m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a863240a0aa94464b8ee84ca7878ebc7 : SUCCESS in 4m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9582dc7164f6493b934eec3770d24ad9 : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2157e63f90334091b8cacc0f6c71b676 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9fffded9e10a48b78605c63c7d055442 : SUCCESS in 7m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/22368b3f9569481e9c6014edc74b88f9 : SUCCESS in 8m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/67af7981a0e143cca6fb3ad6eddc05be : SUCCESS in 14m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eef6d94d227841749f85329bfe40a7db : SUCCESS in 4m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f06fbf7e4695445db6f3b2aabb122b76 : SUCCESS in 9m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5d3c1531f4ab428d83927dff4f6bdfc5 : SUCCESS in 5m 11s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b9e4cae8dd3e4513936a0a09d6a23f27 : SUCCESS in 9m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e0db394febe74d81ae62484c003055af : SUCCESS in 23m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/25b641f6422b450a8048c122d3e6655d : SUCCESS in 12m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4e2b2403adef431498eaedf0eb1467d7 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a34055a723a04adf9bed81b046746cfb : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9e85f61eb624b8981df80fe23b20170 : SUCCESS in 20m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/482b46a384aa404ba16adb59299ded3f : SUCCESS in 6m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d69c768acf8948c3b48857ed17d5d93f : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c685c4209ab4c98aef650e8de31e0b7 : SUCCESS in 6m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ba720c9f2c2949cd83ee7e7838e8d451 : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c95197201e44020b8b9c25bb82b1c8b : SUCCESS in 47m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1195d75e4264574a4d83148efeb66da : SUCCESS in 1h 30m 36s","accounts_in_message":[],"_revision_number":1},{"id":"73f2d558539cfc23f6e743149a900b888a2c1db9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:14:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1ad389034648aeeb309be19a3cffbca3f3a91c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:46:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/275e511b58264b389df16c0bd15d0b97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a198ee0355341c792091a25f051f201 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d161cfd40c447cea27a9aa5f983ae7d : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63a1671d21984186898ea1c547d5e8f6 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa24ca92cbc1424584e66d2dbeb72a32 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c7dd9f65c134197b4eed0bb88a3b731 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e9dbc4d4a56464a925aba38f9f6702a : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a993e9f951364b819298d8a9f3d1ce49 : SUCCESS in 9m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5dea230415b24a9d9a73ebec2a8141af : SUCCESS in 10m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b1b2320c032c471eb30d7f392e2dd147 : SUCCESS in 10m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/eb46521b388d4f239c84297389f6afd3 : SUCCESS in 4m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4663db882ee946f79b9f016b7fcc1f38 : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/99df2e5a7c2f46acb00e46ed280810f1 : SUCCESS in 7m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/af396f4e5bdf4ffe9c79cf1c4e15b26a : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60e8faae87234ea08894e27351994d23 : SUCCESS in 7m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a8d9e40988d74e229cfc171facf0ca4b : SUCCESS in 7m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a6585331b244412ca8454e747d636efc : SUCCESS in 11m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/037b884944ba40b0a5ef2f30423620cd : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e5645ade6e0e46a7a4deba691900b434 : SUCCESS in 8m 38s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/310804af3fb446c2aa3e496c865c80a1 : SUCCESS in 4m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0f1622a17fdd489885d273d050b31c6d : SUCCESS in 20m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3da2bab0ac2e475aa8b18b6bcc9b9c37 : SUCCESS in 10m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/89d7eaf6ca44476f80d107b1b13587c8 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fb4c9f6b4c644536b66eb3a07f467c91 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d465a2dfa6e4031b498638286366d5c : SUCCESS in 21m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/711f7aa1f53341eaaece3d742f2a4341 : SUCCESS in 4m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e6987b4760c8498bb34c607687ced587 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a21ac5f8e33046b5baa626b2987cd725 : SUCCESS in 6m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e65892e008b4f019ed276966983109c : SUCCESS in 9m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d27c3eb903ae4da18c900c9b68e63b5e : SUCCESS in 46m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aafe433fecad4ecaaf98766ca4376e49 : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"382ebac39330274e7f7a633b5dba440c905ead46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:46:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4f618cd0d81ecdccbf5b84daaec84252f1ec9f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:48:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa628c4e046b497fbdf1ee867c88d6d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa59be66fdf46c5903e5f307b13ba48 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ce9b0145ab6750c0159452e0a85c109d4a136ba","revisions":{"5ce9b0145ab6750c0159452e0a85c109d4a136ba":{"kind":"REWORK","_number":1,"created":"2022-11-02 16:47:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/863417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/863417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/863417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/863417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/863417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/863417/1"}}},"commit":{"parents":[{"commit":"00e8f93f7558babfb2460f6f58c2bd44b71109fa","subject":"Deprecate openstack-ansible rsyslog roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00e8f93f7558babfb2460f6f58c2bd44b71109fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 16:47:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-02 16:47:17.000000000","tz":0},"subject":"update constraint for validations-libs to new release 1.8.0","message":"update constraint for validations-libs to new release 1.8.0\n\nmeta: version: 1.8.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Commit: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Change-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\nmeta: release:Code-Review+1: Takashi Kajinami \u003ctkajinam@redhat.com\u003e\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+1: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nmeta: release:Code-Review+1: mbu \u003cmat.bultel@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Ronelle Landy \u003crlandy@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Icc3ca2c341cd4e828f2940a57ef84c3a97f82557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ce9b0145ab6750c0159452e0a85c109d4a136ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ce9b0145ab6750c0159452e0a85c109d4a136ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
