)]}'
{"id":"openstack%2Frequirements~863511","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ief9acf6d570444002e1a8e0f6879cb925cd37790","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-07 20:26:38.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"Ief9acf6d570444002e1a8e0f6879cb925cd37790","subject":"update constraint for os-brick to new release 4.3.4","status":"MERGED","created":"2022-11-03 11:34:09.000000000","updated":"2022-11-07 23:29:15.000000000","submitted":"2022-11-07 23:29:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863511-new-release","meta_rev_id":"49ef99bc67b6d5b374948ac84955c7f0cd35b511","_number":863511,"virtual_id_number":863511,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 23:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-05 18:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-05 18:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-11-05 18:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-11-07 16:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 11:41:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-03 11:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-03 13:07:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-03 21:01:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-11-04 20:16:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-11-07 16:33:31.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"2b8877d311a32db1654e95b8fa96ee25fc22bb4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-03 11:34:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cebd713aae321a3fab43e106edd764c9caa475c3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-03 11:41:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c76ae43be6bef77cb5977ec89ef2baac044b2bf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:49:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/283f65a0ef6b43aa8e17359cc02835c9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e3bbaa22d60542f189425f0c8023a439 : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"28e91b691bd2f7bc3fd1800070615e609776a159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 13:07:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3eb053cf34a44450895e245d0f19735b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b1be5fa7eb44df95481a8030e6669f : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/569b37260af74a669425398604b0eeff : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e847577375346a9a6a8dce85329d988 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e593774152045b78087ed4da7a7bc7a : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18f4158658ec4bd0b2357dba9577db74 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb83f3f3c9c947f697569158894e50c4 : SUCCESS in 3m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ffda812f82b34c92bfb3c45d17e26de3 : SUCCESS in 7m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b34a0ded3d614aed8d5c5bb02bf9e56c : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/176e399abca247aa9299f807d57c0554 : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a5191c748b134002b80c4d593693cc6a : SUCCESS in 4m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/904a8bd1631b4dce8211db06c19c4049 : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13ed9473730046f79ff2050534fdde4a : FAILURE in 6m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0d865569cf014344a59655f5c159937a : SUCCESS in 5m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/263d990580a7432f93045961b0daa321 : SUCCESS in 13m 43s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/583366f2ae48484f8e904c6f184d3de9 : SUCCESS in 4m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ead1f8c20d504e8eaf899f7eb53ca1eb : SUCCESS in 21m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7ea40eb3c8a9432d9a28b8770cfdd96c : SUCCESS in 10m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7d1abc6d700e4321b703dcc16fe3ba6e : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8756d1190d6348f491b0062487411a46 : SUCCESS in 15m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/82c813518dde444cbd75f0ca881114cd : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37b308e18a504978a3f538211bdc4221 : SUCCESS in 5m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/580a18bc866d44f78de7ec745599af0b : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3290fd6fcea6459faa36ed9a99b6b729 : SUCCESS in 40m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/957f26a41457489aa6ca93006a2b5096 : SUCCESS in 1h 29m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4b2efe2f6bf6436483780806edd43caaec748131","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-11-03 21:01:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c91af69d17b12b6eef033d78cb978552f3540300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 22:26:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67445488f00849a8b7bb7ff58c585ec9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a534edcfedaa4e3c8b82b40b1d2a4a0e : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d891bceb13914103aa77be973a376592 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c084fed73974179adde988d4db199ae : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ca6f7c9788f467c98fa8176400b73ce : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00635e9961c74019a264c4eb3078424c : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe7a31125ad04336936e9da137a6e06f : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a15161b585104e2d8014ebd7676d3fa0 : SUCCESS in 11m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2e7d90b67c374b1b80b85ef83d9a18d8 : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/68acea06d0624ba2b70b409ca5033874 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e52cb3ac3c0348ddae5bae22e9f1e3e5 : SUCCESS in 7m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c2366d7548fb47d18e48f7ffe21f94e0 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6432b74f22374f578b744f0871dc4bb7 : FAILURE in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0e535fa4ccfc4a40b45eca08c3831825 : SUCCESS in 7m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3ef61415726c40c3b86b20cc0b848d3e : SUCCESS in 11m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c1280045a7e44234aceeac5c91b9c328 : SUCCESS in 4m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4ca4ec06f18f42cbb56fe9ed7cc7f71a : SUCCESS in 26m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be80688de00e43e0836a68e396412707 : SUCCESS in 10m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd94c1d727ba46cc94500e71a3efe400 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9469b65f9344b45943b29f5e158bc42 : SUCCESS in 16m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0a0df51cd75c44c4b8762d01194eecda : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ebcb80765534a119b53ea1deab2d1fe : SUCCESS in 5m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/527e4cb15f2d404690642a3fcfb3c9a3 : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f265038fda5b439db11f68ddc119a598 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5883734aedf24947ba03dc0158037c72 : SUCCESS in 1h 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5f37c806bf759a77990b3a37b0d173fd50ea2a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 14:46:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1af3a344cd381ee53689f680c064252c1ea76dba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 15:07:07.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce73dd9f339842b6bc19e650ddb3657c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f181df345b3541faab6cb6d29f7dc606 : FAILURE in 17m 55s","accounts_in_message":[],"_revision_number":2},{"id":"dfbb9968db5109d55747be7392bd9b52a95968a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 16:16:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2420ad87af824ea88603a7bf9c18ee9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a012d5f82284880bbc2f9060da350d7 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d5a79b06f2d44ff8558ac5959b9ac03 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ac4a28891e48ee80ff7e277500de86 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ec2f0ffb6fa4117b306b9f33b7eabd7 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a83eefee6c304671ba605bc0cacf8dbf : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b4d5da085a44934b214e2d479c5c202 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4934acf0f5144a32b6176e0024118a46 : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c7fa0a8f3d0488797b2737bdd643949 : SUCCESS in 7m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e1d1feb0e0a8495f848601042a9b6789 : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/45841d0770a2483ca52f1b0c76ce3079 : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/78b7748f232e4cd085f6b2340b970c5c : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d11e1014152b448b9bdf9aed7e9289d0 : SUCCESS in 8m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/43ecaf0f16584b0a9ed4d1873cbe7008 : SUCCESS in 7m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bc944aacf16846b1942303154056e153 : SUCCESS in 12m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fdb133f09a414db98367fe16ea2ba327 : SUCCESS in 4m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c031ef3e881a4ff8a7f5936ccc261d6b : SUCCESS in 25m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2abda327184444188f702388190095a0 : SUCCESS in 11m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5f036d6b5344a709e8fa332029b8c0c : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ae44623c3ef4388984950f322f4f179 : SUCCESS in 18m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/538c08a1ef0c4038a66b6b45cd5e3cc8 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7d7a0302a6804585a8555e95af507d3a : SUCCESS in 6m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/102bad17b68c486d85450852cf1bee81 : SUCCESS in 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/177684def5b740b493f336b90bd38ecd : SUCCESS in 50m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b76929118dc941a28bdecc8bf6ce8c04 : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d89ed3ae3497057d2b8972875ab71f9ebe608bfa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-04 20:16:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7804b28ed580ffa903009c993a4f3670a9b0e859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-05 18:03:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4a6c4e9ff2fe4558ecf9683839813f51542a9505","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 18:18:00.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a862feeade417bb29a49fe86730f81\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6cda1a5f508c453dae4ae5bbb9120c8f : FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":3},{"id":"aa562a24048fadfac2a56ad76b2269f74af9e720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 19:34:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/394e6e8209b74a0ba2210c25f070807a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db694185e02490d98c1ec8c18c02759 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/071a7da453584267973d81d7cc721612 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d57015c879184ff69857942d2ec53411 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e038368416434121820d8149f208bc6b : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a444ced557e74ef7b053085e0abdb27e : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85749f9cec18495cb7d6bbd6addb7698 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ceb413fba624420caa1b8d21b4077ac3 : SUCCESS in 7m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/81681cf6d9104c7a879962b98f475c38 : SUCCESS in 6m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c514d565236428fb0ef9ed0454e6dcb : SUCCESS in 8m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/44f6dc5e6644458cb85252189220d7c7 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d75dd02ec8b54b019772e208500f0a24 : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1a343ed00534d7c92a97a06c3e47179 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bec7d2fa7ede4020a43626ff43857de7 : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d907f81d24f44f80a7a2f6e75fde9112 : SUCCESS in 14m 40s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b03583a5643c44bbab5edcc95400a347 : SUCCESS in 4m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/28434dfa5e1745068e466aace94db38d : SUCCESS in 37m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7ab655c21fcf4a459f574afef27b1b13 : SUCCESS in 13m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1884b4f047bd4618a57fb88bd21cde8f : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4062dcb694d048d5b44901810339f6a9 : SUCCESS in 15m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a102c2a7fc904cdfa615274cad6c5c3f : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5a93475fe934c0fbd62ac8c99217f8f : SUCCESS in 5m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/97cc1e067eca4a50a4e903779f9eab09 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3a0fd7ccb7d43309518849afa5c6b94 : SUCCESS in 1h 07m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48ea097ad9c046baa8ff48167c6f71da : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":3},{"id":"d129322d1600363add799928b642b709f5d12d20","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 16:33:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5d840c25f1c21c967244a7d9cc8e6c809d2e07e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:34:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c6b0c09c13d63a4992824729feb6e46aae977b0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:02:41.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2bf5a46aa914590accd157206682921\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce3cf3421b146e0b2695fbc92f33f48 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e907b6bdb06240b084384bbe4027a0df : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f0591aa5ce459780dc59c42dfa3960 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e76b042573044655a1e6c704a684ad37 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d5d7357e28ea407c943ab3e12e008742 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/363b86502f514b21839f3a83c29d947e : SUCCESS in 10m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc01a621582b448daea885c8df53b92e : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/861e59ee88464fac9bd4d7746247e96c : SUCCESS in 8m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/88f4e370f2b645fab74d57389d5b9c56 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/217dd6cf88fe40e7b8aa7236a6517858 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21eea7e7c71e4d57a5f7959299c30b7f : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fde31e6f3b8b465094bf32cf6d25abf9 : SUCCESS in 7m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b96e18f262fc46a3916721781266464d : SUCCESS in 14m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b9a095a2dee34391ae01da124db9f6e3 : SUCCESS in 4m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5ab44a835205481f8ab3daf34b4ece3d : FAILURE in 30m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e7dc8eb6d284f51a5b143d9872613ff : SUCCESS in 12m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/06ca366512aa4073bf2b5b7095a86fa4 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52dbeb8614c24084888e35d264191ccd : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/538ee525e3424185bd3bb74433b7a2d8 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb692adea7284e219f5194a339329cf9 : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/173fb314458848f1830b589f4a2a7839 : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce2fd0090baf43a89d20171bbc07a459 : SUCCESS in 44m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2340b9143364c3cb2fa37f01c3df208 : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1046f60e9fcbf67e09f8f45a48a6ee924756ff69","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 20:26:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c95f0d87b4bfd77c918abab0ac35c4bdcaabc5d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:43:14.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/925aa6493a734e46aac3d89b3ab083e0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b8d6a5d870de4d838fb843b229c3b4a7 : FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":3},{"id":"e6acf0ba4fe2efb23e5292c73c865c923ad0bc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 21:45:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad07f2f0bff34922aaab1ca9639dd765\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75827b020a847a1b5945ee9ce5a324f : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1134133cc30460ea63bef07d2285371 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/352c74427d7647ceb24a7da868dadde7 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/571e0791f8164075aaf9ab2259e3e258 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc530fbdde1e467786f964a4358a8c60 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/188d8df7b75d483ca2609c8521ecb104 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e4bd3e5a92d4f8ebb79582d4568aa32 : SUCCESS in 9m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa76e390dd634e3ab474d0430d0b6137 : SUCCESS in 7m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/73f9e8507f3044b6a748615ff7a256e8 : SUCCESS in 9m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8310d3fe8bc946dfaff3a0e3aa2c6a09 : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63009f0a7dd64e3ab634b64a6c45ebe3 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/289b939a097e4c28a67f3c2b7fff46bf : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4f68ddcd31a479fb1966468be227b38 : SUCCESS in 7m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0fb32e1a3818443e8ace7f6dd5d34c9b : SUCCESS in 12m 39s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/446ea806ca3e40fb8b09a35102681f9d : SUCCESS in 4m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b0a4efc367174d7f8a28b52f77ce60eb : SUCCESS in 26m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9aea6cdf6ac431ca3763f8250239e39 : SUCCESS in 11m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab90a90d3dda4481bd70b09cc3e554c6 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84dc921e695a49cfb4cfcd8922cb5727 : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/51fcaf8e0d0d45a19aeff551bed2a75c : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff0772ccaf4643cf81a8ff970c31fa73 : SUCCESS in 6m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2dfe058ff774438bae2cbeceeaa3fdb4 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88b2b4d9acb64b12a581a5df541ab9a4 : SUCCESS in 52m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c0b1aebf8814cb69464eb1273bb1e84 : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":3},{"id":"1e84e5fd10ba5392154b28ded380c1596abbda84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 21:46:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"460f1bdf06677ddf52f1deb87c4902727ae271a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:29:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2abc7a97edf449ec850f0c3acdb7989a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ef808f0a04d4969914b77ce91f412c1 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9f848832fca497b975822c2121a6626 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab25e9ff19794057be31828bad78ae43 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9bc8117a3202489fb71a38f5ffa1ac6c : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c396c0045cf4b3688556dec19692067 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b9fe7202eabe4001a65495f1021dbb86 : SUCCESS in 7m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/df3bc06c4e3a4e3dbcd1c6a20418aabc : SUCCESS in 8m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c511acd182d49d1a58d4eb00d3797bf : SUCCESS in 8m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f6a3a7cf9005475baa401c3118738ecc : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b55e664f5f14e238c55ee5ced3de646 : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4661ea2962204d27b92c1823f8330ee5 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cc304d3d8725455e99368dcf09e4330f : SUCCESS in 8m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb59b78fcb7740199c0d325856e4ad3a : SUCCESS in 11m 59s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/39183ec47ca7487da49506dc49b6940e : SUCCESS in 4m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3c0cc85d09e1451881e9fe731c29bb8f : SUCCESS in 25m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/105f54fdb0ee4ba3be38c685da032667 : SUCCESS in 9m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7c532e6162c42d19e6eff0332028dce : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f74e6fed5bd4addbf3f10ad52f5fe98 : SUCCESS in 17m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c2e008decfe5419d80e7e7d568554033 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09a0f1d63b4b4e4d9387ae2d86b5e023 : SUCCESS in 6m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4986a1b45fc1414183bbc6315e286ded : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5a4bc39457f4701bc45d5ac355545ac : SUCCESS in 1h 01m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa579a1a84d44ae5bac1513709ddd8aa : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":3},{"id":"49ef99bc67b6d5b374948ac84955c7f0cd35b511","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:29:15.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6cadd874e2214443609d9efb10cfa4c4d0cd32d1","revisions":{"bd6d236dc6ef2d6ab15dc2b1fb8fe147afbceb31":{"kind":"REWORK","_number":1,"created":"2022-11-03 11:34:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/863511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/863511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/863511/1"}}},"commit":{"parents":[{"commit":"4d8bfbfa366575994eec8affa69cac9d10de5613","subject":"update constraint for python-ironicclient to new release 4.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d8bfbfa366575994eec8affa69cac9d10de5613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-03 11:34:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-03 11:34:08.000000000","tz":0},"subject":"update constraint for os-brick to new release 4.3.4","message":"update constraint for os-brick to new release 4.3.4\n\nmeta: version: 4.3.4\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Commit: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Change-Id: I7d3501a9cd2453e545826d158223dcd1bddeda7a\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ief9acf6d570444002e1a8e0f6879cb925cd37790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd6d236dc6ef2d6ab15dc2b1fb8fe147afbceb31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd6d236dc6ef2d6ab15dc2b1fb8fe147afbceb31"}]},"branch":"refs/heads/stable/wallaby"},"2e61942ceb1e48ffea31ed859bd7cc4b5161914b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-04 14:46:01.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/11/863511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/863511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/863511/2"}}},"commit":{"parents":[{"commit":"2bbdb171101ba558b66212b776567d8852ec8d05","subject":"Update Wallaby branch to run npm jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bbdb171101ba558b66212b776567d8852ec8d05"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-03 11:34:08.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 14:46:01.000000000","tz":0},"subject":"update constraint for os-brick to new release 4.3.4","message":"update constraint for os-brick to new release 4.3.4\n\nmeta: version: 4.3.4\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Commit: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Change-Id: I7d3501a9cd2453e545826d158223dcd1bddeda7a\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ief9acf6d570444002e1a8e0f6879cb925cd37790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e61942ceb1e48ffea31ed859bd7cc4b5161914b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e61942ceb1e48ffea31ed859bd7cc4b5161914b"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"6cadd874e2214443609d9efb10cfa4c4d0cd32d1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-05 18:03:01.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/11/863511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/863511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/863511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/863511/3"}}},"commit":{"parents":[{"commit":"426c2fb84e5876e453efee21d11d8f03b0fec34f","subject":"Update Wallaby branch to run npm jobs on focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/426c2fb84e5876e453efee21d11d8f03b0fec34f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-03 11:34:08.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-05 18:03:01.000000000","tz":0},"subject":"update constraint for os-brick to new release 4.3.4","message":"update constraint for os-brick to new release 4.3.4\n\nmeta: version: 4.3.4\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Commit: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Change-Id: I7d3501a9cd2453e545826d158223dcd1bddeda7a\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ief9acf6d570444002e1a8e0f6879cb925cd37790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cadd874e2214443609d9efb10cfa4c4d0cd32d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cadd874e2214443609d9efb10cfa4c4d0cd32d1"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
