)]}'
{"id":"openstack%2Frequirements~863601","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Idf1607d358817ea8b198382a2762cc6d288801fa","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-05 17:37:04.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"Idf1607d358817ea8b198382a2762cc6d288801fa","subject":"update constraint for horizon to new release 19.4.0","status":"MERGED","created":"2022-11-04 09:20:49.000000000","updated":"2022-11-07 20:17:39.000000000","submitted":"2022-11-07 20:17:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863601-new-release","meta_rev_id":"db2758a065f15ab443f967e7f52748cd816697f1","_number":863601,"virtual_id_number":863601,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 20:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-05 18:02:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-07 16:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-11-07 16:33:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 09:41:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-04 10:57:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 20:16:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-11-05 17:37:04.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"21a2cd2b98850df96200d6eeeff217d25158e599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-04 09:20:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abeb8efdb321a11953ba3e930f5b664386c96b71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 09:41:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726a8867c8894e54aa473dea2205f387\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f92cf5b3e5ed47e880bb7b17ceafe41c : FAILURE in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6b3c238fbbd2ddf2c9a86e7d9eb49dfda5069927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 10:57:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03d0ad32882c4498947e875bdb7894ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b24e9dc368b7437686a41296f328cca6 : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9b007583e6e4b089b8a8a1ff2915ba9 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b43dba6079446e5bf6daba9c57caa3d : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75dcc8f7bbb241d6a5693127b8a3cd2c : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4da8aa7325da40f9a79424f918e1dbb0 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5389082bfad14240bcf23aef5050c057 : SUCCESS in 2m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6de43a8b21a947e098d4ecd1dc711bd0 : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b2a26277848640f6a219b25cf1d8533b : FAILURE in 3m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/47edb2834d2b4de8983188ef96ae7396 : SUCCESS in 9m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7a12f78023624695bdc26e703aa31c98 : SUCCESS in 4m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c3618d36e2ce4b68811cb2ad78e76ec4 : SUCCESS in 7m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9b56350dc27481e976b4a0035f33efc : FAILURE in 6m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9ea6436e57ab434abc8326457c9f687a : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c4aea1242f694c63806c0f7d0bf8e235 : SUCCESS in 14m 00s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a894553ba26d45b6828cacc116b2b52e : SUCCESS in 4m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6e40ce4795a147aea9f9ca4d403722bd : SUCCESS in 27m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/664bf1e4ecba48a7800b97ba739622e8 : SUCCESS in 12m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52210e7eef6b47d3b58c4a442d756007 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/504096fc71574ab798b302b0a52097e7 : SUCCESS in 19m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6ecd9e78ab8e4a0ab0ff22e24a47b877 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e89aa8bba6b4a99bc30515566e0a85a : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b0d6a820863c4d6980ac8ba5b619d42f : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d6f46030f5d4a7f9d6376b9d492c4c2 : SUCCESS in 1h 05m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3060d5eccaad4a31b4869ff4c4c9e845 : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"75fbdd4720299578b9e95b91ce0cf526d253a281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 14:49:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a092518c5347cb46add5bd10bb2ba323e2e9006a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 15:12:30.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276db1d237c04b0cb83d912fb658e681\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0e07e3ef92ec47bc9b827df5a0810491 : FAILURE in 18m 29s","accounts_in_message":[],"_revision_number":2},{"id":"abcc44b2600d207631fe800fb8616e40b2761d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 16:17:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e47fe64896f4cc4acf47f9e7b5da864\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd5915dbde746ee9dfe2aced5b6be98 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e28e954e7ae458d868f3f52dd0f66d4 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd0d79084e1146b081943554a0edf952 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26ebd3a4619d465ba81d435d381f5a2c : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b0dd4f532974cf18dc7a8dd7589cb22 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a115f8b07bc466ca6541ab5c1fdbdc1 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e653fcdfb77f412d95a0dc72d19d3c87 : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a6bd5fd0f1a4b3986c1e71618cb96e2 : SUCCESS in 6m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f86710d73a64e389fa02dd67d4e84de : SUCCESS in 14m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7dc14fa2523746a0a5d7d0c57616632a : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3de25faec14b4564bedd51c8ed9efdf2 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac0ad9539edc48699083072a8b40f7f5 : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e653121650dc43ad954caab247272c83 : SUCCESS in 8m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ddbd08986a254e3d84cea04202b93081 : SUCCESS in 17m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5dc4bd6814ae49c4996d3844e819277a : SUCCESS in 5m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f7d5cb14518544d9b0ad15381705544a : SUCCESS in 27m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/565e360aa66f4787a063830d69556657 : SUCCESS in 12m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1638a19ab3841d98a223d2beb99b6b8 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5214d3d82813470cb7a78b81642a3195 : SUCCESS in 19m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ee941126d2f84583b7ba2c9b7691b806 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/125913c5c4a642369428353edb370d57 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/230dcc8b911349a793471977ac0c7a65 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebb21e5849eb40a1983b66e7c39c4a75 : SUCCESS in 53m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3099a2ecfd4c4a42b167abf7d7fa90eb : SUCCESS in 1h 24m 42s","accounts_in_message":[],"_revision_number":2},{"id":"35b4fbf480ae4a97e0712ae6082cb77bdc0f5f73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-04 20:16:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"15e96226d6ae81126fb4f6b554290a8cc5f08e7c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-05 17:37:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a77a573b2a93dd0d3a8c603447a27e05a045522e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-05 18:02:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"11b9942cc906987ed0f55daf19b3f729b0b398d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 18:15:52.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fa00c99acca49d89ae6f1033f752194\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c7b0174a92ac4895987a5d613658072a : FAILURE in 8m 13s","accounts_in_message":[],"_revision_number":3},{"id":"4c234e92c9c3a62935b70aa5833c3cf054beed0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 19:53:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e5d25f42d24ba2a13a13463f851b45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb15aee571b44784be69d04bf3a94bdc : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa5c797865824ff9ae8e8eb807c5458c : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/836a75d9ae694613af03ef5fc59afd86 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9407af2b3b9f4b049234bbc10944a5cc : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32e904168e8c405eaa91605c70256d10 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/efb5f639eb6c4d26b742421192d424c9 : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/14eee9dfe2c24609822bf607265cac0c : SUCCESS in 10m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3a84ea76ddb2454e8f780e59d83d88f2 : SUCCESS in 7m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a4f668430bfd4e5a92e58890353b1c42 : SUCCESS in 11m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b1ab5bc34701404894534cb317880309 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/176cf82f21a74ebb9a031ec1e99c9a2e : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0bbf7eb611a4c06b6928ed1818a42db : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3f9a52d0fd954368aff8af150b3212de : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cb997869bf1449f3b31d87eccd68de99 : SUCCESS in 13m 14s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/365541845f324147bd37ef60989cd642 : SUCCESS in 4m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a130396865314924b6fc905e3c9131eb : SUCCESS in 30m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9daf2f3503654bdfbe0a3feab28e93fc : SUCCESS in 11m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6f59099d6c345dab8212a9a23b9d65a : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0cb8c88d531348a18830c894e91a5fc6 : SUCCESS in 28m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/64c03a27dcc64552a6bb82d749b9887e : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d56efcc265a748de85f9fc80bf28ba67 : SUCCESS in 5m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f0ea7760e9a34d56be05588ea3015d12 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/561d54471a324889b6a41a09ec040998 : SUCCESS in 1h 08m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8506122fcab46ee81b149a20812f6d6 : SUCCESS in 1h 41m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3365ab03d386c837a6c9febb5570efd7aa13e6fd","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 16:33:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d8ff0bfde7b70bd694b662df84898fd7f25caeb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:34:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ee356789434a5937fe67c4bb6b6c106059ccc4c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:17:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7047dc5797b49ecaff6d4cbc8318516\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5efbafc27b54457b400922e529bc3e4 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da1b61bf84b344e2bf7c59a20e8a5d3b : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/724e7fe5a6424b09b155b03ae8017208 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f2b711170e244b791c736987960324b : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c45313aea61c40fc9452967726684da4 : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4694a2b83db540ea95942707b1166065 : SUCCESS in 7m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d72010ae4b694556906f00d352b268ef : SUCCESS in 6m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3b0f36168cd842538850a0dd4ec3bf0f : SUCCESS in 8m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ac4ea86f1d4c4d1888d4d2f29a484023 : SUCCESS in 4m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8b1d360570574c11b77a12ee3a4fc9ba : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9a721b7cdbcd493c99fc56bfd5f9c4e1 : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4e1c02bbdb784bf0a364999129109efc : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b66162548473488ea101ba0247fd0d87 : SUCCESS in 14m 44s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5d9125f2454343cdb0cb7bd50ee84430 : SUCCESS in 4m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3716f3f89ded44c6b9cad3c8a9ae92bc : SUCCESS in 32m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28b562a0ceb144e6aec926fddb4e87fd : SUCCESS in 10m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/33ab2316c97d412f8958fc5da24001da : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ac5ec8d84794992a3f2294206bf5d61 : SUCCESS in 20m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6c340a323ca24b07ac4e1380ca1100ee : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6535a6b8de646e0adb3a9f3aa2f949d : SUCCESS in 6m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44701162769749e4b6b1f5df6b68f355 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41c5864941404f67ab39b2f0988256b3 : SUCCESS in 52m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/844b55a8cba6425e926491087654849d : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":3},{"id":"db2758a065f15ab443f967e7f52748cd816697f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:17:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7218638384d1fdea75a5c635419d0243f6d5ab6b","revisions":{"eafb02f1bca8e4bdbc9c225d044da8af9016ab9b":{"kind":"REWORK","_number":1,"created":"2022-11-04 09:20:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/863601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/863601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/863601/1"}}},"commit":{"parents":[{"commit":"4d8bfbfa366575994eec8affa69cac9d10de5613","subject":"update constraint for python-ironicclient to new release 4.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d8bfbfa366575994eec8affa69cac9d10de5613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 09:20:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 09:20:48.000000000","tz":0},"subject":"update constraint for horizon to new release 19.4.0","message":"update constraint for horizon to new release 19.4.0\n\nmeta: version: 19.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I1765bab313faf4593ffa142966adb94d09960303\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idf1607d358817ea8b198382a2762cc6d288801fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eafb02f1bca8e4bdbc9c225d044da8af9016ab9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eafb02f1bca8e4bdbc9c225d044da8af9016ab9b"}]},"branch":"refs/heads/stable/wallaby"},"761e5320fd585df129c6bc82c2967c3969002b49":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-04 14:49:21.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/01/863601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/863601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/863601/2"}}},"commit":{"parents":[{"commit":"2bbdb171101ba558b66212b776567d8852ec8d05","subject":"Update Wallaby branch to run npm jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bbdb171101ba558b66212b776567d8852ec8d05"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 09:20:48.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 14:49:21.000000000","tz":0},"subject":"update constraint for horizon to new release 19.4.0","message":"update constraint for horizon to new release 19.4.0\n\nmeta: version: 19.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I1765bab313faf4593ffa142966adb94d09960303\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idf1607d358817ea8b198382a2762cc6d288801fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/761e5320fd585df129c6bc82c2967c3969002b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/761e5320fd585df129c6bc82c2967c3969002b49"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"7218638384d1fdea75a5c635419d0243f6d5ab6b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-05 18:02:45.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/01/863601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/863601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/863601/3"}}},"commit":{"parents":[{"commit":"426c2fb84e5876e453efee21d11d8f03b0fec34f","subject":"Update Wallaby branch to run npm jobs on focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/426c2fb84e5876e453efee21d11d8f03b0fec34f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 09:20:48.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-05 18:02:45.000000000","tz":0},"subject":"update constraint for horizon to new release 19.4.0","message":"update constraint for horizon to new release 19.4.0\n\nmeta: version: 19.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I1765bab313faf4593ffa142966adb94d09960303\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idf1607d358817ea8b198382a2762cc6d288801fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7218638384d1fdea75a5c635419d0243f6d5ab6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7218638384d1fdea75a5c635419d0243f6d5ab6b"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
