)]}'
{"id":"openstack%2Frequirements~863604","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I7e8837e9808b93eb0fa630fa758376d80b14b125","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-09 23:12:08.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"I7e8837e9808b93eb0fa630fa758376d80b14b125","subject":"update constraint for sushy to new release 3.7.6","status":"MERGED","created":"2022-11-04 10:05:15.000000000","updated":"2022-11-10 02:23:52.000000000","submitted":"2022-11-10 02:23:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863604-new-release","meta_rev_id":"92c6854573e9113e3ec7636bd3671ef61e960fbe","_number":863604,"virtual_id_number":863604,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-10 02:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-08 00:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-09 01:21:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-11-08 00:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 10:10:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-04 10:16:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-04 11:53:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 20:16:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-11-07 16:33:19.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-11-09 01:21:52.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"ad6cfd830e75b0a2cabe3f17d237f6440e0f7efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-04 10:05:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d6150ef97f59c5facc65e0c21bcacbd54124821","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-04 10:10:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9f561bf65879422fc415d0049685b46496b82be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 10:16:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a026c8544c5b4c498e78607e212d9a1f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5b45f94caf394ef287c5fc61d1366a40 : FAILURE in 7m 40s","accounts_in_message":[],"_revision_number":1},{"id":"49a7e19881a635452428bf2e1ea9db72541d8c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 11:53:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/caa14a06d85e448ab04809d072b5ec80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154778df62904b5283345662ae360bdc : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd4a5ffdcd0e46ac9f4ffcdffdd09005 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d17dead65094110a5642a74bb535882 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b00041021d44166bafbb40e8c14844e : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a24fa250795b439eb1dd6148d0e27831 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee76fb6f57d34ca3a41506337a0ff1d8 : SUCCESS in 2m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46fc0546072242818554cd0aa148fa50 : FAILURE in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f3c63ca789049e8a6e1293738d3c91d : SUCCESS in 8m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a93aa4ee8724ec481b21c19fa6af2f0 : SUCCESS in 8m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14800372b6624b908a71c70f2b9a30ec : SUCCESS in 4m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/24f09a6798e44e119d41c2490b3ee517 : SUCCESS in 7m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42ceedec13174ae989998c397084ca13 : FAILURE in 6m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23f4e9ec32a74737aa8224b5b1485dad : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da423737e9264d69a99af23b13038b59 : SUCCESS in 13m 50s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2b3f67d788a34f2d8757ed05728a312c : SUCCESS in 4m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a0c843a905b046f5819fdbb301cf2150 : SUCCESS in 26m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ffdb1e8d0c1407b9159e27ded9a4ecf : SUCCESS in 12m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/12035cdb377042678f039791a7120d70 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1482f8e21b65417d8a9a560b0c185f72 : SUCCESS in 15m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c5e622b1095e47dd9bd02c5278dbe578 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6acf8f013894836b2ae9d0e1f9df144 : SUCCESS in 5m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abebdc45f0a545cbaaad9133945df459 : SUCCESS in 9m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81d7c0407bef4cc08cf28b875aef3ada : SUCCESS in 43m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4e71287ec3b435ba596211b61b60bd6 : SUCCESS in 1h 42m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e2a730ca3922e27e9d2ce8d59a1194708adf9b1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 14:47:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3f22713b5c67233bf8890745f035d091ae7e914a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 15:06:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/decd7854aafc4110ac6c98215e71dd0d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9305dad22377493d934d5e7cc6d8576b : FAILURE in 15m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5b6d4007edb87d0499321f69d0268ccde4ab568c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 16:16:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a09c79b9a64095a8991fdce2e26169\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b2fbc242234d47aac1ee78692070a0 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c22367f4c82484091fa887c1c28c1c0 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3699824ffad146caa7f2fdac74c0dc53 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89d5c7c74e544564bfc4a764ae7ffebd : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f792f0459de478b8baceae51d84960a : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42817e77bf00457c8553d6fa5f121cf0 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/33f67f8e88c4465bbc4d7e5b27c7e613 : SUCCESS in 11m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af1aacfb0ee5452884d1246f556ecfb2 : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9cdf47e1e6a9429e9c636755937dc29f : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f6ed288fc3134c7a8ccaf65c181e201f : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a8a3c47af0ee434f97f07f306ae560c0 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd8ec7a9914641cf8f0a71883df9f911 : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/46b64d402c004d56af1fe4820af436c9 : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eeb778167ed64f598f42af9fb836a834 : SUCCESS in 12m 47s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5e2c004d68ef459a92c2694b67024fea : SUCCESS in 4m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/495b74ab5973490ba62c7bdc5058c053 : SUCCESS in 31m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e852f818e5543eda48093b2cc4f424b : SUCCESS in 11m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f1259742dc24b7d8664ed6398a8300b : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5804cfcb64664add99cfff890db20bc0 : SUCCESS in 15m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/660d4d40461249d98a946217a1a2e44c : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93f5e92cbce14073bb1de3d601e362d6 : SUCCESS in 5m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e60220a9036e46e495485fa5eb2acfaf : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12a854e764104d679ac11da6261f9e8c : SUCCESS in 59m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a15b5335772c48e0bddc0c248ea7eee4 : SUCCESS in 1h 26m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3165d3f416e52e2fe76d568c5d4ff2ed884b5373","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-04 20:16:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e7ce27fce6f528b09dc2ea49b6ab1a9a33dfbeeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-05 18:02:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"15d346597c2dbd5a5751f8e2aec9224da67d135d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 18:20:28.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b2ccec5d304f7cb2d429b65ff780ba\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8d35886c4126448c9a9608094cfad6f5 : FAILURE in 14m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a6353e0a1365e9b3659165a4b25c2a09d34de2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 19:16:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e7326fb9214bd9917b8568bf03bace\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56da8bf841443ca9fac04d2557d2a59 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb3f8819b46d453aa10024c8de31c15a : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34db623e3a9149c4b95d68ddf72ba3b0 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5950b91ca7c840cf81456b3f2b87d21a : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3a9bec05e764f669329c3253cbeb0b1 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26d43558e871450d82844d32020312fe : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc1fcb0bbd174b8b9c5e46eb55179402 : SUCCESS in 7m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b877b439b9e549048141ae8ff797f58a : SUCCESS in 7m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/46f65eaf48d9466bbba0597cc09e9da8 : SUCCESS in 11m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5f10590ad620421bac5133270a492e24 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/969798fac7294c6190a8b1291e3e78cb : SUCCESS in 14m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f263233c75234383992cd98fd3485f3c : SUCCESS in 7m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5a8147ce546e447db6e897a2f0108585 : SUCCESS in 8m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e72de875f4464d1db8e87c162554eee1 : SUCCESS in 13m 33s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7e85c7ecedf64a129d6d77ed2bc24fe9 : SUCCESS in 5m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/65d5a866144a4d4f913db32811edba8f : SUCCESS in 23m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/34a2712e60ba46d194c42999171f55cb : SUCCESS in 11m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1cb173d72d014f3890cfdcfd05c56570 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/928f25f919bd42e2a491e11492083f3b : SUCCESS in 25m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/36fbf21d9c504247b8ff03772529ed69 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae0e58a93b914407b8d129f0aedb34cc : SUCCESS in 5m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/123b89d0e1124f64a9a566f93fd81f3f : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e38c7df3cf3249a884434259198ff478 : SUCCESS in 52m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bad8d84ef7c4feabc2c7cc490d67672 : SUCCESS in 1h 11m 13s","accounts_in_message":[],"_revision_number":3},{"id":"86105d46e7eec0a3b8c254f058ec50ecb513ce82","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 16:33:19.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1a66a0f3ab9a616e1207f799a849c252a6df7add","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:34:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da219da264ad8c64dcc54f097bba6284bd9fbec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:16:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c98b3769db9447880c5606ede21463f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b49b39bf382d477b9917cbf9762680d4 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/040fb612b6ab41d3970414de295928eb : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cac8ad14141643328e853831f2ccc60b : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26722b3a42be4503a3676adf6f6baae0 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/704048d643324d0dbc633b996e9c5a83 : SUCCESS in 5m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b518243baa17483da3911f736dcd2688 : SUCCESS in 11m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8a46ccd802bf483e83d2e4279ea17156 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/332e0fa403594fefa2b59dcdde69d3a6 : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d4e3eb6e96b44ac29a84c22af87837b7 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/49f621a2f6da4c1282601612522e081b : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a21b306130a944c18d2191050718a6b7 : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f26a7b9d006d4965a31cb713a095424e : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a2dacc6cc381444fbfd39e16a1188a56 : SUCCESS in 13m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/942d1065931741ae8c78e25489051c3c : SUCCESS in 4m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8658681bde3b43f0be4c1b4d03e13f85 : SUCCESS in 28m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc3b7280846b4b849d29a3b3e17c1aca : SUCCESS in 13m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f2182cb315b43be822ed63a3e8d1817 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74e9036ab72e4adf9798370e8f09baad : SUCCESS in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/50f24690c18c4d5b9fc19c85b487e82f : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a609d4da7d144cbb890b949f8524914 : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/efac5d77462e48dbb33419dde0820c3b : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8d74d8038844528afa76cc0f80aaeec : SUCCESS in 56m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa8deb4cccc541b58ed68c3825d195c0 : FAILURE in 1h 21m 33s","accounts_in_message":[],"_revision_number":3},{"id":"b5813112eedfbc78a678d97856976c54c459e1a8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-07 20:27:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"04358f96eb8efd884e919d2db64c1daa2ac3fff5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:43:10.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/911babee72704a1390f2ceb828045971\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3b01d188660f49cd8105d366c317b6fc : FAILURE in 10m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a87927d52ccd076f8f4292d63f0cbb5704c8e4f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 22:26:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753816e441b640c4b93623e226ffa220\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007894d8f61b4cabb8cfeefc574cf8e9 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a410fad5a06945f8be01622ce6ca725a : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/201fe550d2b74dca9c30e194b115d4ea : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83eee94482984812974c1ea644606a6b : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc7bb222a17d43bd9c538c2429e68b92 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4246231cec4f463aadd7f843025249b5 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2ecf51edb12a45d9b19e0695c3cb5507 : SUCCESS in 8m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5e9af283778e4b20a33f1af1d28cd67d : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e5a4192b4fc445b9be99ab4a676b0883 : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1f17979a03f047f598e3d120c64d03cc : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/52d22ce403a141d19a18565a77d8523b : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/651d0d33645f4ab39853b15280c41d82 : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d456ab985f424d069ed9798e5af59a2c : SUCCESS in 7m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ac68ea66b61a4d95b8f9e93d1ebaed4e : SUCCESS in 13m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/22b860e1069743248afa0b12989f076c : SUCCESS in 5m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/790fda945c6f4d96b15f90f76b221f3b : SUCCESS in 24m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/13c4ea7dab7a410992a74a803871867b : SUCCESS in 11m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee508406af844889b41f91553657c407 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7312f28e86224f4c8f9cb316bab89d52 : SUCCESS in 15m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c6023e50a3104357a1f4a784487c5d00 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d90591ee67ac4fd7b6c3bef7cbd5e995 : SUCCESS in 6m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bea5b7bcc41c473bade765a378a6be17 : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/845107b5289f40c8b2a34bb46f52cf6c : SUCCESS in 1h 07m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bd5d093b2b84abc934c3969651b87fd : SUCCESS in 1h 53m 15s","accounts_in_message":[],"_revision_number":3},{"id":"f71d0cb09197b41e32f1db7a4e427536bfe76a0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:46:35.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 863604,3","accounts_in_message":[],"_revision_number":3},{"id":"db43d2dfff3d26c34c6b98b5fbccc510ae8bb52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-08 00:06:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"816a01a66ade5c6739208ca80fcd9279799f8a6d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-08 00:09:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"07b2bf939d2b6b0868b51866eda7a68ced9c5ae5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 00:20:10.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1bb3fc7c2c3467780b9284deb295bc3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/444bf0b955f44b7db1492eab2802aaef : FAILURE in 12m 50s","accounts_in_message":[],"_revision_number":4},{"id":"f86160d61c209179c1fc3f83cea675d64a421131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 01:22:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27642d08b4c4615a6c47834043050ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b323c6e9b334472b681fba9e484b4bc : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93edce6806994f0f90d91a13f203eb82 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0862dcf197ec45aa9c1041be920b0511 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d308c7ddeadb45268a282e408589efa8 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/71e48ba18c31423892de29150b543932 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4022915341ae4cab9c568c74a40b5ac2 : SUCCESS in 3m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d7d053bf1c894517ae993b475848e9ce : SUCCESS in 8m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8a69ac6192424d4385a11dea03bbc15a : SUCCESS in 8m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f3d9cbe146f48c4b702ac83d4ef4889 : SUCCESS in 7m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/99c50793580d48bd8c7db1401dc10c37 : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ec1b6e890e1c4a05be5897fbe34babdb : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53889935a46945ae9baad3d2888a9afa : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/16ebc51e89ec4435893cc3fef1faf574 : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b89980faa7bf43cd9b46e783dd313dc1 : SUCCESS in 15m 23s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a6d692819fe84f7d867c1c6b661aceb0 : SUCCESS in 4m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/927820f7566c4473a46d251e271b516b : SUCCESS in 25m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c5e1bf29f9a43b29d8e59b662a99e0a : SUCCESS in 12m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/109f3613db094ee2b945f2c3c98c0557 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e45c718ce0a4057be370b30e127dae9 : SUCCESS in 15m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9420002dd2ca484d83d0020ce17612eb : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4db2f5d45d954a34b4ed2120817e3c8a : SUCCESS in 6m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/047c18a3a6544f3da867f53e46d2df5f : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74c959ef4d054193ae3a9c02d9b4cb0f : SUCCESS in 57m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a49525e45fa4eb29717bf1ed97e30d2 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":4},{"id":"130b7357f7bee096b3d5eb345e87fd57ac3d2871","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 01:23:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d48cbd870a144dec663449f155ca8e25cdfa2f91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 03:57:11.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8eeb4733a9d449b5b9b83513383a9abe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9323a1e1b74850aa89131b1c7b17b5 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54255dbf6ba643c1b7e8221f04cece21 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5917ad6882f849c79657ebe00d2917c2 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/240088b6d6b84468902858174a76e0d2 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40a500ac20bf42839f1bcc19ff27e25a : SUCCESS in 3m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a7cbc678133403ca476911d595f07fd : SUCCESS in 8m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cced446aefe34a059a456536dec105b9 : SUCCESS in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/187b56f5163242939d3a703fd14a53cc : SUCCESS in 8m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab848f7770f14fc692703fdba10fe32a : SUCCESS in 4m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ef81204ed2f04e18a2a00a4be3ff9070 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38f729080e6044f2bc65a0456a1c19af : SUCCESS in 8m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6462c7a09c9048ecb68a9610f447e93c : SUCCESS in 7m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/75d8cf641e1248caa7c38521c545a345 : SUCCESS in 13m 51s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fcaf501db6f64d6cb41145e1d740c5a7 : SUCCESS in 4m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/524730ef5b9d421f99c0d5288b550679 : FAILURE in 28m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/41b4b40a689a4846a6ebe57592feded2 : SUCCESS in 12m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a886488064df43f59ea8a6b28c31fb6a : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29f2aa7f099649a397d8cf5f4a4ba4cf : SUCCESS in 19m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/30d9a8d457ad46ef99f93d48a41a2e26 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ba312bd9f944b7db386317d07d3283a : SUCCESS in 6m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/15d2c6bc81bb42cfaf0e7f5802525beb : SUCCESS in 15m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2684b46f5f4647c19b7376fd595191ac : SUCCESS in 51m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bf13941d38042729aad7a1d50cde841 : SUCCESS in 1h 51m 00s","accounts_in_message":[],"_revision_number":4},{"id":"42986c65cbf741948d92eb304bf1801b97221b6f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-08 18:46:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a91ea5b91ebba9ef904c0798ecb7c89b1f3f4932","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 19:04:19.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/251710860d59462e8da2b8a43bff4f69\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/55048c6b88fd42a99d4f5166c5fab841 : FAILURE in 15m 08s","accounts_in_message":[],"_revision_number":4},{"id":"f053cc4606a3883f14d3ed9c9596b0e420c8cbb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 20:30:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94d70f2e52bc47eba1b50514e06f8e3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5c9e073491e4885bdcfb62bc824e201 : SUCCESS in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28390e8fd2ef4f3aafd09074fef7079e : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a43a126d4a734c0390459e7dd586c139 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8f5ffc19a414c9abb53abfc76e3c91e : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c0161df38c44cfcaa3633893d0de25c : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30189f8d07334e14a3e77d9a385aebc1 : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/746a5449af104c27b0b2793005f533d1 : SUCCESS in 7m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bac6baa79b0946db8ace810471658c68 : SUCCESS in 6m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/21e716f31e624242a53da03c1bfcf0b9 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/61dae66f0cbd41c29040539eb62e8863 : SUCCESS in 4m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/72f845a392134b43b77485e0bdc866e5 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0708feeb2026498dad996c9cee628bb5 : RETRY_LIMIT in 10m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c13237697bbe4c858d1e47d54cf34420 : SUCCESS in 9m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/adf5714299344ec1bb2712d7e168e2e6 : SUCCESS in 14m 53s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f4e00025db21491091295b6471762c55 : SUCCESS in 4m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c121112191af49ccbcb0c167d312aff9 : SUCCESS in 20m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f5c392d282cd42d58c69a44dc961a576 : SUCCESS in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2cf0267f3a9846adb356d9e7c925455e : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b77fb9cb17b8487e9775e6058373fbda : SUCCESS in 20m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cc21e990970f459a8e43a9c1f0ee4ae0 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f1893ac90e942b5bbe54d73e1db8c75 : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2a6ddf4c65e1482a810c22ee4a47a795 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b89dd47c75e84c14a0b3de6e890fe815 : SUCCESS in 49m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac248dfc968447799487ed6765fec96f : SUCCESS in 1h 38m 30s","accounts_in_message":[],"_revision_number":4},{"id":"80dd0dad8c75123fa096945ae608ac68905a0193","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-11-09 01:21:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ef2a8ed760c17af400c8955e60591ce46060e872","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-09 23:12:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"469cca189810acd611135427e0c70108383f66f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 23:30:57.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ed02fb0e2d047bd82ba924446362145\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/88b7018476aa46e88dbbdedf19bbab71 : FAILURE in 10m 58s","accounts_in_message":[],"_revision_number":4},{"id":"6b26d8898bd218cbb832227678bcd89fa1b56039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 00:41:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00521f938c744c09804ed4f32931bdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f45e08fe5ee44d7bba2f40a42baedc9 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d076917013246c39f5f9671ff803f02 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c708ead9adaa4a34986bf446a9debe5f : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da7cd429d4a14ae58bed9f062d23f392 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9aa2a2f5b514352867424697a460b57 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca90bb4db10f4a1187f27b6770c9bf54 : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/42d96bbc4d084577be32a2271351902d : SUCCESS in 7m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8bd76c284fb1441da8b4309d96804eff : SUCCESS in 6m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3d09f300777b4f0180bb1ef44dee7045 : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1890db73e09d4616b27ea145c2a25534 : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1639b319fb12437e8eff33f810761b33 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f22baf9380f34559ae2b5b163b42863c : SUCCESS in 11m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a80b9b13cab44a65bed95ca42e53a835 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/00b0ba7e2a504f278efcc09198de15b2 : SUCCESS in 20m 20s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/424b9697e4ba4569b75b6c007709d506 : SUCCESS in 4m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a7f2dcb4c71a4e799c444f17483d6d01 : SUCCESS in 23m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/88b02dd4b4a94e7098005e5e9677ac74 : SUCCESS in 10m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/68597baa72554224bea8297374eb078a : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e324509f17ea460fbfd034570c3bc322 : SUCCESS in 17m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/abf179660db8431c85a017f5f376829f : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89557ce7c4e547bb992e4f6b0413f8ed : SUCCESS in 6m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/23273d8e07554aff955571ced8bf5452 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7385276c857c43f184e7bc286bf13d4a : SUCCESS in 49m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/991284c0e52348ed95ba232070790616 : SUCCESS in 1h 28m 15s","accounts_in_message":[],"_revision_number":4},{"id":"47e1334268bfdf81208c6357ed6258797e45a767","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 00:42:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aa8f071099115af723360bef82d69cc9cf5114ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 02:23:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e517a0df9d4c418196bb8998bfda0a14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73570f761c214dfc8ba8577fd375f5bd : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc38484cee5e4833a4be9a21c2ae8cb3 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba03083ea1a24c8b9220a7acb9eb9c86 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2181109af4742b68ba2dfa7a10371a3 : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4a47d4794b74d3ea13fabe6ab99b3bb : SUCCESS in 2m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2bb32741c0254439b67b58a005723032 : SUCCESS in 8m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c401fb3a924a45278e86f4137d634512 : SUCCESS in 7m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/35e5e8fa965c44368dd59ac186076e9b : SUCCESS in 8m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf2c895065884dd088f9989d77e478d0 : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5d985b252fa648029a52389f31a3bb4c : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29ab28a1cf0d43379d5feba614b77905 : SUCCESS in 11m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ec8f77315e7b4946936c63d1a4e23bdd : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/409a48a129d94ec392da7fefb3f9946c : SUCCESS in 12m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0fcfebe81314483aa38846be27f12fb4 : SUCCESS in 4m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b015df4fa9b4917884d5ab726344a09 : SUCCESS in 24m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/64acb168e67947aca0ba66ac15fae3e4 : SUCCESS in 12m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/adcd8a5964e242128ea538a835c4c61f : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/507ed11f01de41708d6b116bdbe8f503 : SUCCESS in 18m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2879e1a07ddd45b480331de33ec7c177 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01a664df14d343528b028e3b6f2675e4 : SUCCESS in 5m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0641ae72c8a7424bbbbed50b6a84cbb4 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/566a5965e67744b3b2ae7c20cf6ff2e9 : SUCCESS in 42m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2487049bd614346987691513bc5ca8d : SUCCESS in 1h 41m 17s","accounts_in_message":[],"_revision_number":4},{"id":"92c6854573e9113e3ec7636bd3671ef61e960fbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 02:23:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b87589f0fe6d461274a544b95b1bfbb35145d17c","revisions":{"166583666e14586dee0675475c57148da15f46a2":{"kind":"REWORK","_number":1,"created":"2022-11-04 10:05:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/863604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/863604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/863604/1"}}},"commit":{"parents":[{"commit":"4d8bfbfa366575994eec8affa69cac9d10de5613","subject":"update constraint for python-ironicclient to new release 4.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d8bfbfa366575994eec8affa69cac9d10de5613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 10:05:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 10:05:14.000000000","tz":0},"subject":"update constraint for sushy to new release 3.7.6","message":"update constraint for sushy to new release 3.7.6\n\nmeta: version: 3.7.6\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Commit: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Change-Id: Ic99e361478a9504e3b7e30cb29921c5f5f7b1690\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I7e8837e9808b93eb0fa630fa758376d80b14b125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/166583666e14586dee0675475c57148da15f46a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/166583666e14586dee0675475c57148da15f46a2"}]},"branch":"refs/heads/stable/wallaby"},"e540a40f74483e9136ef1c5cc55462caf65a8c57":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-04 14:47:47.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/04/863604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/863604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/863604/2"}}},"commit":{"parents":[{"commit":"2bbdb171101ba558b66212b776567d8852ec8d05","subject":"Update Wallaby branch to run npm jobs on bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bbdb171101ba558b66212b776567d8852ec8d05"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 10:05:14.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 14:47:47.000000000","tz":0},"subject":"update constraint for sushy to new release 3.7.6","message":"update constraint for sushy to new release 3.7.6\n\nmeta: version: 3.7.6\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Commit: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Change-Id: Ic99e361478a9504e3b7e30cb29921c5f5f7b1690\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I7e8837e9808b93eb0fa630fa758376d80b14b125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e540a40f74483e9136ef1c5cc55462caf65a8c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e540a40f74483e9136ef1c5cc55462caf65a8c57"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"2ca3ec2596a8935876ad337f38c971b844f0554b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-05 18:02:52.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/04/863604/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/863604/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/863604/3"}}},"commit":{"parents":[{"commit":"426c2fb84e5876e453efee21d11d8f03b0fec34f","subject":"Update Wallaby branch to run npm jobs on focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/426c2fb84e5876e453efee21d11d8f03b0fec34f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 10:05:14.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-05 18:02:52.000000000","tz":0},"subject":"update constraint for sushy to new release 3.7.6","message":"update constraint for sushy to new release 3.7.6\n\nmeta: version: 3.7.6\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Commit: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Change-Id: Ic99e361478a9504e3b7e30cb29921c5f5f7b1690\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I7e8837e9808b93eb0fa630fa758376d80b14b125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ca3ec2596a8935876ad337f38c971b844f0554b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ca3ec2596a8935876ad337f38c971b844f0554b"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"b87589f0fe6d461274a544b95b1bfbb35145d17c":{"kind":"REWORK","_number":4,"created":"2022-11-08 00:06:52.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/04/863604/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/863604/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/863604/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/863604/4"}}},"commit":{"parents":[{"commit":"c3ea0f1ee39607de2c40a3beb01e95ec9a16ca98","subject":"Merge \"update constraint for os-brick to new release 4.3.4\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3ea0f1ee39607de2c40a3beb01e95ec9a16ca98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-04 10:05:14.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-08 00:06:32.000000000","tz":-360},"subject":"update constraint for sushy to new release 3.7.6","message":"update constraint for sushy to new release 3.7.6\n\nmeta: version: 3.7.6\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Commit: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Change-Id: Ic99e361478a9504e3b7e30cb29921c5f5f7b1690\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I7e8837e9808b93eb0fa630fa758376d80b14b125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b87589f0fe6d461274a544b95b1bfbb35145d17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b87589f0fe6d461274a544b95b1bfbb35145d17c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
