)]}'
{"id":"openstack%2Frequirements~863676","triplet_id":"openstack%2Frequirements~master~I211333a6da03282feba82cbcdf9309f1a17ee7af","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-05 11:36:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I211333a6da03282feba82cbcdf9309f1a17ee7af","subject":"Pin npm job to focal","status":"MERGED","created":"2022-11-04 17:32:11.000000000","updated":"2022-11-05 11:38:36.000000000","submitted":"2022-11-05 11:36:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863676","meta_rev_id":"b3035639adead58d0dc934b22e4fbcfeb16428ae","_number":863676,"virtual_id_number":863676,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-05 11:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-11-05 08:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-11-05 08:45:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-05 08:45:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7ec1fb8b7c94697ba9f998c69cccf0dc9a593bb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 17:32:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9667250f9aeedfdd8819c961194bd7f604046c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:43:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00346a5e801843dbbff59cf84e9dd379\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/addd7d21de6743499ff1fdf356bf42f4 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790cb662efdd463bb6c01a2c3ffd0028 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae2d7a966d7b413190c937e8364dfa70 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b35478d406514d0e96eb71691ed1f331 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48aa922d0b5c43cf9f4ae38ec61dd214 : SUCCESS in 3m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0c93c8b191184e8992d3fd88413b75ec : SUCCESS in 7m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/238ed8ffdc2d41fb8de5e306b7d8a37a : SUCCESS in 5m 02s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c188620590f44e39a7833c99bf46be78 : SUCCESS in 6m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/882cad5ebcb1469584c3c0ed593bbe3d : SUCCESS in 9m 40s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a600abeee9574b929cad5adffb2fe8c3 : SUCCESS in 4m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2d4e10f776fe4da4bb4179f71e76fc62 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f40dca3f33754de1b3ba8d6d0b328c3b : SUCCESS in 7m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f9673634b721413093560805f1a68db6 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1c2f10a07084ab9893ad29c165f3dde : SUCCESS in 27m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1179aabbe775432b9d19426d90ee7ccd : SUCCESS in 9m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f247f5162638407d89c1131919bb54aa : SUCCESS in 12m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/777bad3d5e52493eb97ce18e92051de1 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fcd92494b5284b479679a556702747f5 : SUCCESS in 11m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/39f9d9b47180431dbd6bd1b576f07f9c : SUCCESS in 5m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/551fa65831c44789a2f1b1404a0beadc : SUCCESS in 9m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/adc25e198be442a69a89a8dff9c1d6c7 : SUCCESS in 21m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b5d7f4049f524ec982035c2e28248bb0 : SUCCESS in 12m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4782cf83019d4a51a2422a6cac2e378d : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e5dd7437313430eb3a5458ec5f479ac : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/815603d20cd243819481146cf665f30e : SUCCESS in 21m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/548df782f6b24782bd3983ff6b3061c0 : SUCCESS in 4m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/59f39ba9779043468d3f35c332f6445a : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4db7dde34ff641eba72b083e149e8d00 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e5b69d23383440a0a5fdbaa067ff0f63 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77acabc570004d08aa1251aa9beb0967 : SUCCESS in 46m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ea17196397345a79d422fcdc38cd619 : TIMED_OUT in 2h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bf8c7ee2aea7463793b8cb2210318b0172eda658","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-05 08:45:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a807e59ef4ec6b80dac7b00c75c086f305990b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 09:54:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad4d1d41d46246019efd4bbd2ae32def\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a98fe5d3c44ddbb48c4810ea67673d : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e89fa9730eb42aeb5ddf245bc86d4b5 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/462eef709d9c40bba2992070a414a4c2 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93b685dec356403dafbe4ed4845c1734 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61d4366fbe7d4ea0abe8179b68791306 : SUCCESS in 3m 19s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0764232ab0a54525b3afe018541494b8 : SUCCESS in 8m 03s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bc8c46a43f814acc96789da8a344ce5e : SUCCESS in 4m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3876f29695cf47d6a50e25bea3d06e31 : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f74c006e7e014db4ae8c9b47be306680 : SUCCESS in 9m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d336f97d8830485c90e56ea0ade97a9a : SUCCESS in 4m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7d7a42f14a8c4c38b35fd74ff6407d76 : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7be9f59c259242e5bedd7a944620297c : SUCCESS in 7m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/66c3e19d35484d929514f571ac7d7de7 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f875fef230548f382808dbf7b8d884d : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dcc17811a3e344268e604ac7fece7282 : SUCCESS in 7m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c147a5ad8c4a4ed78d0d68716aaddf7a : SUCCESS in 11m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/26d7e1154112469bbf4e0ba5266418b1 : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4ae926b5d48b4459a5f87e4c1689627d : SUCCESS in 10m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/27cad728bb224bbe9311503bb1298e3f : SUCCESS in 4m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cb8e4ab127ff4075ae93828d3a832cb9 : SUCCESS in 8m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8a3dfe18dea043659ba60a0f1b673bce : SUCCESS in 21m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/19ad3858521e4ab99c0c2126a7a2555c : SUCCESS in 11m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4a2d9d06b7f84da4a7b6855a4b2a066f : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02208a9044544e16a6955b62ca4a2205 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a87fff5efd44350ba3562094669a9ac : SUCCESS in 18m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e95dcae4087a4ac39e9a9ddb0a4a37ea : SUCCESS in 4m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/da4d67b20640467cabe73df1b2034018 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdd6d97c3acc4542b9593bb67d505161 : SUCCESS in 5m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7a465107f97f4eb8bc11bb33a64aa1ab : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec6dd8786f214358bce41c73293e2e88 : SUCCESS in 35m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05068e62a0784ceda463ff68c89b33f9 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4a4b4c2f67fa332e49ba537370e61279d8d6a644","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 09:55:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"224a27d4a7c03a17fa1237aaf3ccaaface59a276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 11:36:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e6fc346b36041ebb2d949cfa33b0a4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1cd2cede93454b94247b709bb328fc : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d9f563810e4ccc84baf84cee7f2948 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4df3b5bd594642578b81581226ce042a : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf949656c2634b73b616a541b41ef702 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c558fbf3a994e9ab2e6c9597489a84c : SUCCESS in 3m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7cada74393d84378925566235fbbd57a : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ff55f10e02c742488c2a0ddb5d0e604a : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/932a08cd9da746eb9944dc21bcb6eaf6 : SUCCESS in 6m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/25c389c8eb7147319dfca2db91e84001 : SUCCESS in 8m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c3e80512b27543a3ae361134d496bd7f : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86c41c4ac8bc4f7b9eb9385e6adaef38 : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3a267577d49841f9811858ca7f95867c : SUCCESS in 8m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d6b2148f8e124498b38b1f7df23c7c2b : SUCCESS in 9m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6b10fd97f9374b878f5ce272109174c9 : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ccd6f0225b3a46b1a493bd94bb083e97 : SUCCESS in 8m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/88cca33a4b2343f5b0767362ac4bfc6b : SUCCESS in 5m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/021a9aec4feb4209b85bc88d50cfb4a5 : SUCCESS in 21m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/422e534108c74869a64659b9f8d89d49 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a105e6df73f4e40afcd7621174fd36b : SUCCESS in 18m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/43bc1eb9c29048f4aca19c7d0f523911 : SUCCESS in 7m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6138a6d0197842d39741fd5b8e3fe576 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/489377bcc2704d5a933676f95ffd45d7 : SUCCESS in 7m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8c2dd1f6303943f890662d4735a627db : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c20edf76be484f278e372484b4c05dda : SUCCESS in 54m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d534c835363247b281682892e22d4fd4 : SUCCESS in 1h 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"36fc26581502b6ac54400a651520baaedf6b3b3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 11:36:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3035639adead58d0dc934b22e4fbcfeb16428ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 11:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c97114e615453a92ee3524e09ad117\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28b1c7cb9244b218dd30893e47b86a9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3074faf79886f5acfcbebbed1272c1aeb43b0bc","revisions":{"d3074faf79886f5acfcbebbed1272c1aeb43b0bc":{"kind":"REWORK","_number":1,"created":"2022-11-04 17:32:11.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/76/863676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/863676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/863676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/863676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/863676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/863676/1"}}},"commit":{"parents":[{"commit":"00e8f93f7558babfb2460f6f58c2bd44b71109fa","subject":"Deprecate openstack-ansible rsyslog roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00e8f93f7558babfb2460f6f58c2bd44b71109fa"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:26:53.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:26:53.000000000","tz":-300},"subject":"Pin npm job to focal","message":"Pin npm job to focal\n\nfirefox on jammy looks to be broken.  Wheil we figureout how to fix it\nwe have 3 options:\n\n1) Do nothing and be unable to merge anything ;\n2) Make the npm job non-voting ; or\n3) Switch back to focal\n\nClearly the change goes for option 3\n\nChange-Id: I211333a6da03282feba82cbcdf9309f1a17ee7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3074faf79886f5acfcbebbed1272c1aeb43b0bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3074faf79886f5acfcbebbed1272c1aeb43b0bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
