)]}'
{"id":"openstack%2Frequirements~863679","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Icb45b4ca0ce04046c26c33ca9d958246b323ad4c","project":"openstack/requirements","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-07 12:07:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb45b4ca0ce04046c26c33ca9d958246b323ad4c","subject":"Update Victoria branch to run npm jobs on focal","status":"MERGED","created":"2022-11-04 17:41:20.000000000","updated":"2022-11-07 12:07:37.000000000","submitted":"2022-11-07 12:07:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863679","meta_rev_id":"97e590caf855a0afd697e67870eca03a4049b064","_number":863679,"virtual_id_number":863679,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 12:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-06 00:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-07 10:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-11-07 10:27:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:21:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 20:16:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-11-05 17:52:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b3a870cf8aa10a129bc2079a6ce8020247e5afc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 17:41:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc8a0a68681e814ad0ff15abac30c32ba4156ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:21:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd455736dd7d4a0399df969892285fa2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23fb99798ac14d0aa68729e77a465f72 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a670ecc1092c4c2bbd2ec4520f9c4f21 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff10a80767774f47a86b34fe8dc3fc9d : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6dce321088d74638aac13faeaf01031c : SUCCESS in 3m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3166957103d146ac89ed65d4db4b02da : SUCCESS in 8m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/573f8794992d409daa075f21b9096d34 : SUCCESS in 4m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0e8e6cd239084717940668b843fb5a7c : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56e621906de04b8b84340288403e1969 : SUCCESS in 6m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3479785bb28a46e58ca091f7fb7e7cd6 : SUCCESS in 6m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5d4373b3823e448599540aaa9172862b : SUCCESS in 15m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8664819b71e64344b4146309649d0791 : SUCCESS in 21m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37d4b9bea56946f88c7e497c12414993 : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bdc95f7fca8b41e8911e104a440de5a7 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/045e5ba2e6ed4cfc846dd053e2fe3cf9 : SUCCESS in 14m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/80bc90fa86f3437abd37e083fedaafd5 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/875f516570454eec98110c7cd20f1286 : SUCCESS in 5m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5def3fc819c04c8ca50ee574d2bb2ecc : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/717b97f8708b4956a7419efeb0219374 : SUCCESS in 55m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c3dd4b1e4764217a256f3a04e0cee6e : SUCCESS in 1h 34m 19s","accounts_in_message":[],"_revision_number":1},{"id":"724be6d2bfdbaaee6013e681401bf8c8a7779c22","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-04 20:16:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed64cad40d5e48f094ddf15ee56187dac63cd7bb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-05 17:52:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3caa4707c64ced603c6707fa62bb340ca65b794d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-05 17:59:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88119ae4edc144e11c45863135276f57f7f94224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 19:17:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b99fa4fc9114c9c9e2061d03b013634\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05d90ec57a04ee3953261343fcfd0b0 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/075cfb0528bf4884af38aa4d7908256e : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c13e2f386d8841ddabf06800ae969f37 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e06bd3fd67348aaa3f25659f7e8de5e : SUCCESS in 3m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd2e8ac1c6e9416ca372d1b4e8ef4514 : SUCCESS in 6m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8384a541f4a446a8b3ad39d15c065d50 : SUCCESS in 4m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ebce0a47757e4f16af5a0aa2a7e3290c : SUCCESS in 7m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4502a664375840cfa5765fe62376ac06 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/abe21808dbfe42e3b58392d557edba5c : SUCCESS in 5m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a05661591a2a4308a37e9600840c233b : SUCCESS in 11m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d18594b05dc54efd9acfc91ae842c300 : SUCCESS in 21m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/326a18f65f1d4e2abe081bb68213783b : SUCCESS in 11m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7b60e842a8944318b6daf8093d96d14 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac9c5c6cb2f34c019d1fdf7a5320da04 : SUCCESS in 16m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a383a84d80d54faabe5cc56e9987bcdd : SUCCESS in 4m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/238344d0f05b4b44aac06ccb59efbbf6 : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/205ce42746924d658b030b6dd694e7e9 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d052cedd48741cab9bb63e6587d4ec8 : SUCCESS in 42m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72c7e0b24c5f4bb09e860a129458d59f : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":2},{"id":"eea13a8d09c80695c3ec249c2749741c5de55c15","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-06 00:00:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"27569ffff1f7de08095f249f517b761bc7c6f366","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-07 10:27:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d232f34c3c5aaf0bbf7a3fd64bd1d5d2a6ede335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:28:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d137538c6290013e4dbc36c43853084b839003c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 12:07:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7320171967ef4334bc9be8011c7a290d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5025aadfe7c47f7a1a4995ef7ef8fbe : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc4fcdf0f7184eefabe371a9a3fb57e9 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/253f77b8ee55446b8ef479ad4645c4c5 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a015516adbdb432d8590ec75057a181b : SUCCESS in 3m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/631071f678a4404fae375ae415b35eca : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/06bfa37d1f5d4f7ebedd8aa5285cdc3b : SUCCESS in 4m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/65794ee81f9d4d388fa3d0f47d3edc9b : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38eb6cc49a824cc8be2525d0e6b627d6 : SUCCESS in 7m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f5081cab730a4503b1db545f47a8b3d0 : SUCCESS in 6m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/27279dee60d04890be5b86e1ce2d1634 : SUCCESS in 12m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7202e4749f3a4778998688e133783387 : SUCCESS in 22m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/837567f6f1a541dda675a47b01d451e4 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0f975c8947d4c079546a7d97e3ae258 : SUCCESS in 17m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ebc56c8f407d448cbda1b271b77a79db : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a61bd358e566482eb4d782eb127fc75c : SUCCESS in 5m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4dbbc42f02dc43979fdce8c9d9b67776 : SUCCESS in 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29a14b558877415181cf42f846b7fad2 : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b71137bb8aca40b3bf5f90f2f48bc56d : SUCCESS in 1h 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"97e590caf855a0afd697e67870eca03a4049b064","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 12:07:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2e710d8214fb0a6a2165c5f0f721514a083008c","revisions":{"d5227063143b2b699dae84f3e313ef1a7a16587a":{"kind":"REWORK","_number":1,"created":"2022-11-04 17:41:20.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/79/863679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/863679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/863679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/863679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/863679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/863679/1"}}},"commit":{"parents":[{"commit":"36716b655d2da2477469f5812ccdc9d6145f127f","subject":"Add upper limit for oslo.utils in stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36716b655d2da2477469f5812ccdc9d6145f127f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:39:49.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:39:49.000000000","tz":-300},"subject":"Update Victoria branch to run npm jobs on bionic","message":"Update Victoria branch to run npm jobs on bionic\n\nCurrently they\u0027re running on jammy which isnt a supported runtime[1].\nI *think* it should be focal but [2] points at bionic so I\u0027m following\nsuit in this change\n\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://opendev.org/openstack/requirements/src/branch/stable/victoria/.zuul.d/cross-jobs.yaml#L16\n\nChange-Id: Icb45b4ca0ce04046c26c33ca9d958246b323ad4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5227063143b2b699dae84f3e313ef1a7a16587a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5227063143b2b699dae84f3e313ef1a7a16587a"}]},"branch":"refs/heads/stable/victoria"},"e2e710d8214fb0a6a2165c5f0f721514a083008c":{"kind":"REWORK","_number":2,"created":"2022-11-05 17:59:00.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/79/863679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/863679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/863679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/863679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/863679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/863679/2"}}},"commit":{"parents":[{"commit":"36716b655d2da2477469f5812ccdc9d6145f127f","subject":"Add upper limit for oslo.utils in stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36716b655d2da2477469f5812ccdc9d6145f127f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:39:49.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-05 17:58:42.000000000","tz":-300},"subject":"Update Victoria branch to run npm jobs on focal","message":"Update Victoria branch to run npm jobs on focal\n\nCurrently they\u0027re running on jammy which isnt a supported runtime[1].\n\nNote: this differs from the python testing which is on bionic[2].\n\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://opendev.org/openstack/requirements/src/branch/stable/victoria/.zuul.d/cross-jobs.yaml#L16\n\nChange-Id: Icb45b4ca0ce04046c26c33ca9d958246b323ad4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2e710d8214fb0a6a2165c5f0f721514a083008c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2e710d8214fb0a6a2165c5f0f721514a083008c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
