)]}'
{"id":"openstack%2Frequirements~863680","triplet_id":"openstack%2Frequirements~stable%2Fxena~Icb45b4ca0ce04046c26c33ca9d958246b323ad4c","project":"openstack/requirements","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-05 19:16:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb45b4ca0ce04046c26c33ca9d958246b323ad4c","subject":"Update Xena branch to run npm jobs on focal","status":"MERGED","created":"2022-11-04 17:44:25.000000000","updated":"2022-11-05 19:16:49.000000000","submitted":"2022-11-05 19:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863680","meta_rev_id":"6f18f1ccf87c0a28ed0e0525b7be772709ec8cc3","_number":863680,"virtual_id_number":863680,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-05 19:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-04 20:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2022-11-04 18:09:34.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-11-05 17:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2022-11-05 17:44:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 20:16:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-11-05 17:44:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d1d4c9bc0680fdad8083014d9f8928f4e5897e35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 17:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98480154bfaf17fe4df4e0eccc7d930f7e6960a9","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 18:08:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a24e23707d29f78c849ca1c360c8f6c991bb59fb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 18:09:34.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c67cf366a0174f4a4bc18efa4d62d313837249e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:13:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d006d5753a24b40a31de511aa73a225\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7b85febaa5471a98869fbb0e923b95 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a12e3b94551a41fbbdb18ea2ef987d13 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514b5cfeb54b4da6bcbbb2119d9ce213 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a31fc8b04c3841738924e3409a59ded1 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09bbb56593b44dc69dd6ed44e0ff6524 : SUCCESS in 3m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e30e27c4abfd481eaf23c054a3f9013d : SUCCESS in 5m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5b05e072b0384789a320cf06e577cb6a : SUCCESS in 11m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5947c02e2f9f438f8db9ca5218ad7ead : SUCCESS in 4m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5d14ea08d83949e3b3550b730ba2d2a6 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e3ed375c26e45da8eeb287e78927733 : SUCCESS in 7m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a59ad87eedd74085bf23500c462a503d : SUCCESS in 7m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/454c0ec07fae481fa8d0be3b3aa0cfe2 : SUCCESS in 15m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/da53383de1274f5689804cb7ee54d47c : SUCCESS in 8m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d5678cf7395446d6aac2dd263aadd05d : SUCCESS in 4m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cc7242aac4a148c999e6f06b6edd563a : SUCCESS in 20m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c8401a96610046d5b9eb9a6117cf6847 : SUCCESS in 10m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/aade3d15fbd941e292907eacc982d582 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d490ef64f1243dd839b1ce673f8c61d : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90b4ca84307d4f95afcda5dc2188122f : SUCCESS in 20m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7483955c908f4d78a909ff9c2430dcf2 : SUCCESS in 5m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d1d8ebded95f4fc9ae287991ffc38a07 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35f0fc4e1b5c44fd9a457b80ee11e8af : SUCCESS in 5m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b72b2eef50ed4b368f2daafb7610cb34 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee33f2ae114e4bbc85b1ec33253a98ef : SUCCESS in 54m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/481f3111f24043cc84b39852bfbe65c6 : SUCCESS in 1h 27m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3af77330b7a4c1db265d260e297813c27583773a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-04 20:16:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81f9497f8d0bb6b649f1b01fedd0443b64d494e4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-05 17:44:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ec46681272305165fe466a418ba5268afe8757a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 17:45:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7c029694b4d4ffc3049223b78e9432661c9df10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 19:16:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8714d3ea6a134fd28eb790ef7ab1a094\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dea541f6cf034aab8dd51c091de2cb11 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0138d9514f6940d782b0d8a0d434acc6 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83f9d470962b43e780a35cbfa7b03ac5 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a17a076d7264fd6b13632be53d35689 : SUCCESS in 2m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f0dd7837811b4a398734b6d6388b453c : SUCCESS in 9m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/75603ef98e78412db6275d4f2ddf0ab9 : SUCCESS in 4m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ad44a4fcc4e84838a67f5aab7a873f23 : SUCCESS in 7m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2697a5b95864a6ab91194b2b69d98ca : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/25eb9f7e60574d0298531cd3b7ef528c : SUCCESS in 6m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c0d95e8d2e9e42a786ac4385d9a71dc6 : SUCCESS in 14m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eff82d57dc7d46999ef8e5ed0b2e56c6 : SUCCESS in 9m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cf00007e79834d9f84f3b995c444e169 : SUCCESS in 4m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bab13ffee65b42dea2ea09d6faa902a2 : SUCCESS in 19m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/782ceeb783b64275bb994955f1418f64 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43243de750d04f18b60bde53d9d5fc9a : SUCCESS in 19m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3d510a25770f4c738477ceb93941a04a : SUCCESS in 6m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3fd60cf99b864a3fa25ffda82fcee691 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61aa7cd06e1f4d9dbdb9a02bb8be7251 : SUCCESS in 5m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cb6cf5f95b274e159c7dd9d3b752c4f2 : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d14f92336bfb4b1db05dde2350e09328 : SUCCESS in 39m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9d73fc22e3f4a8892986d93afd48f29 : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6f18f1ccf87c0a28ed0e0525b7be772709ec8cc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 19:16:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4de99c4c53887aa014675ac3bd8092d7feb10321","revisions":{"4de99c4c53887aa014675ac3bd8092d7feb10321":{"kind":"REWORK","_number":1,"created":"2022-11-04 17:44:25.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/80/863680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/863680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/863680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/863680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/863680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/863680/1"}}},"commit":{"parents":[{"commit":"91d9760ee1849b0c17a8e59a1953408c22027437","subject":"update constraint for stevedore to new release 3.4.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91d9760ee1849b0c17a8e59a1953408c22027437"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:43:32.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:43:32.000000000","tz":-300},"subject":"Update Xena branch to run npm jobs on focal","message":"Update Xena branch to run npm jobs on focal\n\nCurrently they\u0027re running on jammy which isnt a supported runtime[1].\n\n[1] https://governance.openstack.org/tc/reference/runtimes/xena.html\n\nChange-Id: Icb45b4ca0ce04046c26c33ca9d958246b323ad4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4de99c4c53887aa014675ac3bd8092d7feb10321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4de99c4c53887aa014675ac3bd8092d7feb10321"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
