)]}'
{"id":"openstack%2Frequirements~863681","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Icb45b4ca0ce04046c26c33ca9d958246b323ad4c","project":"openstack/requirements","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-05 20:48:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb45b4ca0ce04046c26c33ca9d958246b323ad4c","subject":"Update Yoga branch to run npm jobs on focal","status":"MERGED","created":"2022-11-04 17:45:40.000000000","updated":"2022-11-05 20:48:53.000000000","submitted":"2022-11-05 20:48:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863681","meta_rev_id":"9dde84e86d7b64810e387c87b3827d09c8d632c2","_number":863681,"virtual_id_number":863681,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-05 20:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-05 18:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-05 17:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-05 18:58:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-05 17:44:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-11-05 18:58:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"349d82fc9687c1f7c74405a378b697048ddc898a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 17:45:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"047084899d6047fdd54081748cb429d83982afae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:52:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9282a72981a4a9596ba533ebe07c93e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f017d0d9aed407396f39fa52943f6fa : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7023659a0ed41b7ae1d48de5432a58e : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01e8f94bd03c4dd68f03271df6ffd5ce : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/227d43471801492d930e975db05e97bb : SUCCESS in 3m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b4527ea0753841df9b2e7db39f6f4c6d : SUCCESS in 5m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/50808e254f044cca9d43cfaf20699d06 : SUCCESS in 9m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8378bf59c67b4e2286850b6891ebbdf7 : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bc16b50cba18414dba01ef91002a0baa : SUCCESS in 4m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5983ae3752a544158969b4c93e418354 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef84c30af943456f93fb581136a7d4d9 : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/15abae2dcf88411e97b11275205fec43 : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/76df005549504e808d47443bf729301b : SUCCESS in 13m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/67ffcd0650374b91b33559abcf354ef5 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1911f17b125f421b805999707f5cbc8e : FAILURE in 18m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e9db724bab644287aeeb8e7539c634f3 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/91175ac2314e41168c1e20bcc57f631e : SUCCESS in 19m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/10241af56c1844d39ee78698067aa637 : SUCCESS in 9m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/836215ae4c8c47cab2959144dd3f69f4 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ef980316bfe54b278a9a79c8f22ef40c : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57c0ff00821b4d489b33e6cd20fbf3cd : SUCCESS in 17m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b1d8a8bc1baa4fe9a38bef11553dceda : SUCCESS in 6m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2764644b120a460bb415b692d210bf82 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/860d8ed1f5674894b03b7f27b6382f16 : SUCCESS in 5m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e16863f19e7e42bba94c250e813eb05b : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57208d31ffa64483b7c78fc9f545cca6 : SUCCESS in 47m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d63a38f906654e5da984be1ff54471a6 : TIMED_OUT in 2h 04m 51s","accounts_in_message":[],"_revision_number":1},{"id":"be812f92a2f068822b5c382bbb5a85b62e853c52","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-05 17:35:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1654222449d9741c89e86cb75db64cecd1838537","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-05 17:44:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b7caffa8476961ffedf496af87a18ac1d4cac134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 18:51:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0010a5141295423796d98a46fa781a71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea018d39225e448b9d24f2cbbbab5816 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f224f5d2b7d7461a8276bc097908fca1 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/792b4f265fa54b0e9c5d0cc16924b05c : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5de3da5e072a4954ba8145d043f565c2 : SUCCESS in 3m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c1ca2859cf284ffc8374dccd01a1fbe1 : SUCCESS in 4m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/87a6e40f5a6d415ca58f12426cce6d07 : SUCCESS in 10m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e2e24917a3cf4d83a6fa8812d63667a7 : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1c5271dc2f6c4f19aa82a5e56b1e223d : SUCCESS in 5m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/297d36a010f74624b371923113a361e0 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69dee30eeca847dd8731b868cdcd5272 : SUCCESS in 7m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/78e8246de7df4e309c133c997f680c71 : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d957b8cfe624475489329f8fbc88a939 : SUCCESS in 11m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fdf7f578f8834757b592af853d3cd283 : SUCCESS in 3m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/379f618908c142c2b33e0436c7ccb83d : SUCCESS in 9m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/909a99b64acd4c0ea51e625180542f94 : SUCCESS in 4m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9e599476bdea456fa0ff95712129105b : SUCCESS in 24m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/65383ba0b30345d39b9e3d0b6a149274 : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fc96d8803d0c4204893c3c54c45e7412 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b14486ef84cc4d0cb8e0746328eb0cf5 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/abc98032f72e4a32b7218d426d008220 : SUCCESS in 22m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2f40363e3bd74a1bb7c1d560cdc21013 : SUCCESS in 5m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/35c37e2c631f47dd87627cf859edd9d3 : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4dc5a4096e04045b51cf0eb01a8cb75 : SUCCESS in 6m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/09653502fc1544ada81c2dfb924129c7 : SUCCESS in 9m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22710c8b24fa48acb01fe9e04ad8c981 : SUCCESS in 55m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd084be1b03a47b387f95c54b27d1a1b : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7b6b910a8fc2e7cd897cc1b2af5a82384f1b4435","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-05 18:58:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ddbf33a16a949e6962e82e90babf642d9a6178c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 18:59:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b27fa4776a0d862d201854e92481235e7556169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 20:48:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3289641fed724387a299b17cf247e13a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e67c7482f324ad0891ba2d0db14c1c9 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64bb6ec695904f028dbd05a18435ff49 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/024c18eb6ec348769ecc92c146be6381 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99f1330d656847f69a7a90a706cd636f : SUCCESS in 3m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3f38c2bbe3644646a276cf8da2be7a11 : SUCCESS in 11m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2df756417452443892a4ef3a9b7e4d78 : SUCCESS in 4m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/46152d51e24d42ee89664b6db1bc9dd7 : SUCCESS in 4m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3f7e27a233444594ac7f4eb5b9e32232 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f47b69f8b574f69a42a6d77ecee7648 : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6bccf752d7da40c2821ee08d9f7ca20f : SUCCESS in 9m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6c4d007f95d04b9a8ef2574310cecb23 : SUCCESS in 12m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fe138bdf5dca497db2bc65b2f5b29120 : SUCCESS in 3m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dd53a9ec052240e1b341e7e0a551a2d5 : SUCCESS in 9m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aaa6092eb53a4522b202c4da19f5cc35 : SUCCESS in 4m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b983be8c4ab847788f40409148460649 : SUCCESS in 25m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1281a4cb8f66453b878400a359888a6b : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/766fc051e6544d98916a2f0fbc3175a9 : SUCCESS in 20m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3c86758d155d4064bbe6fc87d7b2237d : SUCCESS in 5m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fe4dc4693bb4441b9800544ba53136ee : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18b7f0202e9f43798fc7dce28d086b56 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a6bf198808b241c69aa9003269cc2734 : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9753fd5d9e184e018aeb31f013c151f1 : SUCCESS in 58m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5f4b50f24494fef90c4bb0af306b7a4 : SUCCESS in 1h 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9dde84e86d7b64810e387c87b3827d09c8d632c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 20:48:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c02280a2018d1c8d02f989a8bdb3adc6ba11f009","revisions":{"c02280a2018d1c8d02f989a8bdb3adc6ba11f009":{"kind":"REWORK","_number":1,"created":"2022-11-04 17:45:40.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/81/863681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/863681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/863681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/863681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/863681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/863681/1"}}},"commit":{"parents":[{"commit":"9d58bffe63dfaa064d44ac59a1ea7dc37ddb0d2c","subject":"update constraint for stevedore to new release 3.5.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d58bffe63dfaa064d44ac59a1ea7dc37ddb0d2c"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:44:52.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:44:52.000000000","tz":-300},"subject":"Update Yoga branch to run npm jobs on focal","message":"Update Yoga branch to run npm jobs on focal\n\nCurrently they\u0027re running on jammy which isnt a supported runtime[1].\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: Icb45b4ca0ce04046c26c33ca9d958246b323ad4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c02280a2018d1c8d02f989a8bdb3adc6ba11f009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c02280a2018d1c8d02f989a8bdb3adc6ba11f009"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
