)]}'
{"id":"openstack%2Frequirements~863801","triplet_id":"openstack%2Frequirements~stable%2Fxena~I605717824159f878df252e4c6d9bdd7af2f113df","project":"openstack/requirements","branch":"stable/xena","topic":"drop-python-dev-from-bindep-stable/zed-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-08 17:30:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I605717824159f878df252e4c6d9bdd7af2f113df","subject":"Remove python-dev from bindep","status":"MERGED","created":"2022-11-08 13:35:21.000000000","updated":"2022-11-08 17:30:52.000000000","submitted":"2022-11-08 17:30:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863801-drop-python-dev-from-bindep-stable/zed-stable/yoga-stable/xena","cherry_pick_of_change":863800,"cherry_pick_of_patch_set":1,"meta_rev_id":"f8688fb5d8994b63dd725faa754cfd26808b89e4","_number":863801,"virtual_id_number":863801,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 17:30:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-08 16:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-08 16:00:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 13:53:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-08 15:48:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-08 16:00:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2356bd4cde448841f15e787aac21c36964901ecf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-08 13:35:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"481aef4ce355eb9413cebdaef4ca7902288b8cc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 13:53:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/441b83885e234ed597bf078751df8a1f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/61c88bddb2124c2fa3dd9503dce57571 : FAILURE in 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"708e849bb3c42534975a6249ed4d2893c6717919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 15:48:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/119754ecbd9a44e8ad5032f3d4e34d14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06c545d79dfb4052aa1b481d77e8d16c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1469d94758f54a818d29f213a50ca1e4 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58d803804f65439a9fcf380e3c18daca : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/970a972fd581424ba3ccd60373b82a89 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/123ca7d2029b4268a80f73bd1fd10a8e : SUCCESS in 3m 10s\n- requirements-tox-bindep https://zuul.opendev.org/t/openstack/build/33a1887a449d4c5280d6c23a4d65f5ee : SUCCESS in 2m 59s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90aff4eed44f45f58437729fd0763788 : SUCCESS in 1h 13m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b8d51ee5dc0429889b52cd6abfde1a9 : SUCCESS in 2h 00m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d05e7bfb55999073648285376ca7ad82bca98eb9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-08 16:00:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9de6235e9c1209e611e130b1e2b399161dd930f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 16:01:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fab1fbc25abd50ad33e6eb4ce4f061aac397f7c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 17:30:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06dbe1ecb9f941b48213e8a963c98613\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71eebc00b1c54ef5977655de92104d35 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed740b247486466388026e0df96d4384 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/832d66b09643479589f45a421ae46e8b : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/291cb79ebf714777842cbf4e0f545e07 : SUCCESS in 3m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34f018584ebd4130b97de7e85120e97a : SUCCESS in 35m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/666e1f00c5da436a93c58836962a1823 : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f8688fb5d8994b63dd725faa754cfd26808b89e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 17:30:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb75f9f0671351060c1cf64fd5af9b6783f6708d","revisions":{"fb75f9f0671351060c1cf64fd5af9b6783f6708d":{"kind":"REWORK","_number":1,"created":"2022-11-08 13:35:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/01/863801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/863801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/863801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/863801/1"}}},"commit":{"parents":[{"commit":"20800243e39ca3771d580a653e9d0edd63e936a3","subject":"update constraint for sushy to new release 3.12.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20800243e39ca3771d580a653e9d0edd63e936a3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:06:00.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2022-11-08 13:35:21.000000000","tz":0},"subject":"Remove python-dev from bindep","message":"Remove python-dev from bindep\n\nIt is no longer supported by jammy and lead us to the following errors with the announce-release job.\n\n```\nNo package matching \u0027python-dev\u0027 is available\n```\n\nChange-Id: I605717824159f878df252e4c6d9bdd7af2f113df\n(cherry picked from commit 90b48934f4f018cf6e04dcd42731731346422b03)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb75f9f0671351060c1cf64fd5af9b6783f6708d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb75f9f0671351060c1cf64fd5af9b6783f6708d"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
