)]}'
{"id":"openstack%2Frequirements~864065","triplet_id":"openstack%2Frequirements~stable%2Fxena~Id14e3c36f7da9e6e19aec04140e56094055a7891","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"Id14e3c36f7da9e6e19aec04140e56094055a7891","subject":"update constraint for metalsmith to new release 1.5.3","status":"MERGED","created":"2022-11-08 21:27:24.000000000","updated":"2022-11-10 04:04:59.000000000","submitted":"2022-11-10 04:04:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864065-new-release","meta_rev_id":"87cee2b839999b99207d018a0518d18e75b8bfc9","_number":864065,"virtual_id_number":864065,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-10 04:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-10 02:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-10 02:25:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 21:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-08 22:48:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-09 23:20:31.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2022-11-10 02:25:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"30dcd315ae08876a3c31c32f14ee173197b5d288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-08 21:27:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d5086f562060da9ac66f88212afdbd314fe85f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 21:51:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d3c15acb9264079a5e1fef72ed06c62\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ee46c5202a8a4c78b55f68ce52555b98 : FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4aaa715f64de3af122db57de55352a6dbe64a281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 22:48:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ccddb1472d748baa36514818472f8bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d6643c2c0343cda6971c848fc9f371 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42f2749080824a6cae1a36214b02e2c5 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99bd4dfcd1424fde9afdae276c6c771e : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c809e46a153d4ec89c2d6b7dfcdf1ed4 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8794f1838377459db5638052e3581de5 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9361a6048844453fab7cad4f75d2a71e : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c57ab6238b74dd08419486e2ec01f03 : SUCCESS in 10m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f230ce74d2045508599fda70d5c8066 : SUCCESS in 7m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cf6b2c49501f4a88878593144e5a9efd : SUCCESS in 4m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dd6eb316adb54b1ba82bae28a81701d0 : SUCCESS in 8m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/701f2b0752734e4a8dfe573182a0f6bf : SUCCESS in 4m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cc5b10eaccb84faa8330287606e7f119 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae3f4de11eec4325be1586e07ee479b9 : RETRY_LIMIT in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1897ec8c8bef4e90b3e3005b5708b56d : SUCCESS in 5m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9aa37e19cfb4476f835564374d4ba700 : SUCCESS in 13m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/206c2b0b0a3e49e4ae8593d57936cda7 : SUCCESS in 11m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c84ce8b90e3d4692a225c954f7503d42 : SUCCESS in 3m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/226e07e56d16414cab78458a75f0c3d9 : SUCCESS in 20m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3ba8f8b1eec74caa9381c9f817c3e0a7 : SUCCESS in 12m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0d2867e274584296a9ac1f30f52e9607 : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab876f383beb4fbea4960774df5baf23 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c693ea91ea74c41a22a67517179fc85 : SUCCESS in 19m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3823165404154401b36aad383b70175c : SUCCESS in 4m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6427bdceb5ca41eeb6537f80cf762da7 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38e35432461347f6a33be168c3b92cdf : SUCCESS in 5m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/01fd9766b0094a31b3051ec7e2aad14f : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f50a40fc53a40d28f9051a5eb6a58a5 : SUCCESS in 54m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1d410976d1945e7ab1e4774ca13fccc : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5ba967fc395146d4892f8d326e86ce7404089977","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-09 23:20:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3a7fb76c7e9b5392b0ab0a02083e8139b642d73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 23:34:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72f8f8b6af5b4790ac2657c6b1e909b7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/92b1960358694c4c8c8caf1dfe39b31c : FAILURE in 9m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2e6d70d4dfdd0e7a8a39867b0fb263cbf88f3cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 00:35:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3503fe47968341598299a4fe0e217c5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3502407bea5441c98c021a349b87424a : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cafce5d1ac6c4619843151efc1c528c2 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bdcc0f07cca462594fb070475a3d95f : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90fcc97f233341688572f878ba520005 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28f99b10cabe426581045ab08552b415 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1581040d107b41bd91a41ac740dc9c53 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e97bb6e159b45e1adefe3d6d0b0e641 : SUCCESS in 6m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c67df38201b3429db90e0d09e9d73570 : SUCCESS in 7m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6dba8c1daf3844619c3905c2dc3274b7 : SUCCESS in 4m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d1962bc53d3d4de48a4a5b7bb60e901e : SUCCESS in 8m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/56e4c0373fd84f0abe63860a71bd5533 : SUCCESS in 4m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/46695711aab046e692ad5a7109c8d378 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6e57f1ac462470d9efb9ec375a78a5c : SUCCESS in 29m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/13f4c0e1e8784190a75430eb9e81b5b4 : SUCCESS in 6m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a6667937f144433c811a704d158806a8 : SUCCESS in 18m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2b7cd70318c74b33aeb3a07664469d99 : SUCCESS in 9m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8b8fbdefd76146b19f52b652bc6dde8a : SUCCESS in 4m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2b7c0f6214d640bd90626b8fc280c65d : SUCCESS in 18m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f764c2b6a6934251967735e65660bd76 : SUCCESS in 10m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bcd334d5e8014a1fb5262655a1256f0f : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5db70f665ef4a128fbcff2ae266ac3d : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/461a8d50538c4e2ca079963a6443573b : SUCCESS in 17m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a1b1750598524671b491d204652e7880 : SUCCESS in 4m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2aea60a0f8a84941b8011844c84305dc : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f58679ac86044624a2b08a4daa1f74bb : SUCCESS in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fff512163f2c4ea39ddd094994d9e05c : SUCCESS in 8m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72d7afcdb864428eaf9daf2c263ee003 : SUCCESS in 45m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/068b24d323054316a6ad2926f069979c : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bfb1237cbbaaf16226632dd065b05f0a8b3f6a35","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-10 02:25:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8bb9ece82e627711096de44c56b04132def4c36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 02:26:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ca5aa2a37109dbee9973f4a6245542410f2514a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 04:04:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628b381478a343938a3e17c508c9c976\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba90b9e45f74690aa84c00306dd9155 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92268fac392f4fafa753e7bab88e411d : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efb8f715c9aa4e1489b091e2d047621f : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d0912b2e82c44d89c56701dc29c2fa8 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3459838065d4634a1573d05de356fe4 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/016a9c6aa3be45d3b9cf805ad934137e : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3540b232ddf14d068faf3893c734161d : SUCCESS in 8m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c6f8bdb601d3481c9dce577b226fd916 : SUCCESS in 11m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/94cd76ce36b34688b6133fccdb99acec : SUCCESS in 4m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ed5ff8e9cce940558406b66899a91bef : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5639efc3a0b428aa004a86f8f790149 : SUCCESS in 6m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e9d1074c72a24cceb22193dbdd767bf8 : SUCCESS in 9m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/85920cab303c422299d199e571950fd9 : SUCCESS in 15m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/83bc05503e2148b48f93bbe80b01a793 : SUCCESS in 11m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6fdc07c7909e4e61b3c4a241b1058b1d : SUCCESS in 4m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/991adf1382614dcc876cc2a86db9ef9d : SUCCESS in 20m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/36f23b9e53fb4359a51de170bfea3922 : SUCCESS in 11m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/83030286fa574b43ab9e538ea3a846aa : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b5f620949a941cea7690bd8ce7a00b7 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/098b4ba462cd4093b958be6e35a0c4b3 : SUCCESS in 16m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e462d59a6bcd4d21870a8b51308b6398 : SUCCESS in 5m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4760d000da114e5885eae003f3da95ea : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc4dde11a3e04cfb96cff631568ba8da : SUCCESS in 7m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4f77bba627134a2ab18d4913f5442410 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5af40eaeca1e47abb62bcc6b15ee5b21 : SUCCESS in 54m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03bec97de868462789ea772c286a1733 : SUCCESS in 1h 32m 25s","accounts_in_message":[],"_revision_number":1},{"id":"87cee2b839999b99207d018a0518d18e75b8bfc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 04:04:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43903b8757b8c1d45dcb97956be63581135845e0","revisions":{"43903b8757b8c1d45dcb97956be63581135845e0":{"kind":"REWORK","_number":1,"created":"2022-11-08 21:27:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/864065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/864065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/864065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/864065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/864065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/864065/1"}}},"commit":{"parents":[{"commit":"fb75f9f0671351060c1cf64fd5af9b6783f6708d","subject":"Remove python-dev from bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb75f9f0671351060c1cf64fd5af9b6783f6708d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-08 21:27:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-08 21:27:23.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.5.3","message":"update constraint for metalsmith to new release 1.5.3\n\nmeta: version: 1.5.3\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Commit: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Change-Id: I54b1c9429f1a007b977c4b6b6376e6c25cb52153\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id14e3c36f7da9e6e19aec04140e56094055a7891\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43903b8757b8c1d45dcb97956be63581135845e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43903b8757b8c1d45dcb97956be63581135845e0"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
