)]}'
{"id":"openstack%2Frequirements~864191","triplet_id":"openstack%2Frequirements~master~I01f040b33122a99d781cb19cb5a62ac724b4dae7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I01f040b33122a99d781cb19cb5a62ac724b4dae7","subject":"update constraint for stevedore to new release 4.1.1","status":"MERGED","created":"2022-11-10 09:15:43.000000000","updated":"2022-11-10 19:14:08.000000000","submitted":"2022-11-10 19:13:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864191-new-release","meta_rev_id":"1bb6c0873f274fba9dedb687e165667c27567bfc","_number":864191,"virtual_id_number":864191,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-10 19:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-10 16:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-10 09:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-10 16:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-10 09:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-10 09:31:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-10 10:55:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-10 16:51:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7719e47a84944c56d9602c63f8756b31e25d0fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-10 09:15:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"107fb5132bce32d05dff6380eb0ca5909fc9afca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 09:27:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0102fac78a2e44e7b2d89011e28d0ae8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3644477a07624865a547e2ebecbb7907 : FAILURE in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fe34010ff139a3225652cbb0f23a1a8ba62e9eaa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-10 09:31:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9aefc9e0ec344a0b4499e54615abbaac295a2862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 10:55:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76bd8b952dce431a8b8bc86c00315e5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1672186c5a45809fe59c56bf81c602 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c7ad9794924beca64516d362754912 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0c6c6909d1c4cf0aacf92c5ec40c2d3 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eea92f0568e74f099b57d66a3ee595e5 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43810942e73a43c4a537d1a8a6ce066e : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9529c3424fec4d62a377bc0d30f5bf56 : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30ed34b355be40858e738a0aa1f5ef21 : SUCCESS in 11m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5a961d57f7b24e8095864439f78a2640 : SUCCESS in 13m 55s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3b42183d9748485ca653fdc33a987073 : SUCCESS in 7m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e095dff0a7e642a583ef30f467784ddf : SUCCESS in 7m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f9ab6b9d013d463baca99a63e749a0ff : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/35b348be24614824ad1862ed1db137a4 : SUCCESS in 10m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ee5a3bb1198444c989d810485766fac4 : SUCCESS in 5m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/65bd28d21f1344ffb082c30891e94704 : SUCCESS in 6m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dfbeaed352044b82895a0432cf65ec52 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/549e77f3af894ea6bf6afcd51ef61393 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c5f62f9475b4deb86486b1c9924493e : SUCCESS in 8m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/32850ae7b43e4ed09739e2a01af72dd1 : SUCCESS in 9m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e4a39d9846ed41648c324cf8561175b3 : SUCCESS in 12m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dd472c76399f47daa7dfeecdcc61feb9 : SUCCESS in 5m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a1d121f0377b472f9eab590a0e2cc37b : SUCCESS in 9m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0587df26f38e49ca8ff16ef532cb4f0d : SUCCESS in 5m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3ef0b7651bb74a3aaf14d82dac03a619 : SUCCESS in 9m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ecac30a1aff142e28b747207b6431bbf : SUCCESS in 26m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/608fff2be61c4041b3e6b5d70d774352 : SUCCESS in 14m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ae8a2dc50205468cbc934632d4609ae2 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f355aef58a54175ab8d512f803b4d7f : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/161ff326cc5a415ba8881577c2aee2e2 : SUCCESS in 24m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6f66700bce4c44059c2d3bbb7c20dc4e : SUCCESS in 5m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a07c2f46999d4107b21f6489c3c4280f : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a224383f4b51475086c0dc8f1306c06a : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/53bef29a702747f4b16103560b89af81 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/876db47e3b034ad89673bef0e726da52 : SUCCESS in 48m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d899fa3775854cb684904d41683942bb : SUCCESS in 1h 33m 25s","accounts_in_message":[],"_revision_number":1},{"id":"97fad26f4ced9d047bcdbe77bdfdac6d04a9aa00","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-10 16:51:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2474c6ad82164f0ad73ba9dbc44592160c64d61a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 16:51:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"277c1a91933d7d37902ef1c42a46dde366e1bc13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 19:13:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20870e632e6496aa266c8b1bc34c2ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09f8a6369ab4abbbfb90de9f18a9731 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803af553921545579b383328a7a3cd76 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2605ef20df3441ae82ac1f480ddfa60a : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95f2996549274acf82c8ad179863543e : SUCCESS in 3m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08d4f8a0e6cb4fdfb3b93b1f534dfe74 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ec51968e0114c3cafc5ade30d1c83db : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e976d6d8156b45ec906bf47b381716a1 : SUCCESS in 8m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/73e9863f59694f2793b7d93429294424 : SUCCESS in 11m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/70f1d454211249b3b929782cae8b45ea : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6da356997bd44772acfc7790d5378248 : SUCCESS in 3m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d2eb1d5056d34a9bb4fee09124765dd8 : SUCCESS in 11m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/623831a7085a46c580d922170fd18591 : SUCCESS in 6m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0a29cf58749a453baf1756771e34941d : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3988e628e53a432292c1407e4bb901c9 : SUCCESS in 8m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ea446c9e5fb34d9abf3b5e14360f94d5 : SUCCESS in 6m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a7e793ab8f564d158a3831128a579455 : SUCCESS in 18m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ee9585ea639445f7aeb231dd2bc581c9 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/938fcd3e542449689a2b13876c35565c : SUCCESS in 9m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e1da99d4fe8941679a395fb762b4d0bf : SUCCESS in 5m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/14b5f71c4d304cc1a75d2576097a55e3 : SUCCESS in 22m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9f873b6a5dbc49a5a1451862c60d8d38 : SUCCESS in 16m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/38cd1b2f81ab47f48d982e0b99a2868d : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ebd52aac9d0f47fb86064387d14acfe6 : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0deae2b3bc014c34a687d0c0d7a7c256 : SUCCESS in 19m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9b7284167a604d87820750debeb3826e : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/953ca94442cf43438b02819f89cd0924 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f71a8e7d68a14e1bb4ba0933f0b87d52 : SUCCESS in 6m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/40e63bb6b37a4ed699d1183d9f655e74 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/615e48f38f4147ce98fda877a2d2971f : SUCCESS in 47m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcb925f085b344108f9622319bad0eca : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8304b1e28fb19a50839f0fbb9dea837c5bce0f1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 19:13:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1bb6c0873f274fba9dedb687e165667c27567bfc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 19:14:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87c62641b674c7f85fc8c1bd4c32d58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544dd556c1a2467db229c09e885a7201 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e30cda8126abcb9df7a49c5c428d461acee9ab29","revisions":{"e30cda8126abcb9df7a49c5c428d461acee9ab29":{"kind":"REWORK","_number":1,"created":"2022-11-10 09:15:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/864191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/864191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/864191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/864191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/864191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/864191/1"}}},"commit":{"parents":[{"commit":"fed9cbbfec37e26c28456c5e83110a208392be01","subject":"Merge \"Drop unbuildable package from global-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fed9cbbfec37e26c28456c5e83110a208392be01"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-10 09:15:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-10 09:15:41.000000000","tz":0},"subject":"update constraint for stevedore to new release 4.1.1","message":"update constraint for stevedore to new release 4.1.1\n\nmeta: version: 4.1.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4f113516e2180ff358f181b04d7b1ab6a8e7631b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I01f040b33122a99d781cb19cb5a62ac724b4dae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e30cda8126abcb9df7a49c5c428d461acee9ab29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e30cda8126abcb9df7a49c5c428d461acee9ab29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
