)]}'
{"id":"openstack%2Frequirements~864573","triplet_id":"openstack%2Frequirements~master~I1ed628b0ff6c98d21d36caf828edfbc7ef822363","project":"openstack/requirements","branch":"master","topic":"bug/1995514","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-14 02:06:09.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-04-14 00:44:58.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-03-28 11:28:09.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-04-14 02:06:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ed628b0ff6c98d21d36caf828edfbc7ef822363","subject":"Blocks oslo.log to version 5.0.1","status":"MERGED","created":"2022-11-15 17:06:42.000000000","updated":"2023-04-14 02:11:49.000000000","submitted":"2023-04-14 02:06:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864573-bug/1995514","meta_rev_id":"8c697f61fe76bf915d5b58c91b93b34baaea1bcf","_number":864573,"virtual_id_number":864573,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-14 02:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-14 00:44:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2023-03-28 11:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-14 00:44:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-15 17:14:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 17:32:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-17 07:40:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-11-20 04:06:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-03-27 09:17:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-03-28 11:28:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-04-13 05:40:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"42f9311fe5db02be42a24fae5c806ffaecbac26a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-15 17:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4609e9abfc120c8c2b4c9211d71f610ce63c137","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-15 17:07:10.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b2bf0010e3a91c3889a2906efa3b685853d099de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 17:14:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/599ab6a0d96941fea599e7800610c909\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ce54221a3fcb42c89ea315a901febee5 : FAILURE in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b44e6c8df10f0c9419ef08e07a4df2d8ee3e8320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 17:32:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fca246f8fff5484bbe41b30f6e02a3ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73dc73740c3d4780b3f1fc0de00643fb : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9af32326ad4ad58c8e3f4055d5e7f1 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7b18c450100477899865bda441adac2 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b9c4e7e25b048dbbe26f20739168e60 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab89c4ba186a432ebb7e2c520ac33b9c : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c877837c55234ab7b07a219f21ba987e : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/487d0f11ca9e483ea0a1a597198642db : FAILURE in 4m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b5659ee531744379ab3b619e7274272 : FAILURE in 5m 33s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/65953a13346c4a539018944ca2ba31be : FAILURE in 2m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fdaf856ca9724318878358692dc4ea5d : FAILURE in 3m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b8d96f193df44d76a06bc23e12b23eb7 : FAILURE in 2m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ab48c564b8134f248538ede5a192b4f7 : FAILURE in 5m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f2e54b95530c4a2b8ae314a6b7cbc116 : FAILURE in 2m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/326581994b0b466ea1be49939870ea75 : FAILURE in 3m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/937e55b56e474631802181b5b0a82e32 : FAILURE in 3m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fc6929b7cce5499285bfa53e49ee250e : FAILURE in 2m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61f0223076694ab1a8c4fe9f161a84bb : FAILURE in 6m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/37c46d70793446b48ae41b15a93dade6 : FAILURE in 3m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/aae0441802f5414ea8c4a602981d928f : FAILURE in 3m 39s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bfb4debe886146d2bf87f2c4a145a6fb : FAILURE in 3m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5e73b143cfb749e5a461a6bb09ad4515 : FAILURE in 3m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3f177039089f4221bf59a1d6e96f8af7 : FAILURE in 2m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/070556c024bc4806b1c3710143d431ed : FAILURE in 3m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/51729d1209e04bf5949a83808f34eefe : FAILURE in 3m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ed5fba377ce34b6fbc52b7eb38224104 : FAILURE in 3m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/961d15370fb24513b2cc48dd07e35701 : FAILURE in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/84cfccb25a7742b882e24df7662995f6 : FAILURE in 2m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbc748ee76824f65a60312edca2cc481 : FAILURE in 4m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/964d8e0c19e846c2928cbabba8104da1 : FAILURE in 4m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e7c828c2a4db49a292b07ca669a5367e : FAILURE in 2m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/365e2627d7ca47ecb4d6b155e5bf573e : FAILURE in 3m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5bcdf5ab176c466c90344c0f25be6ace : FAILURE in 3m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/629c3969fc7444ce925356b23c0c94c8 : POST_FAILURE in 10m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e114a6ddc1cd41ef8c340f4dd0898034 : FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4d63236ff3d122c850a5eb82687834027c441264","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-17 07:40:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8594e3b4ebde84e55de6d03f2aec5e37dc19edf5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-17 07:41:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b7d2baa436ec4253367173389e3de00b3115323","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 12:23:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0c83b1b269dd8016605a7983bdac71ac9924f794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 12:25:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"bd3eeeadb482ec96e72b0032e08e0bf103c86a41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 12:26:27.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9e8baee639cf1aed6002b3ea4aa78553e818a968","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 12:26:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"88bee59b9b855437c2cec5c76adea559cb5b085e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 14:19:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed55d50d9e184c839aa0a0715a0b2d1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7df07b1a1a484e824280ff720dc3e0 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00661867c95c4609aacfe9cad2dc9a92 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a8e950a6c934d4eb5b4210a833f1361 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26b28ec715524e7b937d0fe2992af63e : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d1704d161fd4a188c46782467a7876d : SUCCESS in 3m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deda002b10ea4a498bb489dd0d03e5c0 : SUCCESS in 57m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a96016a5fe0e4726adf2e3bc08b83ca2 : SUCCESS in 1h 44m 38s","accounts_in_message":[],"_revision_number":4},{"id":"bdb47dd0a3a504a54e4f17528033ebc50d720187","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-20 04:06:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"30d58e77263675417b27bf19da8742dcaba503ee","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-26 09:16:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"06e958b00992f07c6be56b32f3859624772da0e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-27 08:52:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"7115072f9fc5b0e208b8502cc765ca99f883c228","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-27 08:55:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"abf30db61ffdcab3cf52b872af02e4d2c7b8246f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-27 09:17:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"87283a3d16a672165b04a1a1e56c8b26b30cd144","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-27 09:24:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0a1733b35a22431fcdd2fc467ada7f6200c9b920","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-27 10:12:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"12a3faa87bc00ae2659b5b8e1712dc727004ab8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:44:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f92bb11ccb14669b5d52a636b1a7fff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11946d714674b84a8bbde585e22dc61 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30bac56bd6f4347bf21ae7bf45735e2 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9eee8aaa433473f94d645a0ff39fd4a : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7e50cfa33a34192a65c24c7788ad1ec : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90a2bb09edab463386264223d9a3060b : SUCCESS in 3m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deb86a1220794bf6ba9bee4f6bb88dbc : SUCCESS in 1h 00m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/621192a898fb4842935e9a5059a139c7 : SUCCESS in 1h 35m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b9644db17ef5e36c58d218d9cb24bd1d0f50c94d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-28 10:05:01.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cdeb67c7e0aa83da813a5f259c5e286d1d18afac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-28 10:05:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"4efef06ebe5752e10dc8ea7334c6663bd390c650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 11:24:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3786b6ce0cd2466caa75524e88cd8879\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2dc93140c8441c93dc7e75b903e017 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c629e633d24e1288daf70fb1a9d585 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab2eadca7bb1409e90f5dcd24f12939d : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7f98810085b4b9d9cfda0d134a2cbe7 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/477b7c7ba66241da869573bd646a2014 : SUCCESS in 4m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c783e41b38940ca936a5b5ac4b89df4 : SUCCESS in 48m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97fe28463a724879b93d1fed292dbe08 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":6},{"id":"84ebb570d95ffb09ede60c41f8a504098c0d604b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-28 11:28:09.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"403b31ce3b1e627e14e068a2c5ff7744eabc9243","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-12 16:54:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bf815c707188b7ee3f9cf8b0b45f7340944c605d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-04-14 00:44:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"daf860077fe79706caa516b93d10f463555377ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 00:45:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"35b1c1c0d91819bdbc9b321b512a81ebeaf358bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 02:06:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12317ad088df4dd4af9d48c2b6460ade\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbbb41d6760c49a9999c55e9b685bc93 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d90f6d16ca48549e90a4dda849dae2 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44f006118b9f4bc39afb0d16dd3a9108 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22e7164f823c40028fa252b84a5f267b : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16c89346e800416f93436bb372ba8ffc : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35041e25f02541d3bf9a7efc2c224024 : SUCCESS in 48m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64112e80e1104e35bae674bfe663fcf8 : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":6},{"id":"997c893ab42374955b4a34e28d3e78f77ae9eda1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 02:06:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8c697f61fe76bf915d5b58c91b93b34baaea1bcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 02:11:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fac5c182d8f4fd383649f06d1da69a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5251f314c26446be8a49baaee06ee763 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e6e6112fdd83947bbcdbcb0e8f8e8b7bb5aefad4","revisions":{"110ff2b58461ddc808b9aa85baa8ed0aa430ef52":{"kind":"REWORK","_number":1,"created":"2022-11-15 17:06:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/864573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/864573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/864573/1"}}},"commit":{"parents":[{"commit":"21eb5cf40a8dea05b2339aab0148e76a15dddeab","subject":"Merge \"Add upper limit for Sphinx and docutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21eb5cf40a8dea05b2339aab0148e76a15dddeab"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 11:00:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 11:00:00.000000000","tz":60},"subject":"Bump oslo.log to version 5.0.2","message":"Bump oslo.log to version 5.0.2\n\nThis patch also blocks version 5.0.1, as reported in [1].\n\nThis new version, without the corresponding oslo.privsep one, will\nbreak the Neutron CI, in particular the job\n\"neutron-functional-with-uwsgi\", that will be temporary disabled.\n\nNext steps:\n1) Push a path in oslo.privsep using this new oslo.log version [2],\n   fixing the problem reported.\n2) Release a new oslo.privsep version.\n3) Bump the oslo.privsep version in this project.\n4) Unblock the Neutron CI \"neutron-functional-with-uwsgi\" job.\n\n[1]https://bugs.launchpad.net/oslo.log/+bug/1995514\n[2]https://review.opendev.org/c/openstack/oslo.privsep/+/864254\n\nRelated-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I1ed628b0ff6c98d21d36caf828edfbc7ef822363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/110ff2b58461ddc808b9aa85baa8ed0aa430ef52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/110ff2b58461ddc808b9aa85baa8ed0aa430ef52"}]},"branch":"refs/heads/master"},"50f98bc6e5114e72c0a65004e2ab150d01dbb161":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-17 12:23:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/864573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/864573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/864573/2"}}},"commit":{"parents":[{"commit":"7750bb3ac9db433690d5dd1948fee653c92685bd","subject":"Merge \"update constraint for ovsdbapp to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7750bb3ac9db433690d5dd1948fee653c92685bd"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 11:00:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-11-17 12:23:35.000000000","tz":0},"subject":"Bump oslo.log to version 5.0.2","message":"Bump oslo.log to version 5.0.2\n\nThis patch also blocks version 5.0.1, as reported in [1].\n\nThis new version, without the corresponding oslo.privsep one, will\nbreak the Neutron CI, in particular the job\n\"neutron-functional-with-uwsgi\", that will be temporary disabled.\n\nNext steps:\n1) Push a path in oslo.privsep using this new oslo.log version [2],\n   fixing the problem reported.\n2) Release a new oslo.privsep version.\n3) Bump the oslo.privsep version in this project.\n4) Unblock the Neutron CI \"neutron-functional-with-uwsgi\" job.\n\n[1]https://bugs.launchpad.net/oslo.log/+bug/1995514\n[2]https://review.opendev.org/c/openstack/oslo.privsep/+/864254\n\nRelated-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I1ed628b0ff6c98d21d36caf828edfbc7ef822363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50f98bc6e5114e72c0a65004e2ab150d01dbb161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50f98bc6e5114e72c0a65004e2ab150d01dbb161"}]},"branch":"refs/heads/master","description":"Rebase"},"1332d4a4a0a004804e3d345992c9d8657352e441":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-17 12:25:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/864573/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/864573/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/864573/3"}}},"commit":{"parents":[{"commit":"45852ca3ecbf9ff11c8cda663b97d04e713f8ec9","subject":"update constraint for oslo.log to new release 5.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45852ca3ecbf9ff11c8cda663b97d04e713f8ec9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 11:00:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 19:27:58.000000000","tz":60},"subject":"Bump oslo.log to version 5.0.2","message":"Bump oslo.log to version 5.0.2\n\nThis patch also blocks version 5.0.1, as reported in [1].\n\nThis new version, without the corresponding oslo.privsep one, will\nbreak the Neutron CI, in particular the job\n\"neutron-functional-with-uwsgi\", that will be temporary disabled.\n\nNext steps:\n1) Push a path in oslo.privsep using this new oslo.log version [2],\n   fixing the problem reported.\n2) Release a new oslo.privsep version.\n3) Bump the oslo.privsep version in this project.\n4) Unblock the Neutron CI \"neutron-functional-with-uwsgi\" job.\n\n[1]https://bugs.launchpad.net/oslo.log/+bug/1995514\n[2]https://review.opendev.org/c/openstack/oslo.privsep/+/864254\n\nRelated-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I1ed628b0ff6c98d21d36caf828edfbc7ef822363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1332d4a4a0a004804e3d345992c9d8657352e441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1332d4a4a0a004804e3d345992c9d8657352e441"}]},"branch":"refs/heads/master"},"ffbcd5dff395c6d724fb8b1d4fbcf1f70b57089c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-11-17 12:26:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/864573/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/864573/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/864573/4"}}},"commit":{"parents":[{"commit":"45852ca3ecbf9ff11c8cda663b97d04e713f8ec9","subject":"update constraint for oslo.log to new release 5.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45852ca3ecbf9ff11c8cda663b97d04e713f8ec9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 11:00:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-11-17 12:26:27.000000000","tz":0},"subject":"Blocks oslo.log to version 5.0.1","message":"Blocks oslo.log to version 5.0.1\n\nThis patch blocks oslo.log version 5.0.1, as reported in [1].\n\noslo.log 5.0.2, without the corresponding oslo.privsep one, will\nbreak the Neutron CI, in particular the job\n\"neutron-functional-with-uwsgi\", that will be temporary disabled.\n\nNext steps:\n1) Push a path in oslo.privsep using this new oslo.log version [2],\n   fixing the problem reported.\n2) Release a new oslo.privsep version.\n3) Bump the oslo.privsep version in this project.\n4) Unblock the Neutron CI \"neutron-functional-with-uwsgi\" job.\n\n[1]https://bugs.launchpad.net/oslo.log/+bug/1995514\n[2]https://review.opendev.org/c/openstack/oslo.privsep/+/864254\n\nRelated-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I1ed628b0ff6c98d21d36caf828edfbc7ef822363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffbcd5dff395c6d724fb8b1d4fbcf1f70b57089c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffbcd5dff395c6d724fb8b1d4fbcf1f70b57089c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6d6a24c597e195b21a830b3eb8b4e21ccf78712e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-27 08:52:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/864573/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/864573/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/864573/5"}}},"commit":{"parents":[{"commit":"be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8","subject":"Merge \"Add requirements-tox-py310-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 11:00:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-27 08:52:39.000000000","tz":0},"subject":"Blocks oslo.log to version 5.0.1","message":"Blocks oslo.log to version 5.0.1\n\nThis patch blocks oslo.log version 5.0.1, as reported in [1].\n\noslo.log 5.0.2, without the corresponding oslo.privsep one, will\nbreak the Neutron CI, in particular the job\n\"neutron-functional-with-uwsgi\", that will be temporary disabled.\n\nNext steps:\n1) Push a path in oslo.privsep using this new oslo.log version [2],\n   fixing the problem reported.\n2) Release a new oslo.privsep version.\n3) Bump the oslo.privsep version in this project.\n4) Unblock the Neutron CI \"neutron-functional-with-uwsgi\" job.\n\n[1]https://bugs.launchpad.net/oslo.log/+bug/1995514\n[2]https://review.opendev.org/c/openstack/oslo.privsep/+/864254\n\nRelated-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I1ed628b0ff6c98d21d36caf828edfbc7ef822363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d6a24c597e195b21a830b3eb8b4e21ccf78712e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d6a24c597e195b21a830b3eb8b4e21ccf78712e"}]},"branch":"refs/heads/master","description":"Rebase"},"e6e6112fdd83947bbcdbcb0e8f8e8b7bb5aefad4":{"kind":"REWORK","_number":6,"created":"2023-03-28 10:05:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/864573/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/864573/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/864573/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/864573/6"}}},"commit":{"parents":[{"commit":"be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8","subject":"Merge \"Add requirements-tox-py310-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-12 11:00:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-28 10:05:40.000000000","tz":0},"subject":"Blocks oslo.log to version 5.0.1","message":"Blocks oslo.log to version 5.0.1\n\nThis patch blocks oslo.log version 5.0.1, as reported in [1].\n\noslo.log 5.0.2, without the corresponding oslo.privsep one, will\nbreak the Neutron CI, in particular the job\n\"neutron-functional-with-uwsgi\", that will be temporary disabled.\n\nNext steps:\n1) Push a path in oslo.privsep using this new oslo.log version [2],\n   fixing the problem reported.\n2) Release a new oslo.privsep version.\n3) Bump the oslo.privsep version in this project.\n4) Unblock the Neutron CI \"neutron-functional-with-uwsgi\" job.\n\n[1]https://bugs.launchpad.net/oslo.log/+bug/1995514\n[2]https://review.opendev.org/c/openstack/oslo.privsep/+/864254\n\nRelated-Bug: #1995514\nRelated-Bug: #1995091\nChange-Id: I1ed628b0ff6c98d21d36caf828edfbc7ef822363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6e6112fdd83947bbcdbcb0e8f8e8b7bb5aefad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6e6112fdd83947bbcdbcb0e8f8e8b7bb5aefad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
