)]}'
{"id":"openstack%2Frequirements~865027","triplet_id":"openstack%2Frequirements~master~I79eb211ca8516f817ccc5697d281233f78ef48f5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I79eb211ca8516f817ccc5697d281233f78ef48f5","subject":"update constraint for os-win to new release 5.8.0","status":"MERGED","created":"2022-11-18 14:37:05.000000000","updated":"2022-11-19 04:58:15.000000000","submitted":"2022-11-19 04:57:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865027-new-release","meta_rev_id":"752b398ed37e6e67577887ac26ac51bd4d470838","_number":865027,"virtual_id_number":865027,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-19 04:57:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-19 02:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-18 14:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-19 02:04:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-18 14:42:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-18 14:49:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-18 16:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-19 02:04:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"642016cdc49c9555bfcf9e6fd240ff6c21e5cec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-18 14:37:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"585fe10b6850b66a54713a0e26ca573ab869a422","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-18 14:42:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"287ad2915db7b88133c1a36e0e5ed22a9123b560","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 14:49:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/176c787ecb4b4f36bd50cde3896e646e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3c569be85112492a9b1107a8d8ff16a3 : FAILURE in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"909206d2855dabd949f55af0eb37396e56304bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 16:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3b73b6e65ad4d1f908943a98ae9d701\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1df9d2863e4a069cb4bcfaf6b8fac0 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261a706fe9b34b1398a904c23447d300 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ffe1ed0640746b0909c201eccbd9b1a : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6dc89c6c6bb74f82bf5bbe68cd08af91 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4417b761d817447c80dfbcf549fe7fa1 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/538188f8e3704d01b5c23089393cd3da : SUCCESS in 4m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/372e205f2c744eaab36985723cf3111e : SUCCESS in 10m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4156c39001464183be5264ec690dcaee : SUCCESS in 11m 52s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/dea9f7b7c15b4773a8ca7efeb00669fa : SUCCESS in 7m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2e5f2ad07c5c4e73b6aefa85addfc239 : SUCCESS in 5m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b3b88e8f7ffa4feca796af7107cf8c67 : SUCCESS in 7m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cec0d971feb94a5bb846fd1b5e3dea8e : SUCCESS in 11m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b58173e94cda448eb1979848a29a7ce0 : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2f22b0fc50d0460a9e35d9dc8ac77e23 : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/276321e554c0442c912cef4a4d9aab0d : SUCCESS in 8m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/916ce216a1e24a538501c13973ec758b : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c74cf4c7a9264fcdbf5287e4e875bf4a : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8e5e85513fc047dea8b9a803d1f65735 : SUCCESS in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1113f1421d7240fba5d7a4dd26a260e7 : SUCCESS in 11m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dd6f9a9e95554cb983e8966afba3c3c3 : SUCCESS in 4m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/36ca30492b86446bbe20d2a7f961a764 : SUCCESS in 8m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3b8d892ffc0c480e856a7da4c64c0200 : SUCCESS in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/46b9eb5e701e442ead5ff35a3b514503 : SUCCESS in 9m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f963a99bc70a4296beca2a3bf376ea53 : SUCCESS in 27m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/35aa302cd0f0456d886b283346046092 : FAILURE in 12m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/55491d056d2b4829ba1d04ca533edeeb : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/235e2c8f26c54124bddc06138051c6cb : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1b1b7a17c584dc5900b96852b6f5315 : SUCCESS in 19m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c7a28726dc2a41db82b40472354311bc : SUCCESS in 4m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/aa93db07c010484aab9a187e89520139 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/382d2c4400fa419dbc1b6a9f974ecc82 : SUCCESS in 6m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/44694f3a0cd84c52b382c3e59b8fc633 : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9686a2fde0d24dd08599fa727c47e4ec : SUCCESS in 57m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cbad7df791942e8bb2b7907568d9f94 : SUCCESS in 1h 39m 49s","accounts_in_message":[],"_revision_number":1},{"id":"eafa363fe135a276f004caeceea24e163017f794","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-19 02:04:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c0534fffea8b11050fb6c444ad50f94f0777fe5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 02:30:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc23af8fa2c4805b3b9ccfe1858d76e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77fcaa72f5bb4fa4a44265121841e0fe : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"87a8b25a1bdda88876aa5a9415ed25adb0c0c161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:43:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be022d6ba65418390268a8488507c44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a001beb1d3c4cbbb1dd62897f36e9bc : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d96879a38abc4da6ae0523ac1f301a1c : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2750253f640d478994ce42299c99d190 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19449959c0c04e1fa3939b401cd7d26d : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f896c1bc59e4cd79503ebfdb4d2b49e : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1be3ffd54e844f7b47a11a84c6cbb71 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3a89d733ad394484a0b21213dfbf395f : SUCCESS in 10m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/58bff5eeeee740a39b4b93ddc16ea1d4 : SUCCESS in 14m 30s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a04341fd99d74a8fa246144c734f3974 : SUCCESS in 8m 39s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1456c8e8a9e34886aa1b18f6bc89658b : SUCCESS in 6m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/89ba8b88769b4afd8ae4f59012431f08 : SUCCESS in 6m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a31a9bea3e704b7c92735a8e1413d997 : SUCCESS in 10m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/98d0a89a047a40cf9957c409ecb1e3ff : SUCCESS in 4m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b1caa0714ad64751beac6ea8c1ed02ec : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d1069bf21d9c4f7cbea21423941accc2 : SUCCESS in 6m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5024dc1db8794ee79a512596429c42c7 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1ed57c896a34deba956657b000e6f20 : SUCCESS in 7m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b4a13a304c9b4635887e844327773705 : SUCCESS in 8m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3806a2191a5a45a3821d89afc7632a2c : SUCCESS in 10m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/97b6fd1f79c64f50aad129fe48e79999 : SUCCESS in 4m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d25d7fdff42b4fb9a8559bf0a77e0a06 : SUCCESS in 7m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b96a88ed2ea1499cb43f957230b7ad9b : SUCCESS in 5m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/45a8337f867f441c8eede1129640673c : SUCCESS in 10m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dd669698565f45c8b324e1e962a07078 : SUCCESS in 22m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0218d899c7be4e41a2b224c698551860 : SUCCESS in 14m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4792adeb56014890abbc47fb1b3ea697 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0fcde37999e4886a716f6f0138b5847 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e01dfc987e5f4be7b55030e73a740439 : SUCCESS in 19m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/75f6fdeddbca498c920b9b7c748ff484 : SUCCESS in 7m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/978398b420484a938d7d3ea44e4b84a6 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3387d19e1b4c46a5ad692a3247a5e780 : SUCCESS in 7m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/059ac53817ff469ba2464858a40f9e10 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efa0810f51bb48aa9e9f3fbe01206742 : SUCCESS in 51m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81dc6ff6dd374c1ab84bbe9d02c62b6a : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d6fe1b7c3e294f8920cb30923891697812383827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:43:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcc70d030b73d4b56d5b2f7d3787808446b994a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 04:57:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540c40d8cf4d41c3a3882d3e18e1e8b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09deb262a6b4785a6a804d022198827 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634068727e384e12a664ddabc2e4df23 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c3f363198644f8885a066498fdc8e65 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39c801f0a80948babdfd4d5badd3b0d3 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43f501d38c914a558888c14d206ef363 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05e0f73409ee4115ac8324811b9002a1 : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/75d6442d85d345c182d79e26fea9e382 : SUCCESS in 9m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ef8f6cdb9aa44089bacd2a37ac2d4452 : SUCCESS in 12m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/67f03515227643d5be5f63eb14718982 : SUCCESS in 9m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bbc42d8e6fd2452889479368716e6810 : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a7b4a49dade34066b4c6fe592b3ec56a : SUCCESS in 5m 35s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/74fd9967dbbc408cbe7c8f00e2b0a783 : SUCCESS in 7m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d1ad4709d0b446308d0648586163a740 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/136dc0c8673f4a8da5b7b04c483397ad : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1ad0d21ecce241c884527099ebe402d3 : SUCCESS in 7m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7c5669d9c68a4734979ffc126a3faacb : SUCCESS in 10m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/90525e08c3be4c6ba306242f2d769338 : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bf5e09542d194a65a98ebd62926f9473 : SUCCESS in 8m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/63c12cbd7f7543cba2e6eaa1c33d03b3 : SUCCESS in 4m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f3b84b9704234870abcf798e7db6ae16 : SUCCESS in 19m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/caf0c76f77644452bebcd06022324df0 : SUCCESS in 10m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e7909527644b4ba6a127d34209f3b647 : SUCCESS in 7m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd723ab7ca534fb3b6a17ec050fff6fe : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28c900099219419a9a0777a0a4d54972 : SUCCESS in 17m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/902cb3a8db664d7491ea0f649899f686 : SUCCESS in 4m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f176666083974cb299a91d43b35e191b : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b4455ccf2594452ba00b64a64cddbaa : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/245f43bcb535447093767c2080a6578b : SUCCESS in 9m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d3ca3ea17c94171a13864baf8e727ff : SUCCESS in 38m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb9557c1827044da8244a645fc395a33 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e89ac380286ffb678dc2b353c04be25ffa666818","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 04:57:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"752b398ed37e6e67577887ac26ac51bd4d470838","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 04:58:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c165002142e42cdb1c180c166726af0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9e62f0744e4e939332c916e382aac0 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"578962f26f2fa999352596381fd8edda792583ca","revisions":{"578962f26f2fa999352596381fd8edda792583ca":{"kind":"REWORK","_number":1,"created":"2022-11-18 14:37:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/865027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/865027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/865027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/865027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/865027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/865027/1"}}},"commit":{"parents":[{"commit":"dfcdd70a71cef20c25865db6f16f35c69244e2cb","subject":"Merge \"update constraint for python-neutronclient to new release 8.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcdd70a71cef20c25865db6f16f35c69244e2cb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-18 14:37:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-18 14:37:04.000000000","tz":0},"subject":"update constraint for os-win to new release 5.8.0","message":"update constraint for os-win to new release 5.8.0\n\nmeta: version: 5.8.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I64d98572a6d0aeef8e169bf44595ec1fbb585a18\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I79eb211ca8516f817ccc5697d281233f78ef48f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/578962f26f2fa999352596381fd8edda792583ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/578962f26f2fa999352596381fd8edda792583ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
