)]}'
{"id":"openstack%2Frequirements~865028","triplet_id":"openstack%2Frequirements~master~Ia6af4f08b4d0d773a9057644221342949c2856e4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia6af4f08b4d0d773a9057644221342949c2856e4","subject":"update constraint for tosca-parser to new release 2.7.0","status":"MERGED","created":"2022-11-18 14:38:13.000000000","updated":"2022-11-19 03:36:04.000000000","submitted":"2022-11-19 03:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865028-new-release","meta_rev_id":"a4613cf455b0f33ec12ca72908ad4397ee8565f6","_number":865028,"virtual_id_number":865028,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-19 03:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-19 02:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-18 14:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-19 02:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-18 14:42:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-18 14:50:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-18 16:40:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-19 02:03:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9c84787bb676bbc9888065f388ec4a4b1dcc5255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-18 14:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1b7ce9b339bf471be84babf3f9f945b35db90f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-18 14:42:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8081f864538790c8d95df39adff9faae3f98c617","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 14:50:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1773c24f80a948b4885f55ec579fe39f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c017d2f0386441de8f70d11880c2447a : FAILURE in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"dcd709551eb09a791e1841ce306330f33a61e614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 16:40:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bc4fb25465b4e86a4834a30d9f7005f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1173e1bc70a8466980b126406732335b : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4219c6e5034a1bbf258e5a6edb4851 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/711e3dd1893a4759977a274cab9d382c : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a8e8e04ca704b1db7c7269bc03590d4 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1dac9412f72b4bf6bedfe7e176cfc092 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b308588a719449fabd9c90a8fe00c1d0 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9bb6d399d90d430ba583628bd9e7d682 : SUCCESS in 9m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/75663e2cb17c4ed2b4524b8b64b7d49b : SUCCESS in 11m 39s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d7e2321ef1eb4bb9acaeb0d2fdc30287 : SUCCESS in 7m 40s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d589ad7edb4a4554ac000f91548673d6 : SUCCESS in 4m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4c8abe1cb7a4485f953ffd41be6639c3 : SUCCESS in 5m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/13e6c9cd323d44fabf9b50890fa1bdd5 : SUCCESS in 10m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2fa351c3d37a4158a2ec190f8e46914f : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a875f833e6b14e32a6b11132a5cd0814 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f09394c0e8e94b5b97cbe64b02d7a9b8 : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b1097e4843ba42b8bb56b47d42b72f1c : SUCCESS in 7m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b92e1f3252234a648e245b59dca8bc66 : SUCCESS in 7m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7304122a28474e3c8786bdee55bd2e8f : SUCCESS in 7m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/799e9b9c00b145bdb50c02c95f60a40e : SUCCESS in 10m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b1079b5184474c5e82ba8cf107c4427c : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ae7e9daa73684d4c97e8c495e947068c : SUCCESS in 11m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7d52d590bf854a668a2f0d4a554a3cc1 : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f5451068723c4ac58f61894bc9c4f990 : SUCCESS in 8m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b405a4b5ba454e9d8936038637f0f68c : SUCCESS in 25m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/09e6195282ab423bbcdc0e9f72222b2e : SUCCESS in 13m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/128e638cec924590ba5ea0327124ce55 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1536b7089009437ca416ffbecbf1fc7b : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/547ec23974a743b5a58b97bdaf77b54e : SUCCESS in 23m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5acd584d8dcf4cffb7aeb4bdbe1c5a0d : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/57bfdb245d3a4b2f88b4b265a796388b : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b5e1a9b8e5f46e7a663f3223daf46b6 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9d5164713ddd4c4080ef85ac305c969a : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/139f7b5d7c464e7ba1615341593f8e79 : SUCCESS in 42m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d830393d7364a1097afff874fb1448b : SUCCESS in 1h 56m 21s","accounts_in_message":[],"_revision_number":1},{"id":"881bf6c682bd01e5fed72fa6efff645f082bd25f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-19 02:03:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b16a904cbf458ff8d4b58a924301eb2fda9b75c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 02:08:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40fb7442276bc2d53b9fd4d3aae4eb64d93ac322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:35:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d664fcbf8174c55a56f42dd2ccda3fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cebfaa0eb19477b883a29d3f13b5ffc : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d2fa4725f5459eacfa1479cfdf865c : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/073e378bda584b7ca6298805cc489fd4 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffd9aab213964dc69f4a80fface442e7 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9858e6fdf154790aee0d3915f02c46d : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92e7c48a25e44336887e70c0ce297e42 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/890e2d5da63b4b35ba743961683533c7 : SUCCESS in 8m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/59cff19f5d6a448dab2873383f174624 : SUCCESS in 12m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6b4be8ba12104737ac53007feab038c3 : SUCCESS in 11m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3a85a7b541e14dd1a74b4edcbf18b66a : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/46eb9a56c44b40c5b8c801715e5bf557 : SUCCESS in 6m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1abb26f8a7c04e8aa088fc1212842895 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bd5fc41697624732a4bc235ab4aaf1be : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/35d5af0c83c94300bdb90f31850dd8cf : SUCCESS in 8m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5e021ae6dfb84439b4d70ed4c66dedac : SUCCESS in 10m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2e50c2eafd6a4888abe12dd51e1a2f06 : SUCCESS in 14m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f011f321a84f4b769a9d6728e066ee94 : SUCCESS in 4m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ad2dd1ea46a046898a7942c541979a45 : SUCCESS in 12m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5e95e93732064f3a9774403c6e029c51 : SUCCESS in 6m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f709f1fe10c84acbb50e66639c3217ce : SUCCESS in 22m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/768f1378a9f443af97357e5c3b102c31 : SUCCESS in 12m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8d880fe7688e4764bc97cf26cfae11a0 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b425e3bba98648e4bb7c8797a06af81c : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aacc2b2fc48a48a5b4f2e272f4a05c80 : SUCCESS in 24m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b858894d41a24fb2a8ec69fe927055d9 : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4806ef5dbdf34f51829abacc2a7bd4c7 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/181a5cba68ca4315929b6993b38b1741 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/39c1a3bf76f64eba9db03e50eedab115 : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff924bc0dd334e739c2f3b210a1b67d2 : SUCCESS in 51m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f63dd24bef94486e94e4d3a27ac30fa4 : SUCCESS in 1h 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"75dc5c958c2ba74e9633cf2e678e00dfdc649ff2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:35:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4613cf455b0f33ec12ca72908ad4397ee8565f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:36:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e47ee82a36a48e3b7364e9dda05b427\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c27f8678fd14c6596ee21f5255e5d8b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb7fb802b346dd81dc8e9080cb54c785ce7ec91f","revisions":{"eb7fb802b346dd81dc8e9080cb54c785ce7ec91f":{"kind":"REWORK","_number":1,"created":"2022-11-18 14:38:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/865028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/865028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/865028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/865028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/865028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/865028/1"}}},"commit":{"parents":[{"commit":"dfcdd70a71cef20c25865db6f16f35c69244e2cb","subject":"Merge \"update constraint for python-neutronclient to new release 8.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcdd70a71cef20c25865db6f16f35c69244e2cb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-18 14:38:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-18 14:38:08.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.7.0","message":"update constraint for tosca-parser to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Id0132b4bc59464071c335ab1df79485496c1abc4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia6af4f08b4d0d773a9057644221342949c2856e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb7fb802b346dd81dc8e9080cb54c785ce7ec91f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb7fb802b346dd81dc8e9080cb54c785ce7ec91f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
