)]}'
{"id":"openstack%2Frequirements~865030","triplet_id":"openstack%2Frequirements~master~I835e90f98d91605b61ccc04cfc986727e1c724cd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I835e90f98d91605b61ccc04cfc986727e1c724cd","subject":"update constraint for cliff to new release 4.1.0","status":"MERGED","created":"2022-11-18 14:55:38.000000000","updated":"2022-11-19 03:39:51.000000000","submitted":"2022-11-19 03:38:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865030-new-release","meta_rev_id":"a379834f50f5e1acd4f39d53bb467d53b9bcecc5","_number":865030,"virtual_id_number":865030,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-19 03:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-19 02:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-18 15:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-19 02:03:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-18 15:06:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-18 15:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-18 16:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-19 02:03:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4de7691d427e6f9cda69eb7254c14f2f8d7f99af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-18 14:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91eb12d960bcef6bf98e968a638295ba54f0ebfc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-18 15:06:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8dff97d5c49d038654e71a453ace6f30321352e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 15:20:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf9becf8d944fb8a2f30a6cb2eca59f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7e4e64a1580341c3bc0bc2f455ba89cb : FAILURE in 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5fec383156d8eabbc77c1734758fbf0d093cbbd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 16:46:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7103c3012ff4d028aed916f0bb20063\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b15c503c4c42f39be5760b844f1d9f : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51840938ce3d4c338f7959d2e99b04b4 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f3f8b12c97d48d9aeaf6a85e59bbe58 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/298dc154a9604b64a1969345fe173e4e : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c05f28a9d1f434a9ffb3dfc97799753 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5d38d70b7cf4133bd20c7909495a7d0 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ddf6b4f8a96e49d285463774da0d41e2 : SUCCESS in 10m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e69bb736002a4fbfbc5070be302922dc : SUCCESS in 9m 44s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/78991d59276340c9919e3e6c18efa37f : SUCCESS in 8m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7a94260b1e054188ae2f32ce3c95a59d : SUCCESS in 5m 38s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/91357f67c02548bbbfe25d5ff8953018 : SUCCESS in 5m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/76984dcb278446179c30ba621ab54ec1 : SUCCESS in 10m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2644ace18b254f1bb7a6dca4ca9b2f41 : SUCCESS in 4m 36s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/029ef3e6aad04fb894b5736923c253e5 : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/90e9231afe21460a863c0de0e13f3eef : SUCCESS in 7m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/05863ff9a0334b19a33a5ab1b336caf9 : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9833b250c4544d94b1a618c1ee2024b3 : SUCCESS in 8m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/64462c54deaa4c458707c65c697c96d6 : SUCCESS in 7m 47s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dbc748fa426c4284917336a116c18042 : SUCCESS in 11m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0d4d7be8c74649bfa5014da2c94a793a : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/eb00e53c3e924dd5b0754fbbdf4e5972 : SUCCESS in 11m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/89d79d7255c141a78dda62f197e40373 : SUCCESS in 5m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0d1c559fa3f74145b0282b1cba75c83f : SUCCESS in 9m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3a4fcd565ed7449a8d5cb099e9eeaac4 : SUCCESS in 31m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/199f90b522344357864839339e0da8e8 : SUCCESS in 12m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d5f3cc0424d44493b2af121f1a89297a : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47d97e491cb54fa5a95f846310eb061c : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1f974ba2b3f47a1b5f989afaaaa5c3a : SUCCESS in 21m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/99c7a528f9584ba4935bd172c5eb2f53 : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e1684d889ee84266ad85befd465defc7 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b41ed5da180448183f1ddd06f7e4e89 : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a4333f7efcc143de8a7fac4f5d3787ee : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0063c72e5f0e424a864e1206e33b9fb0 : SUCCESS in 50m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5860aa7cc0c5416e8117bd2b30ce0be9 : SUCCESS in 1h 43m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0d91c1b1ba371a7f305060683ce9387ef262ec91","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-19 02:03:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ab2d4c636ac9146c8c39cc9659c307627f98511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 02:08:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6d1595d5b20a1b0a08fdba1df23489fd87e2f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:38:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5802096da9fb46a9ab3cb86cd522e3af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb25701954d4b86be12e4b5c54a518f : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd178c5fa2c4c75a028b34cfbf313b9 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee3df08582ee46a28f3fac1e2772bf07 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8376085fe8f44451b37c1f890e76aca7 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dade107670134346a265d5342a655254 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09b66b13ca1c4727a4eea5171b6becd0 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10a14a94244141cda7335268edcecf33 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0d94358cbbf64c10b6eca9f7f8b82bff : SUCCESS in 13m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9b2782f121e14cb8bbafd8e612279e11 : SUCCESS in 10m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0858f3f66f0843669d4285beb7783be4 : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6aae3267386d41f6b31b407e9b7eab41 : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3dd65f16ce05481d9e6cb6a9ea9a8c3f : SUCCESS in 8m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fce0e6dd495945e9a736d20881fc92e6 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7732b5b37f214d33a6f2304edd976f24 : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f3f52afdfa5f43b2868bb12489a8fb25 : SUCCESS in 8m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/596c03e7a1af419d8ab988e6e68315f5 : SUCCESS in 14m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/13d5494afc5545d3a495c026fd152c03 : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7c7d83e79db3477cb5554dda06f3995a : SUCCESS in 9m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2dfdf63c605e429a9c5b417bb91ec9f0 : SUCCESS in 4m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/62b98b1813d046d0b4fba45c8371de6f : SUCCESS in 26m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f07ac04a7f75495ab6046643694a40e4 : SUCCESS in 12m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c56aea93a42e46da859c0b5a9833ac9f : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/358059641cfb4ddbab06adcc20b9f22d : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4dc6aad79d3842a8b31082a81e7a61e5 : SUCCESS in 24m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/933a2e36d4ab47c585d40363c03085c4 : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4832c4a332b142748a0f6be6f4e8392b : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84c773d6b3764a259365fe6dd3297c44 : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f6b34598238d4c1281e9b29efa58f544 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1cde91469774886bcefc60b97ef40e7 : SUCCESS in 47m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cd52720309845b094bf2e4077ad468a : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"32ff345f80aa8a3df460d8e5d118cc6c0893e424","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:38:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a379834f50f5e1acd4f39d53bb467d53b9bcecc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 03:39:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92f30bb7d1e483fb449ea4e3af0ffba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f06f9309dc488c9c6931eb69d30f67 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbf8635fd13e724aed3170e6dd3de9af63c53af4","revisions":{"dbf8635fd13e724aed3170e6dd3de9af63c53af4":{"kind":"REWORK","_number":1,"created":"2022-11-18 14:55:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/865030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/865030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/865030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/865030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/865030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/865030/1"}}},"commit":{"parents":[{"commit":"dfcdd70a71cef20c25865db6f16f35c69244e2cb","subject":"Merge \"update constraint for python-neutronclient to new release 8.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcdd70a71cef20c25865db6f16f35c69244e2cb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-18 14:55:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-18 14:55:37.000000000","tz":0},"subject":"update constraint for cliff to new release 4.1.0","message":"update constraint for cliff to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ida5d1b174a0ed222c167f5fdd46cb7605a15f673\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I835e90f98d91605b61ccc04cfc986727e1c724cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbf8635fd13e724aed3170e6dd3de9af63c53af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbf8635fd13e724aed3170e6dd3de9af63c53af4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
