)]}'
{"id":"openstack%2Frequirements~865526","triplet_id":"openstack%2Frequirements~stable%2Fzed~I1562e0fd60a6340e26a77a2ab6102595dc7ee979","project":"openstack/requirements","branch":"stable/zed","topic":"zed/pyOpenSSL","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-11-24 16:19:47.000000000","reason":"removed on reply"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-03-21 17:41:27.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-11-09 17:17:50.000000000","reason":"Change was abandoned"},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2023-11-09 17:17:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1562e0fd60a6340e26a77a2ab6102595dc7ee979","subject":"Bump pyOpenSSL and cryptography upper constraint","status":"ABANDONED","created":"2022-11-24 10:53:02.000000000","updated":"2023-11-09 17:17:50.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ecc22977fb91c7df147d86e5ad1e0b00e67a3293","_number":865526,"virtual_id_number":865526,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-24 15:01:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"disliked":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2022-11-24 16:19:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-11-25 02:06:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":-1,"date":"2023-03-21 01:01:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":-1,"date":"2023-03-20 16:32:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-11-24 16:22:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 11:01:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-24 11:15:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-24 16:19:35.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-11-24 16:19:35.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-11-24 16:22:56.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-11-25 02:06:25.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2023-03-20 16:32:43.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"1bccb1f9040d2b8506e5d7bf21f7cc776165cd7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-24 10:53:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3e22ef058a7717f1e2fa5162295355ad76a6eaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 11:01:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/663d4a4f2f6f4e359abaa6cfd3f98c0b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/208b8aa158214245996aa6e4982b1b66 : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cf14fea7bd63ecf27314033c80f0888cdd555edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 11:15:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d1571b3acda4ab19e9e3af364d12902\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aab4a0b23c34dd293700a494125d0fe : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a708ced169c34e5da2e2dbed0d6a75e9 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/831a1da447ad4fb7bbc2f95f9769233f : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e2e09dc4db746c5ba84598fa77090eb : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb1bb4ba552a42d5ab129886054b595b : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4b9e97e85194789aad3c0af89dca00d : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fd0ead576704141b14b7089701a6b26 : FAILURE in 4m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e94253f52230424bbb992e2b6628769b : FAILURE in 5m 17s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d61631ec3cef4eb389b88cf3197e27b6 : SUCCESS in 7m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1f25caa42cb24172ac26884c54877063 : FAILURE in 2m 58s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2614a05c79c64e0d9a8925cab9e31112 : FAILURE in 2m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ff40ba7b55914763b6a08abee8de83fd : FAILURE in 6m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c9ef010467ec41e99ad2996b5d780f2f : FAILURE in 2m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1a1b557d40a8415eb16b878221cd63ec : FAILURE in 3m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/53a58b7ce1c84ebc844810d1d12bd836 : FAILURE in 3m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ffa1263b15f14f61b5a7a9e739a09ae7 : FAILURE in 2m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/413723ba66ad43a2a712fa0e5a437997 : FAILURE in 6m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a929c9f2c82440fa2f6d2ab6d64abc4 : FAILURE in 4m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/31158cfd31974dfc8a78c47670c53e6c : FAILURE in 4m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/900df7fbdf9d49fb8b3b87900f483c11 : SUCCESS in 3m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/59f2af0258b640bcbee2ce8d34a77eec : FAILURE in 3m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0a1a336e7a964b0ca8b096690f8a699f : SUCCESS in 4m 07s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/2ccf43c1e6284b6294d5f47c757c013a : SUCCESS in 9m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a756384c3915448ba66d197312150640 : FAILURE in 3m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3aea8627343842b2855652f044276edd : FAILURE in 4m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/961f8b11d9ca49868a0d034d1f01722d : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1556c22a68104eccbc020df4aa6f07af : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0aab4bd0e74548aaa84900a76d036c0e : FAILURE in 4m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d1a42d6edfc74a8a8c6725c0ccd1fe3e : SUCCESS in 4m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b1354d8633ef400482082643085ddba3 : FAILURE in 2m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f22ec89615d14d51a3151997a0b8ab23 : FAILURE in 2m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dbcd527ecf6146cc89bba2a3ac99f220 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0501ab0c1ef04f26896ff35ed600f21c : POST_FAILURE in 19m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01407bc9d6b24c7c913d1b8f4ee25ec1 : FAILURE in 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d2ef32fbcbc62cb62315cb5ec57d02386bcdc05b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-24 12:14:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76406cf064d38990379c4c71077a7352f3136cc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 12:29:15.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07b08eb6ceee4c0c92c83b9e63f69b0f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b44e5a824997492a83cbeb9cef37bff4 : FAILURE in 7m 28s","accounts_in_message":[],"_revision_number":2},{"id":"bd18b7669698a8383130199f4c1a0d9be43c719d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-24 12:52:29.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"35c7b14434bed168cae99cbc7e2b09b46b762fcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 13:08:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ff681dd76b4c9c8f438d95bcf24398\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f6b299f2a8b44e53a5bd86db310c7573 : FAILURE in 12m 30s","accounts_in_message":[],"_revision_number":3},{"id":"01d493e958e45dc2386ac9e868ac1a18bed49f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 15:01:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0007ced18bee413bb1fe1fdd235f981e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71cc19a5e81e41e7a58be24cc2ba7921 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/938bb4a0038d417f88dd4cce6340d8d9 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/168047180fc2456695a1869fed7e78be : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4ee55246b3a48cd9ad7b138c8e8bf10 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e9f57cb12704fe8b86bbb6d09197353 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91ac8b9e460d48bc953ac25ddc8f825d : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03236968f60b40b7b19d47a0b68e2864 : SUCCESS in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f8eb3371e182460c9ed83209ca5411ca : SUCCESS in 12m 57s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/75d1e18b20034b26a8e238d690c14cf2 : SUCCESS in 6m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e9fdee62cb2f426cac4c559c5f47826d : SUCCESS in 5m 06s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a5a521d7f22a461ba233bf51d3ace2d7 : SUCCESS in 5m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/56d489e4a7714011ab8f84fd883f705a : SUCCESS in 9m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4076bd9144c54f7c853701200efe48b1 : SUCCESS in 4m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5cc5d858d9bf41b585f2987995f640bb : SUCCESS in 4m 30s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0a157aa8e3494141a89f021545c6a9f6 : SUCCESS in 7m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0d4fe55ab49847e4b4750e46e6083098 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34a3df8e470a462db3cb1b8756075e72 : SUCCESS in 7m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2b8fb5ece43346539bfa7af5b1cc53f8 : SUCCESS in 7m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/90fd55095ed143e69a38e05f7587175d : SUCCESS in 10m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/885d0121835c47b999b8e11c30dbb5ab : SUCCESS in 3m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3c259a998a134a94ad75231a323871f5 : SUCCESS in 8m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dcc5069f522d4d4cad7574941c48e0c3 : SUCCESS in 4m 20s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/1f200962cd944bd994b33d97c0bcef93 : SUCCESS in 9m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8272c01c0d2040839e7c7b14fbd28907 : SUCCESS in 28m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c484a441d7c84571a773e013c40bfb0e : SUCCESS in 12m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8abb03833622454f9e3510439fb9e3e0 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89aeb6037eac418cb4f9628175096bf0 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/989d870d89c8415fb4557080e5833112 : SUCCESS in 22m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5ddbc7dfa8ab4620b8029e750d26d05d : SUCCESS in 5m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fa87cb90f34c48d0b51d97ad20b678ed : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44c3c17a226449e99200512ce29a8fbd : SUCCESS in 6m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fe995806ecc94c35b53198aa0e8da601 : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31ff2d8289514dafa07cd5096e8024e6 : SUCCESS in 43m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f44f7516f385414bb566f9e213784e47 : SUCCESS in 2h 02m 04s","accounts_in_message":[],"_revision_number":3},{"id":"a92c629cada773a0ffbc5d21cd2891a43067561d","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-24 16:19:35.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"37c0382629ba28d97b87d9b18e54755cd3ccfa01","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-24 16:19:47.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the patch, it should be good, but only for the security concern.  While requirements does try to work with downstream we exist primarily to assist in helping gate (so downstreams can take requirements as advisory, not a requirement).  In fact, the security reason isn\u0027t even listed, but I still support this for that reason.\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#stable-branch-maintenance","accounts_in_message":[],"_revision_number":3},{"id":"e3446a1d62d55b6c50e1784a53196760f085cd7d","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-11-24 16:22:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b8985589f9f6d72066ff751fbc5323411112ac2c","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-25 02:06:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d289574cc6cbd6a0b69dcc8e9b8e7b5037cc7074","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 16:32:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e28bd7d60727d0f327a1040e0a6237992bcbf25","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-03-21 01:01:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ae3dba05dc5d71a91e77adff5dff963bd79f28ae","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-21 10:36:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35880644ed0b77eda86191fe3094f1b7a2c88829","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-03-21 17:41:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c1447edc676b52e22e1cae5bc07586ed91c818c2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 17:17:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ecc22977fb91c7df147d86e5ad1e0b00e67a3293","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 17:17:50.000000000","message":"Abandoned\n\nI don\u0027t believe this chnage is needed or appropriate anymore.\n\nFell free to restore and help us understand why is is needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8076ce92b829866dc96028db62947bbe3e0014d5","revisions":{"20bebe17bae4ff4d8927d01bd4e642c006a62518":{"kind":"REWORK","_number":1,"created":"2022-11-24 10:53:02.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/26/865526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/865526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/865526/1"}}},"commit":{"parents":[{"commit":"7526a3c0193f6a694624c4ad25941e11f9b4e2db","subject":"update constraint for sushy to new release 4.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7526a3c0193f6a694624c4ad25941e11f9b4e2db"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-24 10:51:09.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-24 10:51:09.000000000","tz":60},"subject":"Bump pyOpenSSL upper constraint","message":"Bump pyOpenSSL upper constraint\n\nSince Zed is also deployed on CentOS Stream 9, it would be best to\nat least use the same OpenSSL and derivatives releases as master,\nespecially following the latest changes in this world (CVE fixes and the\nlike).\n\nThis bump will allow to remove a hard pinning of the \"cryptography\"\nlibrary, allowing to make the whole project more secure and up-to-date\nregarding OpenSSL ecosystem.\n\nChange-Id: I1562e0fd60a6340e26a77a2ab6102595dc7ee979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20bebe17bae4ff4d8927d01bd4e642c006a62518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20bebe17bae4ff4d8927d01bd4e642c006a62518"}]},"branch":"refs/heads/stable/zed"},"83c042feeadf60075763329caff056affb601388":{"kind":"REWORK","_number":2,"created":"2022-11-24 12:14:50.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/26/865526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/865526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/865526/2"}}},"commit":{"parents":[{"commit":"7526a3c0193f6a694624c4ad25941e11f9b4e2db","subject":"update constraint for sushy to new release 4.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7526a3c0193f6a694624c4ad25941e11f9b4e2db"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-24 10:51:09.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-24 12:14:30.000000000","tz":60},"subject":"Bump pyOpenSSL and cryptography upper constraint","message":"Bump pyOpenSSL and cryptography upper constraint\n\nSince Zed is also deployed on CentOS Stream 9, it would be best to\nat least use the same OpenSSL and derivatives releases as master,\nespecially following the latest changes in this world (CVE fixes and the\nlike).\n\nThis bump will allow to remove a hard pinning of the \"cryptography\"\nlibrary, allowing to make the whole project more secure and up-to-date\nregarding OpenSSL ecosystem.\n\nChange-Id: I1562e0fd60a6340e26a77a2ab6102595dc7ee979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83c042feeadf60075763329caff056affb601388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83c042feeadf60075763329caff056affb601388"}]},"branch":"refs/heads/stable/zed"},"8076ce92b829866dc96028db62947bbe3e0014d5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-24 12:52:29.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/26/865526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/865526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/865526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/865526/3"}}},"commit":{"parents":[{"commit":"7526a3c0193f6a694624c4ad25941e11f9b4e2db","subject":"update constraint for sushy to new release 4.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7526a3c0193f6a694624c4ad25941e11f9b4e2db"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-24 10:51:09.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-24 12:52:18.000000000","tz":60},"subject":"Bump pyOpenSSL and cryptography upper constraint","message":"Bump pyOpenSSL and cryptography upper constraint\n\nSince Zed is also deployed on CentOS Stream 9, it would be best to\nat least use the same OpenSSL and derivatives releases as master,\nespecially following the latest changes in this world (CVE fixes and the\nlike).\n\nThis bump will allow to remove a hard pinning of the \"cryptography\"\nlibrary, allowing to make the whole project more secure and up-to-date\nregarding OpenSSL ecosystem.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/865406\nChange-Id: I1562e0fd60a6340e26a77a2ab6102595dc7ee979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8076ce92b829866dc96028db62947bbe3e0014d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8076ce92b829866dc96028db62947bbe3e0014d5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
