)]}'
{"id":"openstack%2Frequirements~865547","triplet_id":"openstack%2Frequirements~master~Ib9d02f46501024f77fa27fabf26b52450d6f698b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib9d02f46501024f77fa27fabf26b52450d6f698b","subject":"update constraint for openstacksdk to new release 0.103.0","status":"MERGED","created":"2022-11-24 13:49:08.000000000","updated":"2022-11-24 17:17:58.000000000","submitted":"2022-11-24 17:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865547-new-release","meta_rev_id":"addbe1df56731d7a1821500b807dfa65e7fd4f09","_number":865547,"virtual_id_number":865547,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-24 17:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-11-24 14:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-11-24 14:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-11-24 14:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2022-11-24 14:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 13:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-24 14:16:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-11-24 14:20:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-11-24 14:27:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-24 15:29:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e263042755c121490e35d657f42d5aed1c2c3bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-11-24 13:49:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa82773eb864e0f692ae197bca06040a1681525b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 13:58:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7cb4c899da84d6abe368919f5c0d38e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/eb6bbfeb443548d7a623d1ffae30e694 : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"80d74cf7106455124cc8588be8984a7f49f4707a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-24 14:16:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bdfa4f2b636206a42695f925b1efebf746434e38","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-11-24 14:20:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79bb1a92e333e774cbfcdc01f8f2b4ce2c1432eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-24 14:27:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f936eb6ece06e0571cbee867618a5ca725db138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 15:29:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ee3094ca6744e1489a0a987cd75c168\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa99d4412b4494c80ad51ba5ec5165d : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b45e7d409b2a4885afb7b8d028740fd9 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4215d4625c4c78ae14ba3927ea6e74 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca878dfc3cb94615af2053629b3214be : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/004d24823f734bf08384fe9d1db3d77d : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9ab29aa0f494ad9ac16b2252c89dd69 : SUCCESS in 2m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cef72ed6a20a4dcf8d5a74b0532f20df : SUCCESS in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4672b87422b740bd93f4c37da0993867 : SUCCESS in 13m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/771f22e2abf641539dedfa8546bdd7ff : SUCCESS in 8m 16s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/06fb43bed00b47fa8e2ddb16dd56d01e : SUCCESS in 5m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6275c08c6f0c4109b16a901c55e08e71 : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cfebd8db45b74762a8404bfa3cada1f9 : SUCCESS in 11m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/37df7376c6d547f89ff49202f12e1fc1 : SUCCESS in 5m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b8e28e30ea7f40e39603367ae0556d5c : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3b359bcb62cf4b29addf2f24f5777e0d : SUCCESS in 7m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/43ce1df5a69c44dd96bfb76a2f0667ff : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aaa2470f03564c3dbf7630df64fd8553 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7b0afbbaa0f64d29ab9d78749b55a19c : SUCCESS in 8m 05s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e51631689bce4b209fece1452a11b1fc : SUCCESS in 10m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/32f60d5285ce40ce9ee9090ea1aa0dc5 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/689c43ca0c734ea892ab2f2b6605f8a6 : SUCCESS in 10m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5994e3cff7b94f3a8f1d3a03c4cc2b45 : SUCCESS in 4m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fb50d8d93a1e42cc88331b767e21cbfe : SUCCESS in 9m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e8aab4efd7434c73a4ef90fa1f08ddcd : SUCCESS in 22m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/45ea0288f1e742d9ac8f61cf65f5c4d9 : SUCCESS in 12m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7f7b48ba494e4e2d951366e663f45815 : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a2d5eb7effd448b88d1b8b7ab80560f : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee1cbc2d73974067bc7b41de12cd0fc6 : SUCCESS in 19m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0e8b180c1f2842a2a143971ad771503a : SUCCESS in 4m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9bf1a08e6de84658979f44da2d7278bb : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7febc37336b0462680659dcc8b56d296 : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/91ad6561c1e7437fa8efe8e22df8f55b : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5f889efb9fa44979de858e350d1024d : SUCCESS in 53m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8177c24fee414295bd3673d3355555dc : SUCCESS in 1h 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"311b90c563706a1fafe29d194fd328e36136c262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 15:30:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67c14a0c45510230e44f935c99ace68c8bedaff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 17:16:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a9b5ab0cdb64526be0d6f730dbef1d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1d270bd12d481c858ff20f52be3e96 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e1dc9d326d4a5c871070abec31f956 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f953353fcc7412dad997f795975ca10 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a1606ef3e494235bd610ffc9f5788df : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a504b6f33c2456f9466f38e95d34329 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/123d8985761e42448049dbd060f6e3fa : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e048f4008457436aa17ad1af729c66f9 : SUCCESS in 10m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4d1328b823874936ad80d3ae36920f38 : SUCCESS in 13m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/932c6e7296a84f0aad1a4915f660eafe : SUCCESS in 11m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/54d7f861df4e41809ef2bce9a196cb8b : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cc12f364131e46d48f0f721a761db935 : SUCCESS in 6m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/75a74d975da54fd48e43423318853b94 : SUCCESS in 6m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1a96f5d2ed394d1b9246d4707d3060e3 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/104640eda77b4df88458a58d741c333f : SUCCESS in 8m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b3408c5c77c74a9697cc41bc943f8695 : SUCCESS in 8m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/74baab748bf64a1e912f75e4a6447370 : SUCCESS in 10m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a5ce688e80414653938fcf4fad07213b : SUCCESS in 4m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ebb71c2f04d645bd85bdf3bf6dfe6b95 : SUCCESS in 9m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d3c76eb4c8504beca8d67b8929c1e4c3 : SUCCESS in 5m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/37ff25a44a2747e586e75f8f1d198de2 : SUCCESS in 25m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/37e83e7a09fe44bda2237fba4e55d29c : SUCCESS in 10m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/65fc2e9002304d639d201792609b6c07 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2ab091109b6a4f7fbfbae0aca85d079d : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d249a9761dc479b993b84223f7ae3d5 : SUCCESS in 22m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/51e8774aab1a4bb3bb84c003b285c7be : SUCCESS in 4m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/72b7fc3702db4b44b77b9cca28d66432 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13ad994e0a004ac4b8153442a0d06450 : SUCCESS in 6m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9046b6deaff54050b5eac61320bd43f9 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b867a0876928418a88c3e93602e08843 : SUCCESS in 54m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85f5d163fe9442a2bd84b8df68a8a28a : SUCCESS in 1h 45m 14s","accounts_in_message":[],"_revision_number":1},{"id":"daed103980c256dcde75260d174a80dfddad1fa3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 17:16:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"addbe1df56731d7a1821500b807dfa65e7fd4f09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 17:17:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8bdd2bd99884dd6b2adf3c9ca328b72\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc12db18449c4258a556eb20a48863e0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e81ba13b8719448e89245f5541a1fb61484ce4c5","revisions":{"e81ba13b8719448e89245f5541a1fb61484ce4c5":{"kind":"REWORK","_number":1,"created":"2022-11-24 13:49:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/865547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/865547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/865547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/865547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/865547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/865547/1"}}},"commit":{"parents":[{"commit":"8662b464ecd56ebd750cea80b951a305a5136e76","subject":"Bump upper-constraint for cursive to 0.2.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8662b464ecd56ebd750cea80b951a305a5136e76"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-24 13:49:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-11-24 13:49:07.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.103.0","message":"update constraint for openstacksdk to new release 0.103.0\n\nmeta: version: 0.103.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: Id4ff2853464e9fc544b7da1d7f1fafcadf281303\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib9d02f46501024f77fa27fabf26b52450d6f698b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e81ba13b8719448e89245f5541a1fb61484ce4c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e81ba13b8719448e89245f5541a1fb61484ce4c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[]}
