)]}'
{"id":"openstack%2Frequirements~867227","triplet_id":"openstack%2Frequirements~master~I75a7e43ee7e175bd172e3343d967eb29753fcb70","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I75a7e43ee7e175bd172e3343d967eb29753fcb70","subject":"update constraint for sushy to new release 4.4.1","status":"MERGED","created":"2022-12-12 14:01:01.000000000","updated":"2022-12-12 22:06:52.000000000","submitted":"2022-12-12 22:05:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867227-new-release","meta_rev_id":"210922e41fa74148c96eaeed2d1119346960ff9b","_number":867227,"virtual_id_number":867227,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 22:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-12-12 18:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-12-12 14:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-12-12 18:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-12 14:10:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-12-12 14:12:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-12 15:48:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-12 18:26:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"075f824a1abc4e8b4f46745711e49bcbaf42c6a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-12-12 14:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f7c1a04f08f3c0dfd812c114ab6bfddb646e59d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-12-12 14:10:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"789719587520aab25993cf05c00b3ee62c478559","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 14:12:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d081ea41f9884916bdab14d6ce938d24\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/17f08924ac2542f5a3e0c384b7c27b92 : FAILURE in 9m 31s","accounts_in_message":[],"_revision_number":1},{"id":"86de247144b58a1c3b7ea29d5dfe40c324585dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 15:48:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313e5a72a93f40b9a1b4d108a4dcf6da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4e54af86aa4f2988d12929b5136445 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5b12e4cd8443f0b33844cff3c8452e : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b12d8e54715c4b34b7e4c04fdceb74c1 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9fd8c3638f0465db64ea7904921397c : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e53b1514acfd427db65ff8147c75dd07 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30fa035c7cef4f26bf0617d03960498b : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/27ab74d9e06141e9bf449130226749ce : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/84ea89b4ea994463863583b9735443ad : SUCCESS in 12m 52s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/536f781ba7544705a1de59e59c72a0f8 : SUCCESS in 7m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a9823a1c05234afab590a54927253ed4 : SUCCESS in 5m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a7a36c7b359d44bead310e4a27a19e38 : SUCCESS in 5m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d80ef65e34544ffc8154379c8d2d868e : SUCCESS in 10m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8dcc179c386443afaf14aae1d4d470c7 : SUCCESS in 4m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/54810617455d46e3a5df5039c5b2e25f : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/47f58e81060d411482fc29ef58f79b7e : SUCCESS in 7m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/293adcfbc4c6427695ec8a9f338f6576 : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cac611e51fdb42339023e564b0f87376 : SUCCESS in 7m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c64a23d3f6bb45448a7b04448e0097b5 : SUCCESS in 8m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c9d5ab2c610e4b01adfb657e7ac74c6d : SUCCESS in 14m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d89cd6e2dfd4493f8996604f86b822f6 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7707fac1b2464e559c01aa1f73603f11 : SUCCESS in 12m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ef86d0075f6c4eb4b7ec066210daadd0 : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5c0b09efe4924d6b88ecd54a29ed1137 : SUCCESS in 8m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4f8504c1a2f54cd9a0c99d57b9b9b4ef : SUCCESS in 23m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/11ee053c70f74acca1854e0fb1428fd9 : SUCCESS in 13m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c73a8bcb287a437a8e91e43ae3dfbf8a : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b2c89bb9afe4649902336a2823c1830 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53538a16ed5a4f4ebc8fc9244ebd02b3 : SUCCESS in 23m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/812f45f483e0406a9bf025e459e29133 : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/03df9ba549e043e49a12300abdf76b99 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/549e028ae5ed4b76b0268a7961dfa0a9 : SUCCESS in 6m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/26cef37f15384e72bae29aaf90823829 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5cbfb82db19434897f913a6fa0354aa : SUCCESS in 55m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f87bbe0f42e46a5919328f1173dffbc : SUCCESS in 1h 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"47ebd0815b3b747243395836cc4136adfcc76729","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-12-12 18:26:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f3a93144d34c25bb3edc87332e2a8cb96a6125e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 18:27:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48754eb54fa26a5ae463b5de897747e0fc5b81d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:05:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac999bfcee6d44f885ce94a9715ed01d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da0139611434c70a307039a46779862 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98ec433f2844fb8852c4555007af794 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c29c3394cfa46828d2694d64808e6e1 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6e86b345b074167881d205b13411c62 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60da18ab941d4f4ca0db47d56afa1be5 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93aea7e052b74cc7b747e8b70f1c730b : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b6d54a5ad9a4c0fbe571a85a68cbe41 : SUCCESS in 10m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fdc1c876b26242ab8632b5862d58c5d2 : SUCCESS in 13m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e52cb6a3699343bdb6289323f274fc11 : SUCCESS in 12m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1d5d2fb711984c24b9028ac9b0563915 : SUCCESS in 5m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d11623b244c746c8bbb65fa42fef45fb : SUCCESS in 7m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/757766cc92674dcc9a2cd46db7c7adee : SUCCESS in 8m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2bc1cf25c333443db19f9c1c562dcd6d : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f378e3156ae5468ca0d58041d6adef05 : SUCCESS in 8m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a53dbdc3c0464812a5f2d16f98a608e7 : SUCCESS in 7m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f543fe76820b496c8e5a2fed7e2fa231 : SUCCESS in 12m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/67e6a93af2b44ce8a2561090305f46c5 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4f4871511385410893080328ab6b16cf : SUCCESS in 8m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6e750d125bc9463bbda052ce4e2b7bd2 : SUCCESS in 5m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4df75007caf7425b95a873362b260ff2 : SUCCESS in 28m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fd6911c8f49f45b1bc034ffc21c8fe7e : SUCCESS in 11m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d09ecefc1f88481eaf8862858fd8bf78 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2037e3793e641f9a3991b1d478356f7 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d3f2d6e4eda4e359a61029955833ce9 : SUCCESS in 18m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3924241db8f74be4a53d0ba0c166116f : SUCCESS in 4m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/21cd8f544f07420e8daf9ce26f7ba906 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c9a842e3ed34351a6c0c583ea01ebfc : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/baa5a142b9ce44799eb1148e0a12b86b : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b233552dbc342ada92e1434c623b7eb : SUCCESS in 38m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47e2415e00c941a0b48b2abe2d261510 : SUCCESS in 1h 48m 14s","accounts_in_message":[],"_revision_number":1},{"id":"13a5ae89402d57a261d7c253cbb234c3d201e517","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:05:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"210922e41fa74148c96eaeed2d1119346960ff9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:06:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9537048894764d55a01f91cff377dc70\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c1c6c659ce475590e3ce84dc42837c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98d8436315b35b092d5d5a25d3ac1cea3d631475","revisions":{"98d8436315b35b092d5d5a25d3ac1cea3d631475":{"kind":"REWORK","_number":1,"created":"2022-12-12 14:01:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/867227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/867227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/867227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/867227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/867227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/867227/1"}}},"commit":{"parents":[{"commit":"10afcf9e3916a8edadc2f20b2548d11991d3175a","subject":"Merge \"update constraint for metalsmith to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10afcf9e3916a8edadc2f20b2548d11991d3175a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-12-12 14:01:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-12-12 14:01:00.000000000","tz":0},"subject":"update constraint for sushy to new release 4.4.1","message":"update constraint for sushy to new release 4.4.1\n\nmeta: version: 4.4.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Ic00436bba576ade4b4d8ecccb1868d2b7521c7c1\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I75a7e43ee7e175bd172e3343d967eb29753fcb70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98d8436315b35b092d5d5a25d3ac1cea3d631475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98d8436315b35b092d5d5a25d3ac1cea3d631475"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
