)]}'
{"id":"openstack%2Frequirements~867975","triplet_id":"openstack%2Frequirements~master~Ic147ce220a5fafaab74f14e6408602bdd1449de4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic147ce220a5fafaab74f14e6408602bdd1449de4","subject":"update constraint for python-cinderclient to new release 9.2.0","status":"MERGED","created":"2022-12-16 14:44:44.000000000","updated":"2022-12-17 00:58:06.000000000","submitted":"2022-12-17 00:57:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867975-new-release","meta_rev_id":"50521e6e0f04d11a79bfcc33fa23c333c9c07ef2","_number":867975,"virtual_id_number":867975,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-17 00:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-12-16 16:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-12-16 17:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-12-16 17:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-16 14:58:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-16 16:00:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-12-16 16:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-16 17:11:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-12-16 22:10:57.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"189574116d2417752e020a5a8350829490ef362d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-12-16 14:44:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b1136aa03bb352bee5fec8cbf5e677ab1c5b7d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 14:58:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a4bd95efbf43a2a82c40697657748e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/de2ed3d22c534bb18d4327575d90e31c : FAILURE in 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f3e151a54b20f55bf356625c6e3872fbdacb9949","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-12-16 16:00:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5433bb563bceffc32730bdc97a7bb0cf6003ec55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 16:28:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d2b299086314f7e8ddeb3ae24c592d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8a24fa086c4a42aefe92cff6c0c50d : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c38ed8aa06904d16a572cee8c87be327 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edf3139b70634ae9ab129ce63f653722 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a65533df0f0f4069af760061218ee41c : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cce064af0f5947fe9bb70f1a24326a86 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec4e0240cbb543a1a9ffa28540ace9f6 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c54c5fcd94c4f258b130a1e21043235 : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/75a5fce0c2ac4df5b73af55f9778b2ab : SUCCESS in 15m 21s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f6653cf04d0b4ff690dd48fe33fe465e : SUCCESS in 6m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a80f7a4e7d3343a3943d09ab262d736e : SUCCESS in 5m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1bc3843b31004448b455516329e9efeb : SUCCESS in 6m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/64ee78d3b792426db7e69042b9ad1042 : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8fc6acd909ad42d095012fbfb62e1cb0 : SUCCESS in 4m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/afa5f10133a34592ad7e950db8c77269 : SUCCESS in 4m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/289072cb6a0d4e4ebe6e05dba408bc8f : SUCCESS in 9m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/75f81e4a94f54dab8563b9561db71206 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b3308e101db470ab062f4fbd27a51c0 : SUCCESS in 6m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/235b5d6ccbf2426cbf1b631c8350c8ec : SUCCESS in 7m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d84fa4f5558842a3bec7a11c54d30466 : SUCCESS in 12m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/478764e169e0479a95a1764f474603df : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/de519867566849dfbfff7df61f20085e : SUCCESS in 8m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8efb457184a84055a06efaa594616cfa : SUCCESS in 6m 02s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/01f244795d67431ebbbcac798094935b : SUCCESS in 8m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5a3229bac6b14566b990ce2c4380f974 : SUCCESS in 25m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ebe6bb84ea104bbc96bdcd5fab6ef416 : SUCCESS in 12m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d43faaab387f4893bda29056c73b164a : SUCCESS in 6m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93539e98689f425c8c5443a0a42045e8 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63900594dfdd4ceb854c66658aa911a7 : SUCCESS in 23m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2d547aa6e6fa4a66944971666e04b2b2 : SUCCESS in 3m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ab4420fc184440e1947bbc33c309cf05 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0726c539252546dfa3c2d5eabae1eb7c : SUCCESS in 5m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9831232312e940b99aa7859142a1c72e : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/574a61b4f7dd4c55b86ae5f2772f79ee : SUCCESS in 56m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/933d3084feb64f2bb45ed65e8099333b : SUCCESS in 1h 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6dc32864c6a7016ddb3056a3aa98466f5b3f2e59","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-12-16 17:11:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd16ebcb6c712f25c5dae58e95b03c561445fe31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 17:12:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a39decbda1d7e2643e0293f92231a100a908d1af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 20:18:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d07ffd67a7de4d33aca69e981203c2fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21ff0f4298f4d0abf50fe86e9cad9cf : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9f9c4c832c4409a199a4f512ec6d07 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/078fd0ea3f454e58a1623cd6de5489d3 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d35ba38d4aa54734aa2308b02329e006 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad69793556f04b9b8121c1613f330f3a : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1661973ed1744442a9cce8035b066374 : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b31852e5f0e94592a4d66de215658814 : SUCCESS in 7m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e3355536a1db43fa8b9e2f12523dba56 : SUCCESS in 13m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6debb2e7990745b59f7ce4d04d81b908 : SUCCESS in 10m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/048318ae4afa49d59d2ca941fe41a28e : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/85002096949d4b6da673c5d8da6b7c0a : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/700b0ee53b704cbda485a11068264bae : SUCCESS in 8m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d97feebdb35749f292ccfaf4ab3cfd0c : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e355fbace3214dab8803a49bf449e724 : SUCCESS in 8m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/380ab1c8ea854713b04717a02651641b : FAILURE in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0a83a43c0f52412a8ca1d24a98947223 : SUCCESS in 11m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e1617fca3b5840efa85e3388f75e3004 : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4a115347639148f8aacba7f24d35b7d5 : SUCCESS in 10m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/85e00d676b68490094d5811eae10246f : SUCCESS in 5m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f3a34e0662b649388ae3bf0d6fad797e : SUCCESS in 26m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/63213aea8aeb47919049ccec4d565015 : SUCCESS in 12m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d258416a14164f1195e13eac809f1174 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4fb4af2c2a7d4b6ab343e0e2facb5139 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/212ec18e49a24aecbda8fa89474ad558 : SUCCESS in 20m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/095bacdd63914051aa1a198219139301 : SUCCESS in 4m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e1a4ac2817244c42a5046f890731673b : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0763ee155bc64211af2117bd20489ee2 : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1848ecb2cd0d496596bfd07e29a4fc97 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/181c6cc038814bdfbef420f2c4fc4dc0 : SUCCESS in 48m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/950ae5da266b458a87da4bec3d86a442 : SUCCESS in 1h 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5ba134309898b8ee81ec6358a85535b934fef887","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-16 22:10:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90b0281268ee1ba4b94cbdd05dda63435ea6cca8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 22:23:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01fe6d2e186149278249555f3c2f9dbe\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/297f727d144449cb8eab683555c9bf93 : FAILURE in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e7bd5e06a3f8e8e6f01544601ffcedf108c47dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 23:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793cd5dd9cbd4b1a9fc1ad80998f1c4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/796661dc8d9843868e27ef1239f9486b : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bc78c083da4841aa3f14c85a19b010 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9a90d6eb9d943829ed1a98eb79c0ea9 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9a66d8cd2a543ec9f3eb040633f89e1 : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8d8d66e750a4828bed5c4cad039d479 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e752feb0fcdd485daf4f59a776b255de : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28391566a9934b69b41ed62a98c98596 : SUCCESS in 8m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/509d77523d9c432cbe25d4f8e263c505 : SUCCESS in 11m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c6697f84beb049f3b30532f630640dd8 : SUCCESS in 6m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/73883d84657c488b9452be4f42bed432 : SUCCESS in 4m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/cfa5c821afe74c11a59e31cce2934d42 : SUCCESS in 5m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9be78ba9a6a74befab42a27c42a2dfc8 : SUCCESS in 9m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/688109d60efd4d07b9294530ad19b63e : SUCCESS in 3m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f54d08a8d3ee4da1a1af37ce2d3e1158 : SUCCESS in 4m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a35bdb81d7fe44faa38c2a8032dd24a6 : SUCCESS in 6m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3b0f875fc0fd4334b57eab00a0c0562b : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71e78e7d9f524f89bbdb367b8ba94d28 : SUCCESS in 7m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/390707a95f9849a8bc72005bebe26195 : SUCCESS in 7m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/549961c4ec074db2b2bcb463ec4b3310 : SUCCESS in 9m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d0c6b77642ae46f099706c816b76f5c3 : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a5e50be6fb254b38abd1b0fdc383d82f : SUCCESS in 7m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8599ed35b43d4fee91881c53b7ae0913 : SUCCESS in 4m 08s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d2b46bd9715e4987816b6d6226ef1eee : SUCCESS in 9m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fb98082af7f345119f1eb1d22cff0156 : SUCCESS in 19m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f934e695e87c47998da5d3cdfcb75cc0 : SUCCESS in 11m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e828bb19d6c741c0bb7ec223e85865d3 : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de499e7e8a344fd4b65c6519e744aba4 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aabc9cf16af249a1954822859a2333cc : SUCCESS in 18m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/54a18b7b47ca435799d7fed7df9e3bb7 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/686567533f474aa1b420e51bdd562c4c : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a58f3061d66742fe99b159d5a5931e46 : SUCCESS in 6m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/262c01b7fcb54d8f9845131580fd5f3c : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/748a8de1c66c4136909b2035d6b4fb96 : SUCCESS in 1h 04m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d34d25f9f3246939a2890a50b33b850 : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e31d1da7d4264661ceab8f97caea38cdebc35953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 23:30:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f83cac1dc42c51e6b5a89afc4e84821401338c6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-17 00:57:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14670d235f8c47d4b0a0247d130d98e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d81d5ac6ccb4fe4b1e25dd2f61ea4d4 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac50d1c485884112b370cad29f65359f : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eda13fdfa6444d219d282507d7257924 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60aec50452314bc78f8f20dd0030912b : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b591d68d54cc46aca614ef19507cb65c : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6d038485ed84c13a8171536e392c0a1 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e11f62db0c74771a14fbcab4f44e455 : SUCCESS in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c6b61300ad3148bf9533d720285a5608 : SUCCESS in 10m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/671ecdcda9214f16a7b6164d2ca56d92 : SUCCESS in 10m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d20cd882f15745d19558dadd94337cf0 : SUCCESS in 4m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8b726fca04eb44db9f271d1f37f8f7f6 : SUCCESS in 4m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0c3c8b5b92844477869f317affb30d6a : SUCCESS in 7m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/45765e41631b4aeeaec2a41ca639f0a8 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0730d358ee964a869cf5a916f97d3d5d : SUCCESS in 10m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/66d6eb28f0664147b0d1602da24c1e19 : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/24e5b6fefa824eef99c0b595c6364173 : SUCCESS in 11m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9c346c8c63e1491395e1515a36f9774e : SUCCESS in 4m 08s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/20f5d9edfb744cd49beb46a403007909 : SUCCESS in 10m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0bf2fe6aadc44f068463249f8e216b81 : SUCCESS in 4m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c029e5672edc44ad814937dfaef8e4ca : SUCCESS in 25m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/187dded4a24b45ae9118b9a8a436a6c3 : SUCCESS in 12m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/650421e1b2b84915b20630eaba9022ca : SUCCESS in 7m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/717fd76ed5514910b74f0d0a905bba9f : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b3fa73f1d6b4639ab5e1ec2839fdec2 : SUCCESS in 21m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a70bde9e736e43e2b740aaffc3d2b392 : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/eb09093c3f2e4a4e9a89493bc5021be8 : SUCCESS in 8m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ca9b19ad9b24410a9147e2b50e24f41 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/61c44be0f7ad412684cc85b975fb8cf0 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f49bf9a002a34ad4910e5e2ba7e5431b : SUCCESS in 38m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1031b753ad14385afcd14cfceaead3d : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":1},{"id":"43d6a7a865187d90d43658867900ded9dcd860ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-17 00:57:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50521e6e0f04d11a79bfcc33fa23c333c9c07ef2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-17 00:58:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87a182ad591b4c0995adf4196e191f11\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc44d465f924271a5bb7e7e9b1a42c8 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90da966b4f283f140970663e0ce424e40ddf94ee","revisions":{"90da966b4f283f140970663e0ce424e40ddf94ee":{"kind":"REWORK","_number":1,"created":"2022-12-16 14:44:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/867975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/867975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/867975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/867975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/867975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/867975/1"}}},"commit":{"parents":[{"commit":"98d8436315b35b092d5d5a25d3ac1cea3d631475","subject":"update constraint for sushy to new release 4.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98d8436315b35b092d5d5a25d3ac1cea3d631475"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-12-16 14:44:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-12-16 14:44:43.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 9.2.0","message":"update constraint for python-cinderclient to new release 9.2.0\n\nmeta: version: 9.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I57049b81353c8363be86ce3710d0c7a6c460ffb1\nmeta: release:Code-Review+1: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+1: panhongyin \u003cpanhongyin@yovole.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic147ce220a5fafaab74f14e6408602bdd1449de4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90da966b4f283f140970663e0ce424e40ddf94ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90da966b4f283f140970663e0ce424e40ddf94ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
