)]}'
{"id":"openstack%2Frequirements~868902","triplet_id":"openstack%2Frequirements~stable%2Fzed~I7bc7c98954395765f16cc943e5c826983db5dba0","project":"openstack/requirements","branch":"stable/zed","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-30 04:21:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bc7c98954395765f16cc943e5c826983db5dba0","subject":"Fix tox4 error","status":"MERGED","created":"2022-12-29 17:13:33.000000000","updated":"2022-12-30 04:21:00.000000000","submitted":"2022-12-30 04:21:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868902-tox4","meta_rev_id":"9141c0c11dac3de25661f7a2448fed5539c936fa","_number":868902,"virtual_id_number":868902,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-30 04:21:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-12-30 02:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-12-30 02:09:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-29 18:36:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-30 02:09:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c0d4e040257568ead6129923aecec7f2983b06b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-29 17:13:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56f8ab2783c679740054be832bd52a59eee45609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:36:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1475ef0c51f04233b7072d34feda190a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d87bdf9ec174ef0b437590ebd382704 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0add27abe9904f889c6534acefce137c : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35551d59e78541738059a5616ff5605d : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aba3f3ae5dc45568a6b721f9710ac7f : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12deb8d104954249a836f62fde35d884 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0d557b6e5ae04e7589502b8823caec01 : SUCCESS in 3m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89c791dea4c94ff8a7aba11c928d226f : SUCCESS in 54m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9b40c156459413ab282fe1f7b01b41c : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c97a9102e6208420a895abccd3180dea3221ebcc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-12-30 02:09:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f3ecc68a593718ff8507df2d9965a42d09b0bfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 02:13:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a33633cee93b8f024420a2b9bf47661f09fcf3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 04:21:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffc90509c5f44f4a531cd2c92160165\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bdca63a4d1d447da7f564124f39b54d : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f66ff0aae4249d0b32f5c4f9fad0ca7 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22bd765d6d2947a29cfc5cb63263ec5d : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8f46725f92c4d61ad38e7dfdafc654c : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e351e9f74e034098a7f822e138f9c1c2 : SUCCESS in 2m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da9e721b183b41fb859fd38412e19ea9 : SUCCESS in 1h 06m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5188cc59cbc48c19344e87478f44df6 : SUCCESS in 1h 28m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9141c0c11dac3de25661f7a2448fed5539c936fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 04:21:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4e42a86d9253351be2da7e0cf31262546c51d3e","revisions":{"b4e42a86d9253351be2da7e0cf31262546c51d3e":{"kind":"REWORK","_number":1,"created":"2022-12-29 17:13:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/868902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/868902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/868902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/868902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/868902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/868902/1"}}},"commit":{"parents":[{"commit":"7526a3c0193f6a694624c4ad25941e11f9b4e2db","subject":"update constraint for sushy to new release 4.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7526a3c0193f6a694624c4ad25941e11f9b4e2db"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-29 05:17:58.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-29 17:13:19.000000000","tz":-360},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One of them is to allow\nexternal command to run by adding them into allowlist_externals\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: I7bc7c98954395765f16cc943e5c826983db5dba0\n(cherry picked from commit c393eb75a46bf73339299ab658173ef316283260)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4e42a86d9253351be2da7e0cf31262546c51d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4e42a86d9253351be2da7e0cf31262546c51d3e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
