)]}'
{"id":"openstack%2Frequirements~869063","triplet_id":"openstack%2Frequirements~master~I99c20cf35d6e832a8b341d431a58dff25118b4da","project":"openstack/requirements","branch":"master","topic":"tripleo-ansible","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-01-10 05:07:21.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-01-03 17:56:41.000000000","reason":"removed on reply"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2023-01-10 05:07:21.000000000","reason":"Change was submitted"},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2023-01-04 07:57:12.000000000","reason":"\u003cGERRIT_ACCOUNT_28223\u003e replied on the change","reason_account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}},"hashtags":[],"change_id":"I99c20cf35d6e832a8b341d431a58dff25118b4da","subject":"Adding newly released tripleo-ansible into requirements.","status":"MERGED","created":"2023-01-03 09:59:58.000000000","updated":"2023-01-10 05:08:18.000000000","submitted":"2023-01-10 05:07:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869063-tripleo-ansible","meta_rev_id":"f6619d0a5760c931158fac1747c1bd9417e65573","_number":869063,"virtual_id_number":869063,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 05:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-03 17:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-04 07:57:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-10 03:46:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-03 10:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-03 11:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 12:10:17.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2023-01-03 12:11:07.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-01-03 12:11:42.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"deebd44bb27f362c9a6ee4e7ecd858a062356bf3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2023-01-03 09:59:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"297802a87323af793f55c367a327777b02439485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:16:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd62c90cb099414a86022ef7a9b469b3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dd04e43aa0114908b426752724a33125 : FAILURE in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f73f1fd0463e7073ea2e69bf796ab562ab5aafc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:23:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a04bfb7be4f488e980ca55d5fb5d9fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc6d536136647fda8658d3dc3d74d3f : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14033c8fd9844200b38e469496c8c505 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d11e0f9299340bc851a67859931fe91 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/179a0004a960416b864bf0469fb4d90f : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87097c25de85468d8b549aa570199219 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e0e5cfa45254da9ae6096f5040df5ce : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c95a05b109984e3db5b69c360d8977e3 : SUCCESS in 8m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f805d55153f245a2950fafdc6d94d276 : SUCCESS in 12m 11s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f7ef11b49ac5402c8cde990dfda671ee : SUCCESS in 6m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/23c3c3eda76d49e189a4f4b45741f8a7 : SUCCESS in 5m 59s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0a20113c94fd4d6784f97a6e0608a848 : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ebaebd2f118c41de80e3d03e7b00fe02 : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2724c3a97f8748c48e4fe1e283946c81 : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4ea99a348d3d448481219a9307dc9ae1 : SUCCESS in 4m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6602621640624573bb4215fcd6acde8a : SUCCESS in 8m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/37de65136d1d492f919af82eacf3506b : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52120d6073cd4b7dbc5a5165d8156be7 : SUCCESS in 9m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cdff8d949d8d4e43b5f07c4c2e6807a5 : SUCCESS in 9m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8b5d87b174bd4220b86f0da6a45605b2 : SUCCESS in 10m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bd10867523924a89ba23d2a2a120baf1 : SUCCESS in 3m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f81e67de5e0b4adea09b65eef800c4e5 : SUCCESS in 10m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/75e0a5572577495ea36a5b0a4dc14679 : SUCCESS in 3m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fea9d0911a9f4847a2f7f1514064cc7c : SUCCESS in 11m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e72c27e1a0d3404caf9754a60218d30b : SUCCESS in 24m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5c595437535141b69c225cc72f48d290 : SUCCESS in 14m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f9c16bce12c64d309b1e9bdafbd22228 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dff76fc6ecfd489a9efcb3daa9f5c5cd : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f2ad8f2d72ec445ebcb2d1309aa098ce : SUCCESS in 22m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/993e5ad49f96469b9bd082b33920a442 : SUCCESS in 4m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f226c7bf3cab4cfcb7d25268ed26fb9b : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/647b658baf1b4fe68b24ffb3ec991134 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/600bac9d8f8c4be29bed269e4726bb2e : SUCCESS in 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d53ef99ea0c24382bd4f2aa1008a643f : SUCCESS in 59m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11fe1d08b64c4d4a8c1786cf2628d899 : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":1},{"id":"609b87d65b67a87439be6305e4ce550a7c0173ee","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2023-01-03 12:10:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"8de4eb9e274f3958038bfb0b6a24a3ae5048c7fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-03 17:56:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"175cc5bb7422b6da63a4e3c82272b83b0d346ef5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2023-01-04 07:57:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"afdd479dd2fe75e3252ee1b2491455608969539f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-10 03:46:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75b30c3f7371de7d3080fee83330a7eac177be5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:46:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc9de370fd160686fccd0d94b8316b310c4146d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 05:07:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f0ca598c851431ebcb2f074f27af7e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c2bd6f7ffc46f38b8fb26ab6ab4644 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40f04986cc1476b9f6d5a08dd2bec66 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ee04961c57b4a85aac7ac254fdb8fce : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00603ed9cad3401f8f42aa7efaadb6db : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebebc604cdf34a9a92cde242f46e89d9 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6156201b362c48e2b5a520885c669cc4 : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dde69e58c75146f085cc8a995f45aee8 : SUCCESS in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb4b10e80c65438e800eda1cc18db7b9 : SUCCESS in 11m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/08eee03c8c70429d94bbd9dce2d1a5de : SUCCESS in 10m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/adfedb870af3400d80278684f75ea033 : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c65e6a8a069c4218b3e883328f6b42eb : SUCCESS in 4m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2a1748b769ab40c78a776b025c32d712 : SUCCESS in 7m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c74ab4cc5bb34d35add2da30618e997c : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7e3fe5942dd4cbba377f7454873f423 : SUCCESS in 6m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3571def6658740fba4ca30cbbb4b593d : SUCCESS in 8m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/627a12e7381b4a0f9bdb4bebb1c080c4 : SUCCESS in 11m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/84d154f0adcf44868483a62f8e8d695d : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f24c0302d2fe4a62a38ca2b1ff28da86 : SUCCESS in 9m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/138f180523b2435f97f4a1ddcf3e57e1 : SUCCESS in 4m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b682b6a51638402281732ce856d7512a : SUCCESS in 25m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e1c4f6e485a547f3aa7ea26c851320fa : SUCCESS in 12m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cb9cbc5352d442f5afd2e17faa2d6185 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95a1000d692b4e9e8358fdd1790cec3d : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57f2e57ab05d4131bfb627726426609c : SUCCESS in 26m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d3e4b9b52f0b4191b642510bed9b05ec : SUCCESS in 6m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1c792c2e6ebc4f5c8823c78abc3fc0c7 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d315dd5ee4d74a3f89358c5733a5f86e : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9276aed76a8b4db4a754c358b523ea19 : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d977eeeb825419f8d79331c5e8bb1bf : SUCCESS in 45m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52a1dbb789194632877be5eedd9b926b : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ea74848340ff33c46364e5e975548036e0f149cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 05:07:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f6619d0a5760c931158fac1747c1bd9417e65573","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 05:08:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c825d36cf3a04226a678f66b5be9358d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bcb0b92c5694584992a12a386f394aa : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57f5f705e84289ba9fcf8fdd479b555bfd7b4168","revisions":{"57f5f705e84289ba9fcf8fdd479b555bfd7b4168":{"kind":"REWORK","_number":1,"created":"2023-01-03 09:59:58.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/63/869063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/869063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/869063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/869063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/869063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/869063/1"}}},"commit":{"parents":[{"commit":"b3d1412728b192eb29f36b0440f72fe3e5ecef75","subject":"Temporary pin tox\u003c4 until projects are fixed for tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3d1412728b192eb29f36b0440f72fe3e5ecef75"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2023-01-03 09:54:29.000000000","tz":60},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2023-01-03 09:58:21.000000000","tz":60},"subject":"Adding newly released tripleo-ansible into requirements.","message":"Adding newly released tripleo-ansible into requirements.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I99c20cf35d6e832a8b341d431a58dff25118b4da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57f5f705e84289ba9fcf8fdd479b555bfd7b4168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57f5f705e84289ba9fcf8fdd479b555bfd7b4168"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
