)]}'
{"id":"openstack%2Frequirements~869334","triplet_id":"openstack%2Frequirements~master~I4453bbec1143af6c9947ab82eceed6b2e2906427","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4453bbec1143af6c9947ab82eceed6b2e2906427","subject":"update constraint for oslo.config to new release 9.1.0","status":"MERGED","created":"2023-01-05 10:15:47.000000000","updated":"2023-01-10 08:36:43.000000000","submitted":"2023-01-10 08:35:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869334-new-release","meta_rev_id":"bf98dc033e07ea54e44dae77b764909187a7cdaf","_number":869334,"virtual_id_number":869334,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 08:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-05 16:53:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-05 13:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-05 16:53:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 11:11:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-05 11:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-05 13:23:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-05 16:53:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bea6ee05aa0ed88cef7d8c4822201b9fb05ea9fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-05 10:15:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06d7cc27bb98dd550331f11bd91f0b14ee2bc1e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:11:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e56cb39b57d42b5bd182f1dd1cdfdcd\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/89c501126b2444309f96f87012a5fc16 : FAILURE in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c4b01e3940083f03d32b0b280684d24a5e55eeb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:35:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/987731fd9a264864b2bf17ce17ebdbaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e54cf10370470b82a66fe57c5dc316 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6783eee5a524a23b62c810b8fe2a986 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0848e188a80941a39bc308e6f26b5b36 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2217b353088e4ac5bf856c477233c1e0 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/752a8e9377e7400f85067966d8c68169 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59587164f88342149adb8436ff28b315 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59b5fda6dc38400db441c7b5e5a412fa : SUCCESS in 7m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5a73aaba211f4cf7b856a2077d9d61df : SUCCESS in 13m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5d614c285c51451a878893c3a74a48d4 : SUCCESS in 6m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4abc342b7de6468087d8f8d28ccb76d7 : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d32f26d6e72847338eff9f090af77cc2 : SUCCESS in 6m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/496d2d4df75f4a6ab8be13bd4e0c5d51 : SUCCESS in 9m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8298c56d58e64ee6a869875d65ddb670 : SUCCESS in 3m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4e1c30a65807437386fc21aecb846a2c : SUCCESS in 5m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b47bbde1934841c6b5634cd43e71aeb5 : SUCCESS in 8m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fb98ed039e4f4fca852b6480db77cc12 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef51eaac45e04b1fa4b480ce9b18980f : SUCCESS in 12m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/be45fa61306245fdb686148aa1f6015a : SUCCESS in 7m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6d546306f40545418a9f1e52b502de23 : SUCCESS in 11m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ade7a33a617f4d788d4b8ec2e78e8440 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cea8c51f56404471b5936458902c7494 : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9ada9e923cd24598969484140763c2f2 : SUCCESS in 5m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4176b9c3cd234be28656d981358318a2 : SUCCESS in 10m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c1ced53343a84f92ab196fd5583ad54b : SUCCESS in 32m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/14828bdb29bf459f8d0dd46946356a3f : SUCCESS in 11m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/38ce0f80656f4e76a30ccefa0d02b25e : SUCCESS in 7m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70176c222bdd402aa54a4d7044acd2c5 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3389233757d64cd2977dfd7764497483 : SUCCESS in 24m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bc1069b8503f447ea1b02c0ab28f5ee3 : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/44042872a22b45ef8902026688b3edf0 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/147dd9f58d194046ab22c9b537a016b7 : SUCCESS in 8m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/37f7d720535745259a4e390362d67063 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba769593705140bcb5943dd17bd50c93 : SUCCESS in 56m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bda4e895df524b1c850ba0121db348d8 : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5fba21fd386d6f1a27f672d882e880ab989879ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-05 13:23:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc5cf86e705180b751dc4f103314e041b44c9b68","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-05 16:53:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d80104f48868587c883d997c67edd18e139ace6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 16:54:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2664009649f569f7e95fa0251dfd85001f7a676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:29:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e31fa7aeee4e4b61981e7e9ee8548ac5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054928e52712474fb0ec185fc17437c7 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0a1373433e4621963cca1755ea839d : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbccfa67d3d8422fa1b6965ac1a712ba : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9905ddd439242f6b64f5e51eaadd72c : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc94d85a77584034b6ff55c6fceaeba3 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a166ef1c3d384fd18166f8aefceddfc7 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73461728d97d4982b18afeeda7e1c78d : SUCCESS in 11m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f997d05493c64bb58ab8e97160d03dc0 : SUCCESS in 16m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8d8e7c44ed5d447fb46131195957f89d : SUCCESS in 12m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6c7c1aa15f5d45a4b896d25cbf8ac427 : SUCCESS in 5m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e1ea400986a34474bd74f685463790c9 : SUCCESS in 5m 39s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2552b6741ba242fba19a856c542ef370 : SUCCESS in 8m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/58751648a862478ebc54f8dbf6da0cec : SUCCESS in 14m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22ab639497374def876fc0930d11b514 : SUCCESS in 8m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e6074b5860df4ffaba39f95b2a4d9c68 : SUCCESS in 9m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3c20e093c54f4c6db570e5e7a4ac1e9c : FAILURE in 12m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/98a263333a5b470d8a048752428fd1a7 : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6fd83166996f484f9f1c479e2b3494bb : SUCCESS in 13m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6fe2600021134383b03b565d78dd3b7c : SUCCESS in 5m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bc169035932f483e9b2fce6d888089fb : SUCCESS in 24m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f69fd3bd747b475f87a932a3743d7919 : SUCCESS in 14m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/09e28589116344ceb29bde2bc0a08cda : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86243252e7774babaa3d201347e2d0db : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fced16c6076744c9ba21af4defcbe463 : SUCCESS in 26m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/93f1d34b6c944b97835a9b9bb8bd4d49 : SUCCESS in 6m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b82b76632a544b638e878d4d59f085ea : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab3bb4b1ca14444ea09d014a6c580c21 : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3e7492f4f7af48cc971a7a587f7f4ebf : SUCCESS in 14m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98f9650437ba481d9a752498b4c23c27 : SUCCESS in 1h 12m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e0de0449b3d4b3ea37c4b1ef7d62426 : SUCCESS in 1h 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"297cfb03f3ac1c4fd0ad088750b75150394c998b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-06 03:18:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c6f9f73efc0e73eac7557fc8ff203c8baf8b8073","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 04:13:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf184ec5ef254cdf8de91725b2e02b5a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9ce89a9757214de79f2bc47571460b6b : FAILURE in 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"acff3887af1acfa11845c2e8813d37a5b9366fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 05:13:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aefb469295b7403fba2438ea1535aedd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaae5fd656fc4b17974fd5ebf0fc9d9e : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e09856faf044ee91ea69ce5b886c93 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd0c845ef7c2435cb6445d33ea46040b : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/075b58847b264d4e875a849055226a42 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d4ef22c22384d2ab4e4017d9c81e71a : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e99763ed32b548dcb2134fc0211d67a6 : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d08d799cdd0246a5901e8ade8d64588b : SUCCESS in 10m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/33bd00d35813452e845903a1aefa0cce : SUCCESS in 13m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f5069762fb794db88f9db955ece554e3 : SUCCESS in 7m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/36120bcdfe324aff9927e9e978227654 : SUCCESS in 6m 41s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/62567eb15bef40b29f2d94200801b45f : SUCCESS in 6m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8487b8579e244fe3a1b22c2ddec30638 : FAILURE in 4m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4127dc6cd3884c57963eb76b3c71b36a : SUCCESS in 5m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/28a97281c45c4fe0b5e201a98efea166 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5c4103fc8c12432d8b5c824af3364fde : SUCCESS in 10m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8ec6a64f8f674e60ab29c5ab7abd7ed9 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f4126d98cc543b2a6f82283dab40c7c : SUCCESS in 7m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/39b52bd7469a4dd7a2d6c431361c259c : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3e33b7f397a8496786fad8030eaac544 : SUCCESS in 11m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/64e8c34edd7540fb88a5604b7c4ab970 : SUCCESS in 4m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/86167936b9da4d0180e8f9ecd985241b : SUCCESS in 10m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ffec26a58fe44aeb8c37c964f1567ec1 : SUCCESS in 4m 08s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a00900ae958d4034b70746396029e4e3 : SUCCESS in 10m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1559c12e56a14ef8b8dbc772393e3108 : SUCCESS in 26m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/95d1c80551884127aa96acbf4b641e18 : SUCCESS in 14m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/190e27b6385244de9896f3d9e5c06c07 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b11442cf2683495cbea216b06ffa9d69 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/daa144f6375d4078a2dc06979e95d2ec : SUCCESS in 27m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/245856d6c6ce4e1390d3c15822bef443 : SUCCESS in 5m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7a4dc90d0025441f9313f30232ca6ef9 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a49521a617ac4f608afe338648bdd444 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b6a1bac97db04d06aba913a8bca86214 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f7a73c8c0e64bf1ac4e9a5f3d500686 : SUCCESS in 1h 06m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a1ecccaf6df4d57abb2cf8d7bf91387 : SUCCESS in 1h 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fa4900ff389df83c0de8be4c8468fdf5f2711263","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-09 22:55:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64ab1e1ed8af0d41713e058bafd2070edc00774b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 23:49:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b21a41fbc5fe4910a6d2ee3b8745033b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cb5d944ffa9e437aaa60f5ecc52c6bcb : FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5c91d61f5be78e6e99dad98cbef703a5a608ade6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:32:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d438ad3918431abb99e1f1787e0baf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb952fde6974889a74b75925a068449 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4883a293d74c4339bfe90de6596b6259 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2850a77aad0049059d742bc3540d3c57 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cd91fcb29fc489284a6c33c698e986e : SUCCESS in 2m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e989595fb43a4a7b82f8531af104cb85 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b9bd4e1b20f4e40bcaebf28a437d4aa : SUCCESS in 2m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/55948dc3a7a8477faae3b841244130c1 : SUCCESS in 7m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2cb4f5fc145e4c83bf89e645aa7ec92b : SUCCESS in 11m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bdf360813baf43b498463cfd566d225c : SUCCESS in 6m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f6f688d7ae2c481b8836533ab1ec5e7a : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/968ac04ac0a04c6583b72b573ba253dd : SUCCESS in 5m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a92704233b1540b7a47aa4b2ded48a1a : SUCCESS in 9m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/eb450580d3ea4e5c926cf6cde14dced6 : SUCCESS in 3m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4d2facafc8f248788e0cc07648aab093 : SUCCESS in 6m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/17d46e0d59854abc98efb7279bdcac48 : SUCCESS in 7m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/83979fb663b74186ad32a25f472ea4be : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4956919414f545a88a9faea57649a554 : SUCCESS in 8m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0bcd8ea24fde46e6b73edf586faf8969 : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/11dae3262538486480a254b4899f1330 : SUCCESS in 11m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/332d0279e72d49bf85aafabc8970e0ce : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/33e74446140d44408bc0b942c078335e : SUCCESS in 9m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2260cd43216d4daeb91bbd1613d47a6d : SUCCESS in 3m 56s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fdf2f81a834243fca4cf341d6ff729fc : SUCCESS in 10m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d35cd57dad9844e5b298f1bd8e679267 : SUCCESS in 26m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1d6cea8f8fb246e0ab22ead624057073 : SUCCESS in 11m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bb88e94323a349248b51de7d4e4d1bf2 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29285186c05f4059876979041ef1388f : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/810e2bd72f5b41028fdf76717e2e8d67 : SUCCESS in 23m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f3042693f2e745a885b2a4285c378fbd : SUCCESS in 4m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bfbc056f9cf0486485b34d7d582e4bbc : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/788038622ba748b5b68a1cda86509063 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3d619e3da0b84b6981271017948c85d6 : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/501e1d73e6c948a2ae33c1a6b788b2cc : SUCCESS in 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8ecd2b5b0d040f6abc7cde0ded4e927 : SUCCESS in 1h 34m 04s","accounts_in_message":[],"_revision_number":1},{"id":"47a9807e01fbe2a15fed5354ad4ccd1484f5acd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:32:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b8970c9eeb801fecfe274245fb695a5613a6487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 02:16:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bee1b7241f64754b9bdf96d04873d87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f969881b4c430b8be9be100050d504 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb5fccad5f374d13a87aaa89bf2ceb1c : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b66cfe98ae464c02a772136d6e48c4e0 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11b27f48a6584b06be69778eefd0a760 : SUCCESS in 2m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e30265c7eca4875b67c7f84d070b8b5 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddda2373e339446898d3c26c823a9c7e : SUCCESS in 2m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a5397ddeffeb4703972bd04b1655812e : SUCCESS in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/070fc758c60c406d9aa0696711785f3f : SUCCESS in 12m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/761880e487024219970c76f460eb5805 : SUCCESS in 9m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3abde6a8abc24351939a678804809d23 : SUCCESS in 4m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/125853c5b1e44586809bd76db1a5fda7 : SUCCESS in 4m 39s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cfbb14d83d2e486bba769dcd3e37a595 : SUCCESS in 8m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/48d1ed42b448480da6639b28ee73a456 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3dd689eb9dc044deb71fd34c6f6dbaba : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ce373c22e1a648bf921bd024ae1d90fc : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c19d8bf449d74e3b93f6d3beea58bae0 : SUCCESS in 10m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/99f35c1315904bd4a91ea808c5bd40c6 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b8cad3fad27e4395bc061520a2de985f : SUCCESS in 9m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/570ca94dae034b2fa5d83d5ad77d0e8c : SUCCESS in 4m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f804860cd3804fb69adc81f15840106e : SUCCESS in 25m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f37a20dcda78400aa5cdeb199e436053 : SUCCESS in 14m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2572deda054b4b5697f1d360070631f3 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6764e9e68534cafbb2f132f4e21befc : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3fd5a8ca7284a0ab2b280a91195a479 : SUCCESS in 25m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c2fcc2bc253d42b8bbbda39ab0dc4544 : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cee3f3e9191f4cd5b8374c45a815c94b : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86ab492b66564a908e79ef1919a100e6 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/83ccc0ad82004110874f2ed2622c043e : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cc82d7321e3496aa500793ac72bc964 : SUCCESS in 1h 06m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de459d7f7a0041698f520e691b6da697 : FAILURE in 1h 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8ed39134e14c93c790f72782dd488fa1b3ed6477","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-10 05:46:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1350d97f6c781a04af663e5cafe784bca9e66b83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 06:41:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d4d0198614460b94f257f5cd542f2d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bd3fb0eb42c24dd69f86fd8cd1c4cdfa : FAILURE in 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"aa669017a13d8e80683f632e866fcd95175e024d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 07:11:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5f398c419364583a6906f49349ba041\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ce27642a8e4f07bd25c48e69351be5 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5adfce07ae4bccb630cb3ba41b465d : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ef68190fddb41048d0d2923adae9b26 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fad7166e3ee4eadaef6914ac0fdcacc : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d793e7ef94004248aefd92d644c99932 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8f1078c3be44e3ba92c2a58499cb589 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f22586c4a884da4ab3a091b42143dad : SUCCESS in 11m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a417266d749493e8a09eebb389cd814 : SUCCESS in 14m 46s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/255129b7b6a84649bbf0cb7852071e30 : SUCCESS in 6m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9b66825d5fa04a43ba868e4497a1c962 : SUCCESS in 5m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/09a2bd8ea28d4997bc174f93a134d331 : SUCCESS in 5m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d814f53705104c328a2cf576392209eb : SUCCESS in 8m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8ecadac4bae9456eae07c281a7f8162a : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f1f95879b220406fb416a8e61a536750 : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f30f4d97b466473480dd970b4aebde79 : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8502963fd45f4b57b16916f8b458b784 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0f4e1ca393c4a38a4b8007da2015d92 : SUCCESS in 22m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d13e8bcb0479451491854f518cd9b73a : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0395157c0b34429a8f22b341a695d1f1 : SUCCESS in 8m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/add749e6c5ec4f6f83399df58ff60cbd : SUCCESS in 3m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d9dd274800f540efb5a4f38fb0307da4 : SUCCESS in 7m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/76ef1a5772b744e894ba85dcf3fb9e1d : SUCCESS in 4m 44s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0b51b2bca15d442db8410bc08af7f638 : SUCCESS in 9m 44s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e7a344b1e2994883b0cadd3fc3f7ba33 : SUCCESS in 27m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dac11fc4c6af495fbeae58d69f647dc0 : SUCCESS in 12m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/047c68d768284c4aae4b04967c6f9e43 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8806eb5a7e86403a82570df926d6c1ce : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79d04b55cdae4dbe8ee0ec42cde190c6 : SUCCESS in 21m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3d4c572c27404249bf5f1d3e3ac52e3d : SUCCESS in 5m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4df3f90891284a0298cd17ac0800dc12 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/718417eb5ed74881953f214eee69cbff : SUCCESS in 5m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a4aec038acb045788751e21712b521f8 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d541580cc1a24546a34a418b32de037d : SUCCESS in 54m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9985777d686424da716b0435ecca317 : SUCCESS in 1h 15m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8ab9b6cc0867806a5a78e70e27b2c57a3662fc01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 07:12:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5c2ce2eba0fb13602da60f08766950bbc6066b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 08:35:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a30215214c74f818941d0f24d90035e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a13595f7034a61817a85405394c7f1 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7093cf484c45978d08df7af3efc148 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9edf2e2225db4081bd28930365a23452 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/505b5b24377c415d885365a15191969d : SUCCESS in 2m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d404402c6f8240509dd367b0ec274088 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed176bbfd5de4209a19d84f0c5b899ba : SUCCESS in 2m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c535455f57904d188b220857cbe9667f : SUCCESS in 7m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8580145ba9e142519d739d608356fab9 : SUCCESS in 11m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/09ac931388d64b6382c24fa2aaed9b74 : SUCCESS in 9m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ccc6c89361844284860feafcce3a706d : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6a07bb68ee6344b990438308c0ba71fc : SUCCESS in 4m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b5e3292aedfa4a039b021bfe04615319 : SUCCESS in 7m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ca47c0528cc449bd895d03de5dbafb77 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec1b64d90a6043f392cb5ce76db88b9a : SUCCESS in 33m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3147928cebfe4338af8afde8b13ec7fa : SUCCESS in 7m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d3926eca38d64216bf79ede66f66fb50 : SUCCESS in 10m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c88bdfe69d85419098c00c5bfc1edaba : SUCCESS in 3m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7e9d19cc4f66483e8d2a4ed475f8fcf5 : SUCCESS in 9m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3739f342bc604fbe99e43b39670ea6a2 : SUCCESS in 3m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e9d0dfe9df6847678abd73ba54dddf8c : SUCCESS in 24m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/99bd489232cc4087a05c33baa4bd7e7e : SUCCESS in 11m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/366d7993127c4fcca1253b8cbb8a3a98 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0fd9af5220334c418a7fa18849ca78c3 : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eedfb6b399f347c8ac5934b42160cb0b : SUCCESS in 19m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c9a5c8469bbe4d46967d726b12483113 : SUCCESS in 5m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dba8674d815341fdac3591f119de3cb6 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37d9e763a0ea4f479ca2838e4569de21 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3fb44aa458314d7190cb7b988fdea541 : SUCCESS in 8m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba95ea58819a4d0ca0b65c734f413541 : SUCCESS in 53m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bcf0a5996f442abb25aba5ab8245795 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e22b42b2297ebcb797862dc5d1c2ca8160efad54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 08:35:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf98dc033e07ea54e44dae77b764909187a7cdaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 08:36:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c4a843745ad406da6abf93c27ef3c7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b0d5bbceaa490a891bda75d49b3abe : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8d62044e276e3915e31cab4a8f8085c3b55b1ce","revisions":{"c8d62044e276e3915e31cab4a8f8085c3b55b1ce":{"kind":"REWORK","_number":1,"created":"2023-01-05 10:15:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/869334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/869334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/869334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/869334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/869334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/869334/1"}}},"commit":{"parents":[{"commit":"27174c1170d666bf35de05c56547d3e25551f3aa","subject":"update constraint for keystoneauth1 to new release 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27174c1170d666bf35de05c56547d3e25551f3aa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:15:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:15:46.000000000","tz":0},"subject":"update constraint for oslo.config to new release 9.1.0","message":"update constraint for oslo.config to new release 9.1.0\n\nmeta: version: 9.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I266d0cf9440aab0db6fff74a5a3130f9eaa1def6\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4453bbec1143af6c9947ab82eceed6b2e2906427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8d62044e276e3915e31cab4a8f8085c3b55b1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8d62044e276e3915e31cab4a8f8085c3b55b1ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
