)]}'
{"id":"openstack%2Frequirements~869335","triplet_id":"openstack%2Frequirements~master~Ib747cc54b7a5f4aea07b62af3deab02fce56f6c3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib747cc54b7a5f4aea07b62af3deab02fce56f6c3","subject":"update constraint for oslo.db to new release 12.3.0","status":"MERGED","created":"2023-01-05 10:16:08.000000000","updated":"2023-01-10 03:32:31.000000000","submitted":"2023-01-10 03:30:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869335-new-release","meta_rev_id":"455074639ff854802590e1a3f6815ac5eb2d0baf","_number":869335,"virtual_id_number":869335,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 03:30:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-05 16:53:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-05 13:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-05 16:53:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 11:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-05 11:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-05 13:22:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-05 16:53:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1e8cf2ec2d53dc8b20f19ef7742c619d19ae25cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-05 10:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de1f04bb467766946f73f1b5a77aae1334197535","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:11:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6e0e4c676f4fc095a4810cb69abfa5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/64bf0fcfaad04797b2efa435be510123 : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1d02f23a7eefbcdac44c694a80a335511184868b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:55:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77db758515e4e8c83d3514ba3d88c00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b888ffc56f3f45d5b60490dec49c9d0d : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75aa4e53c50d4251b43306ba9be96fbf : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26cdd7ae22864a4494e01a50f2a8677e : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8af1b13047f4232b4b95ae4ca604892 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af778aa766994a3b8453298fd090e2e5 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b0de4e0100347ffb322af2d0494393a : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f93d27baab0946bcbcbb23e50b5cd8cb : SUCCESS in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4be15739766d4168aacc48728e6cdbb6 : SUCCESS in 11m 25s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bef5c8966af54173a929f1166cd8be2e : SUCCESS in 7m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/73cb95543b3c421baab5c3d1aaadb0b2 : SUCCESS in 6m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6d412e7d8abe457cb0b4dee7ef3b80ba : SUCCESS in 4m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/669c4582efcf47dea706189dff597280 : SUCCESS in 9m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e4db0c98c83e4e558467e720b5f8fbba : SUCCESS in 5m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ddda4c810e6842e8a980d8383483528a : SUCCESS in 4m 35s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/58a9aa12a73b471ca32b4c3693a2f75e : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/543431d1967a4418af862a8eb1e88959 : SUCCESS in 13m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac2acd1735f8430fa003aaa5727fbfe8 : SUCCESS in 7m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9e6a0530376f47a4a24ca7bb8235a73f : SUCCESS in 12m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/aa650fd69b2d4a39b05cc624536ee674 : SUCCESS in 15m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dd19c04acb004a10ac059eb471d4c71a : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bbbdd61947e84db89f5058da0465e522 : SUCCESS in 9m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5110a93c6b1c4dc1b740098df43ed0f0 : SUCCESS in 5m 11s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2840e60c51374f52805d68a3e1199136 : SUCCESS in 10m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/93ced585ff3147a69ab77b1abbe1fb35 : SUCCESS in 29m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c6d3cfa247b2446fab21b3187c67801a : SUCCESS in 11m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5a4d0362836345f7bc9eba02d6b30d19 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f81b060ae3e842b8940527496b139c89 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e15ee485c65a4a39811d7df02d6d869b : SUCCESS in 23m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d005738dffad45dbab70ab4b2b96042b : SUCCESS in 5m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/19b3f3f56bf844e58e62f0b602bd25ec : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03a793049cae482dab5333607aaf547a : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2c1696d918d44b6188f5b98c377c1feb : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e640417f97f4be49dd2edf89fec0241 : SUCCESS in 1h 24m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/205ad268ea194aa183f054f6d4561b47 : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e9d032125913788dac4de753da996fa26cca9f70","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-05 13:22:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23e1ad90072abd2029c1b3dbc5d563989e039280","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-05 16:53:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eebb8f428ee1d8e9387157a102a6fdfb470d1c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 16:54:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"993bd6537f5c89efcf65cf79cb56e6dac84d4fd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:46:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e2c1d6d64424de2bd45cdfc71e7da8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84d0645566b43c7a88a237064f18ba9 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/262912fe3b234338bec25c7dbbc54175 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e70a92790a514fe781be45320d5f356b : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79cf929a896b413c8c5b7be4e121357a : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a683bf6a496e48c1bd58fcb0660bdac8 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc80f3518b404263880aeed32a6b9155 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64212ec2535149acb05757821d8395df : SUCCESS in 10m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/12455a215c5647f19cc4516d3577ac7f : SUCCESS in 12m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f253db99c99d42d9830dddc9eb64ee4f : FAILURE in 3m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/12099c1b42b246688e9ba256ef761846 : SUCCESS in 5m 36s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2141ce69ce4d4b90b2f94d331af95e7b : SUCCESS in 4m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7713403f61c04b3aa03fa8c129a613ca : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/894670cf654f4906b0c1263f350dd5eb : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1780c5848584432294fa30068972fb78 : SUCCESS in 8m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/30757e77c45d4bf2a11d8fa9fe8418d4 : SUCCESS in 10m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e46c07e6a95d418082e92d165656cd0d : SUCCESS in 17m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7e8b5eb6a274474080d6b0852c829d01 : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d09451d710de4cf884100a500241587d : FAILURE in 17m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/69349e351b9e4c9cbf62c35f001d92f5 : SUCCESS in 5m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f1bbe4745e494c69be5e4b13f1ff79dc : SUCCESS in 24m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8e93be663b4e4fe08b1ff10044791ee6 : SUCCESS in 14m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ed8e53dccacf4e78ab0ee5369aa0389c : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40a7a6305226426d8b288d30795c66c2 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7dc3aea26b0a4e25a2d31391e9cd28b0 : SUCCESS in 21m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d96fa0ac0d6649cc88789aac85906025 : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d546ae2149774429927d2622cebdaaf6 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e918b5da4724a0f967614dbde99f246 : SUCCESS in 6m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/82693c8ec34e477c8345a5157a841c60 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c89d6d94df84c9390c9e2cf02e0683b : SUCCESS in 49m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4426325b46b0431dbd8003dcc6a0bd5a : SUCCESS in 1h 51m 17s","accounts_in_message":[],"_revision_number":1},{"id":"461760df2235385dcb9ca424f70e82e811481fc3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-06 03:18:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ceddba59b7b72a95297a35802a21a8986c1b32b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 04:13:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/125a09916a7d4b998dd65f97e5f50d87\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/44fad2132c634af4bdecda44a025a134 : FAILURE in 12m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b9bdbaa5e56d44c28353f2b673861b188e2c1c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 04:47:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/954850c7a799408d91c5d4421965ae09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93fc9cb502a4739a804fb3a91c14219 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679f3a48d66f44ae9c65020cb27fd2ef : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb6380cacd9a40d791fca3ecaf12c66c : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a43e75d8527240a48160e9cb36e04144 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c8dd274995545abb634d909c2f3d760 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6374eccc0aa44e8d8a8acff7c655a15e : SUCCESS in 3m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4fc16856dc584e539cc06edcbe681b64 : SUCCESS in 8m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9bebfcffd0c14c16a1a03fb15a703fa8 : SUCCESS in 13m 06s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1d3714e518df438c87da9dbd8e8b2051 : SUCCESS in 9m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/af8eaa3a008c455780d7000ae7a0d52d : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c037ec42c04f4505915b7cde599312cf : SUCCESS in 5m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d7c7829f0efb4070b0ac5d36004331e9 : FAILURE in 4m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/38d04089465f4d468928ee95367c5df5 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/29ae1444ae9d4634ac0e96489958483f : SUCCESS in 6m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/44b094e29d9e4200899efc60cf89f2d3 : SUCCESS in 9m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5f31ed3fa35646e5becede71f8815a31 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e74b64a71acb4f12ac10e2d8c6537bef : SUCCESS in 7m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/49c2baed5cba4e3385ec2210db567ba9 : SUCCESS in 8m 05s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/85f20f2eb38c4edcab078022e8ac3ece : SUCCESS in 12m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/991dd74283734fad8d5f6fe73f20c67d : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/93b10ffaa38649099a7a73fedb8550d7 : SUCCESS in 9m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2988edeaf5cd48f2b67a3ccc4565b584 : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2a0754f2997c4a7b872cd72871198a7a : SUCCESS in 10m 40s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/69605352b9b543328339ddb8e993a7bd : SUCCESS in 38m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2e6b12da29fe4676a07fb96f544388ba : SUCCESS in 14m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c4e4400bdf0d471290ef2f06e199896f : SUCCESS in 6m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4704038c1ec34cdf900a468f9fa8b30e : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc84778e52e14a328988714ad3c1cf11 : SUCCESS in 24m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/807ffcbbb2f6483abbd0abe662f75c06 : SUCCESS in 4m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e919816da9e2481caca1a192a003dba1 : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68db59a48f244a1baf3c9f1eebe1086a : SUCCESS in 6m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/220888083aab40d797874f223e309b35 : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96683d51556b479daa09038b3ae9816f : SUCCESS in 42m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58db7e642f174d87ae7fc0045bb29703 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e131d59886e245a13aaf1863b4279db74562ef68","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-09 22:55:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bddf75d9ee015153d9d0c8a0b785ddfca4665b32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 23:49:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea913e348abb4c5b8b224128fb65decd\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b47228c949004e6180f0c62c57de4663 : FAILURE in 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d40bd994a86418c704101052fb85dcbb34e0fb6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77f8a001fe2c4698b9c13bc4c242e1ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afdef6d1ccfc407db443792507e1f4e5 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e669e5f77574f648ed706b006908a95 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5015f401d2df4105b8328d56f4693841 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d27f197704124866b4a72bad8c7d3098 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07d4cf17e97d445bab1f703bf3d9607a : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9e85e5d1ca4460e870662769b63141e : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a4fda3a7e3aa49af8d1f5acce01d75a5 : SUCCESS in 11m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/20f98dfb6a6f437aa6077306098496b5 : SUCCESS in 20m 47s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d8abc82049c94fc9ac54f5063d085701 : SUCCESS in 6m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3ff125aa31a047d49c2fde38b9c15e50 : SUCCESS in 5m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b4c050c9a11d4ed690aecd266768aefa : SUCCESS in 5m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/dc7f9514f86d4d6abf4d6b1e0e256d02 : SUCCESS in 11m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1c00d4bbb2564f03a1396a485e58662c : SUCCESS in 3m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8e6f1a7d380c41968e677d16659381e9 : SUCCESS in 4m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9761f583c0774e7d96e042e3685b2d81 : SUCCESS in 7m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bfb716a7766a4b21b310bb15dada3931 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d38bf534ba91449d9893b3f12c2708ff : SUCCESS in 7m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c63bd9d6011c4fefa71cf567c00697ae : SUCCESS in 6m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9585cd8c04e443fd901ef5d6fa4676d8 : SUCCESS in 11m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7cc25c54ce214ba7a099d2a159731220 : SUCCESS in 4m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cd3e1e1cc69647d380075395c1c8c479 : SUCCESS in 7m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fe4621834796409ab35782ea0e43cfbd : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ff0c05e65f1a46c8bfb1494a91c8bbf1 : SUCCESS in 10m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/77a35f3ee8854d74bf24660a74d3ad52 : SUCCESS in 21m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5b9a714953d442da8bf4539495cd81f4 : SUCCESS in 14m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ddbb2cda590c49eb97e1db74362eb05d : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/911d41347c3c424dbbf20252bc693b1a : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08ac4354e7e24453a767a493649893ce : SUCCESS in 18m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/db943fcc237c44c48c7a7d1be3a752bc : SUCCESS in 4m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8e431122504844d2a1ef5201a5633291 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f44d5de493ec4288b3ecbf452288567f : SUCCESS in 6m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/18221239c7864f63b61e5640f7ec20cb : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c2b9de6fddb433ba4c13e958590bef7 : SUCCESS in 50m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/930e9cba06c64ddab543090d264f1813 : SUCCESS in 1h 42m 31s","accounts_in_message":[],"_revision_number":1},{"id":"969151fb0a544b2c20a976f074e81ae21f20fc36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:44:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a95c421f8de26c71dfaba526a65489bbf7257da3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:30:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e887c25226c4a4d926e9079009f8e9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab385c11961463f98df2252877190da : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07b243273804c0194357d08cd7b4af5 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3de545605af34cf991784c510e1adeb6 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25f8a9d21a2b4e89a1b23361eef95066 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1913f2fd49be4a84a76eeaced4944556 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4487cbb71da647f4823084ed73668c4f : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2062721758974e85b7c451c3180f03bc : SUCCESS in 6m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5378c0d3ca1647dcb927b22f99214e85 : SUCCESS in 10m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/08c22ec7526643b28fc93daf5507bddc : SUCCESS in 9m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4216ca94007d45b1a60d26aea7bcfdbd : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/50283c24362f41ada8d7c89dfb081c71 : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c2196f3a67714ae796c5a6be70b4003b : SUCCESS in 6m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/288f9bdf8557422182d2a0380a525f68 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46a3dc2c624c46898f7097ac9032aaf9 : SUCCESS in 7m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9d1ac9df742c44a78471f64cec4b428b : SUCCESS in 6m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/50bb5edc1b6f4cbea23bf1a5e5ee239d : SUCCESS in 9m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3949d4d4eec2474dad036db7709c7624 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f1aa9554832a4f40bcbfbb53ff4c9890 : SUCCESS in 8m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/81a86699d9e34c0d85561017fcee6f88 : SUCCESS in 4m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5756dc54cbc34be8afd853c81480747a : SUCCESS in 23m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c094c8cde6c64a9884d45154aa3401a4 : SUCCESS in 11m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/77d1472e91dd42c9b34f2d34a9dd1372 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e7bdf726289e411ebd050513e45b286b : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cba3641aeef441f9728488b14a7bb6c : SUCCESS in 22m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/30367500dd93465098affe655d0d1915 : SUCCESS in 3m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cbd1e09efd5a4718ad728f555e1c6be4 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52ca8615860b427fad9d8d4e0c8dc684 : SUCCESS in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0f08a508346744e39636277cba407fcd : SUCCESS in 9m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45895f3382d14e739dc494746ebf3ee5 : SUCCESS in 53m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89a71aad01e14c0aaa5988249c694390 : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6888187650fe38681f8090b82532e4582d3c10ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:30:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"455074639ff854802590e1a3f6815ac5eb2d0baf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:32:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2812c3cb9674ab9a8a35f92f6f0d99b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079394590af344bbbc5d8675380cbd64 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06d48115e9f48988747bcf224664a4ca69fd4458","revisions":{"06d48115e9f48988747bcf224664a4ca69fd4458":{"kind":"REWORK","_number":1,"created":"2023-01-05 10:16:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/869335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/869335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/869335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/869335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/869335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/869335/1"}}},"commit":{"parents":[{"commit":"27174c1170d666bf35de05c56547d3e25551f3aa","subject":"update constraint for keystoneauth1 to new release 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27174c1170d666bf35de05c56547d3e25551f3aa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:16:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:16:07.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.3.0","message":"update constraint for oslo.db to new release 12.3.0\n\nmeta: version: 12.3.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I3a248d44d9e2340ce3591a4825a6db312934b247\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib747cc54b7a5f4aea07b62af3deab02fce56f6c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06d48115e9f48988747bcf224664a4ca69fd4458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06d48115e9f48988747bcf224664a4ca69fd4458"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
