)]}'
{"id":"openstack%2Frequirements~869336","triplet_id":"openstack%2Frequirements~master~I022777a80ae27c531897a00216af439fa0b143da","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I022777a80ae27c531897a00216af439fa0b143da","subject":"update constraint for taskflow to new release 5.1.0","status":"MERGED","created":"2023-01-05 10:18:33.000000000","updated":"2023-01-10 03:31:41.000000000","submitted":"2023-01-10 03:30:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869336-new-release","meta_rev_id":"3de507a428b25906a9e2e617f00b7f0fdac4f141","_number":869336,"virtual_id_number":869336,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 03:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-05 16:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-05 13:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-05 16:53:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 11:12:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-05 12:02:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-05 13:22:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-05 16:53:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5c6df827a48b5263a940a0e281ddfb386450fd34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-05 10:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"268dbad874bb9b6e360505a79e361c83815927d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:12:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad255960d88433e9ba0e25a082f6593\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/95ee66900f2843c8a9b598af1d2da474 : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7cc1e3158a191358c23d5386a383621f8432f447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 12:02:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3501abc403f041b482858b6b99be4f41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a9d2ce5e4f244a3b804690607e21caa : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2a4d3b625e4146a1eedc9031fea79a : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a8fbec9dafc4b85802c73d845035f1e : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abc8afc9fd57464c962074936ef77e11 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c08e8db859984db48f16dff6121ef106 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8096662d8f9840758f3314a0d57c632b : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/90c5a68e2962421eb4e19af325dceacd : SUCCESS in 8m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f6ab099f22bd4be694da3e5c4a70a158 : SUCCESS in 13m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/acbbee5ea81941e7aba791112c73421a : SUCCESS in 7m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2cf3f71db8ba4e64b3fa74d90c09ab25 : SUCCESS in 10m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2c40c042c82244f9ab0ba141ebd0ec8e : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/28326cc474ce4acd9e00199483fa1b40 : SUCCESS in 11m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/597e16cf3e92454485c0610e6bbff52b : SUCCESS in 4m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/658c234370bc48e880aec5d0954c9607 : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/61c26c1ba07a4a92a2ac255e51ee9650 : SUCCESS in 7m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7a7a8810190f4454b6a2acd266a076d9 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ecfae99229443dc90c89f2071458228 : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3bd619f302bf49ceb55e0920ea293889 : SUCCESS in 9m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8850457dd03e4601b9b20af2c5407b31 : SUCCESS in 13m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/166de19d9914472e924451e0d5f1e023 : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f0f3c5df3e304a62914c1ffae14c4083 : SUCCESS in 10m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bc910dcc96f34f3a851f46e8b651f0de : SUCCESS in 4m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/53599044b260425cae8fdfbc01277c27 : SUCCESS in 9m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cee8e03808524227bb3cdec92ac05009 : SUCCESS in 24m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3069cc9151dd4a24947fb616fbadaa07 : SUCCESS in 10m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a23cf5f7aea14b088606a05637d913a5 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b30a54034b149c1abcbf88cac040c43 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3acba6b24b0d46beb5c6f8516f8a0eb0 : SUCCESS in 24m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4e7c5124147d4619bcb7e7a74647f46d : SUCCESS in 5m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7a082fcd117f45b1b6791228e5483ac3 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66e2a4e9b85f480dab92cc273ee534db : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c382370cc3494eb2bda6a44c89209204 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f88c10b8daf409782c36c563feddebd : SUCCESS in 1h 04m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/609a16bb453b4f4bbb815374c7a4bb1a : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"30c1407d9dd99e6459c783906b0c45bc98cd8c53","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-05 13:22:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc52ab5a3b246a2906cf3ebf24b2b004a2a98ca5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-05 16:53:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66324ec296579998f0f6b08c0837d94be0f96009","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 16:54:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db5a011ffaabd10f3be4e8d3bff63a6ffdf55a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:59:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83071408140243749d3d168c5826a3c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7741997ac93e420da7c8abc2e626c103 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919795e695844eb5a62353f560b3a775 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74fa5d7d4e784f58bd42e5a87171cddf : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea16f79c3a2a45c19a65b515d4530673 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/261cbbebcdcc43d6996042b8cfadc6be : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d1575f2879e426ebdd386f8152dccfa : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a836f4c79cea45a88aeaeef46d4199ff : SUCCESS in 8m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7a563ef20f114cd78506762bf1098d3a : SUCCESS in 10m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7c5376dba0604e468459823beb18aa02 : FAILURE in 3m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1ed84ea50e86402b9c70752402df3201 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/074dc6602f2446ffb13b78b131c8be3e : SUCCESS in 5m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/30e49e7ffc2f4223997aee5a97573b6d : SUCCESS in 7m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0df2ebe685aa4c83b271bb64109111c5 : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95bb6fa35c874b5999fb7e7abb18ddd6 : SUCCESS in 8m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/244069fa5ba04f3e92b31152b8822ada : SUCCESS in 8m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f61707e08a0449bf95a096dd2a1e703f : SUCCESS in 11m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/26f02a4d6edd4794819589c171c86d8e : SUCCESS in 3m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fbf846b5b90a4161846a365a1e2a3004 : SUCCESS in 15m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/809601c9eddc49a7a88eda8ab070ee71 : SUCCESS in 4m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/46ee9f867ef14bd49cfe96d37d2b767d : SUCCESS in 27m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a4e8acdf3fa34e0b9f52051f2a630ecc : SUCCESS in 13m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ab2d6acc0d6c4dceb0a18a261b5a4df3 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/341e7370c760432a8d5f71857fa854fb : SUCCESS in 2m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9887af5682a64ab9af98669db89bad1a : SUCCESS in 21m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/73bafb471cb44ff6b599ed11010db3f5 : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6ba0c382da33456da127a713244e55e0 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea1f9a75fd7e40f58d125ebc13180f2a : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3af7b6c4207c4360935553174b3de0b4 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/789163da88ee4842a4f5853f2dc44b72 : SUCCESS in 46m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d0ea1a085734a368fd82f55c90585b2 : SUCCESS in 1h 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"940457fccf22899c1d21ae93c05236ccbfffc0b7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-06 03:18:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92c52d1493b9a0f0001219db40ed058144edae38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 04:19:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/728991bafc5345d7989c7e38fdfd29d0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7a9d37620db442b1aa5af9e0a37cd34d : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7a3864af410d705e01aac3a2b9088e26b78ce7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 05:26:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/724151ec79a346e4a2d60fac2fd84a62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc5896e35144ebe8a0271a459111b90 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6780344ac2451f9c86ec90b5b47c10 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2076a9bc9a249edbe085de3d7f74f65 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50c323cc7710478d9e6dc09476851cdc : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e1680f88c164acaaf51a9139fc01e76 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45de95faa28c4e1193bc589fc2b4e404 : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3fb6890682854b0cbbd0513bbf06ac9c : SUCCESS in 8m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bdbd7798f7ba42458041b7e0a1b5f5d0 : SUCCESS in 11m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/afcfdac76aca42fb9907b1b0a1368d43 : SUCCESS in 5m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ffdec57babbe446a9863eda90bf1c433 : SUCCESS in 5m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1b9284da863a427e817d54710f5b55dd : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/92af76987fb841478c2a2132f93aabe7 : FAILURE in 3m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1f81bbfa189c46fb9b56f07b5f6df381 : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/72898cebae2343a7bf1dc675ee933e13 : SUCCESS in 6m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0e67387cbe79431e87de0ed907552d82 : SUCCESS in 7m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7d6c7afb1d3246f59c471059d589733e : SUCCESS in 14m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc051940d67d41688b88e12a70d2597b : SUCCESS in 8m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/eab2b2997d9148be966339457a260f85 : SUCCESS in 8m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/56c16c565ec3432493df509ac03c952d : SUCCESS in 10m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6aa16a13a4ba4c25a0ab04f3f21412ae : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7f0234abb47d469f9cbcd8ab5809504d : SUCCESS in 10m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/23bb108c7501435d89d51d53ce72b55a : SUCCESS in 4m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4b83553243424dc6b449079b476443a4 : SUCCESS in 8m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bb9252a55ae54da59fb62c2cc5dd3637 : SUCCESS in 27m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1e5481764c8e4b32853e73c2274218d1 : SUCCESS in 12m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/92ac8d483bff49adb17bcdaeeb0dc069 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b58931ae514e447c9ff353ad4825126a : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/668405326eef429e84be1b823c86a0cc : SUCCESS in 20m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/530d7b00d29b4339ad9ffede94697fd4 : SUCCESS in 6m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3e42a25f1793476382c2ee9ddcbe554e : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a936880abc94db7be790af465e36f2f : SUCCESS in 6m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7feee887cff943e49f336f63e7e26652 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4ee7d9c852a44ce97b1b5b963f95bdd : SUCCESS in 50m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf04beb6020845e0983bd37927993e52 : SUCCESS in 1h 57m 36s","accounts_in_message":[],"_revision_number":1},{"id":"389cba19db1e6ec890044f905dbf63c618478fda","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-09 22:55:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c68fcb41f0dbd4c2fa8dd9f5cfd4152c1c26d456","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 23:49:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e1e182854a4859a6106e2057e79b77\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b4c28c5162874aa3bc575b2e05d7b7b7 : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b4d65ee7651f3e6b7a388f7976983d69fd051284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:34:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f6d5bc11e9744438efef3b7a9fecbb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010b7e81310248ab8615af80530cf223 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f03dfab9bbc64d019b5a53d63bcfaadd : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7254841d52ff4143ae0298277a982357 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/568ab37f044c4108863419e12dcf0418 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8990ec4348b4f22b010d5453a548690 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac0057600dd443bf94c5322b89d75f26 : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ded333420a9b420bb01f50a08be437a1 : SUCCESS in 16m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d9b681a128ce45088ff6feccbedef131 : SUCCESS in 11m 28s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/adc2d871567045d2ac1d9ca7639eebe2 : SUCCESS in 6m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5105c77fc1164694a624081e3848eb10 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0727a3b9c0b445e086a7ceac15d4a4f9 : SUCCESS in 6m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9c5ab0aa495f46b9959cb632d1abcd47 : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6495c8c555454ad8bdbd90fe5ce06217 : SUCCESS in 3m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/76a67b2b62e64d8aada7f26daceca07e : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/45df3ce146c448cebd0a1ad4da1d6e18 : SUCCESS in 6m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1b216b258ff944f2b3dac592b5af5def : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d91407a6665445ad8148e268ad6d4dfc : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f472f724a044439ea121e9fb0d0e1270 : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/236c5bfbd59f411d9a327da641b4cf11 : SUCCESS in 11m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5dff72c8e27b41e4b5d3f0074884a35c : SUCCESS in 4m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1622db0cb18f47aaa0658b54af38c689 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d71d6e09bfa443e79c35a00e4fbcc44a : SUCCESS in 4m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7643a5763ea1486ab61fecc8ac454bb1 : SUCCESS in 9m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3a2be2a681ed42c9a5633b3bcd3228aa : SUCCESS in 27m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6bf6ad28deec416394de5e919f7be3be : SUCCESS in 18m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c45cde41ae1f4a5a9ddbdef86fbdebaa : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2736c655f024ca8bf3d675709448ec8 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ebc4f58c9c8f4fd2ad0998fa1f5786a4 : SUCCESS in 24m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/909558e248254b1597e5a1e647ce618c : SUCCESS in 6m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2c4e8b1e46684ab19d163c095428cd95 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c33de77939f46b889de3bad480b12da : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2b9d0113dedb409497b70ff5163de346 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3642995da3aa4c87859aea495cfd9bdf : SUCCESS in 55m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b44f8eef3e1e4cd5b621294df7ccb896 : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ccb46873ed5f9ea116ba0112982a24f4f7e635c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:35:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30d9b5e12e0c9483250ba14c1b30808c2ba41379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:30:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b450660c1454fd28d5e8af279c3bfda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe8d34b4c0f47fba1bbb03d77008c7e : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c50662f9bcb4e7cbbde85ff427012a7 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97e7d5c1db9c4f21afefd80671fe85bc : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7866a2a7b3e84a768beaae7a50ab7a41 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0613ff23600741bb9cc532426b2f019b : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7df52ec1ce5426581532f2bdb7134b7 : SUCCESS in 2m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e3bbc9600d1452fac6bf6f3a0ad68b0 : SUCCESS in 7m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e9241237f36144cbb4c65b2c434d05c2 : SUCCESS in 12m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/28b841b1516f4c80ba48615100c5007a : SUCCESS in 9m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b31f61bdd0694b7ca41326e6d41261af : SUCCESS in 5m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fa2e13cfc58449a59e717ff456e5e23c : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/79666c744cc147b3976734e37e943683 : SUCCESS in 6m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4d30a1e6b35a4bd898313a0d7c262cbc : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5359babbd7fc476e88f517ae73b3ded7 : SUCCESS in 6m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8fbe2e63ed41490483f7faa5622076cf : SUCCESS in 7m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/81f0771ad5f84a758f5a35976d076434 : SUCCESS in 11m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7619fe2fde4145f78985778eae52cf11 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/53d9f0d9d04a43f395f93974cf0745c8 : SUCCESS in 9m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5e93c67e89334e6da6b34da49dd73edb : SUCCESS in 4m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dc6146b542b94f50ad980c7e524a94bd : SUCCESS in 26m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/84f9794845f34329960118cfbc9c8a88 : SUCCESS in 13m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8e89af8f59944c69ada1720bf7658fce : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/13d010ec7ac6496c81f47806371b8fdc : SUCCESS in 2m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bfd9b6ab212c43d5a2b9414ed4d68013 : SUCCESS in 21m 03s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cce44902999245ef87ba338519e1eca9 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/81a3bde7a5cf434f9f2cf4a86203a586 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/968a571cd1f94077a6a1f903f5f7e649 : SUCCESS in 5m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/56901e65898146309c02ed7c7e8f9dd0 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6745146e5b54376bcc968e3bda4d546 : SUCCESS in 54m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a846e48021a4cbba9d746e871f61ecb : SUCCESS in 1h 35m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1d9404acd64c1743a6057ec3d389e6bfd9f64722","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:30:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3de507a428b25906a9e2e617f00b7f0fdac4f141","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 03:31:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abd0ed8d9e4549ebb161bcfd7e096777\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4599646a0b9486bb53da1c49bfcb7f4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c024b134d82a108fb4d92420118398d8cd1adea0","revisions":{"c024b134d82a108fb4d92420118398d8cd1adea0":{"kind":"REWORK","_number":1,"created":"2023-01-05 10:18:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/869336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/869336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/869336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/869336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/869336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/869336/1"}}},"commit":{"parents":[{"commit":"27174c1170d666bf35de05c56547d3e25551f3aa","subject":"update constraint for keystoneauth1 to new release 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27174c1170d666bf35de05c56547d3e25551f3aa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:18:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:18:32.000000000","tz":0},"subject":"update constraint for taskflow to new release 5.1.0","message":"update constraint for taskflow to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I590a6ac9f6f1a340aa7740f3bc1322ce38730342\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I022777a80ae27c531897a00216af439fa0b143da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c024b134d82a108fb4d92420118398d8cd1adea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c024b134d82a108fb4d92420118398d8cd1adea0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
