)]}'
{"id":"openstack%2Frequirements~869338","triplet_id":"openstack%2Frequirements~master~I89033024d6fcaaa45815732dc2684dbedaf8efed","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I89033024d6fcaaa45815732dc2684dbedaf8efed","subject":"update constraint for glance_store to new release 4.2.0","status":"MERGED","created":"2023-01-05 10:20:52.000000000","updated":"2023-01-10 09:43:54.000000000","submitted":"2023-01-10 09:42:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869338-new-release","meta_rev_id":"48952472101af7fba492cd6e74cc7123e67b88af","_number":869338,"virtual_id_number":869338,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 09:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-05 16:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-05 13:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-05 16:53:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 10:36:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-05 11:51:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-05 13:22:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-05 16:53:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"12af30e42fc4af91522154ec44db3c304cec1a26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-05 10:20:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f374909a164d70f65054084753cf43e8994fa95b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 10:36:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d64b38a6eda4bbdb8f021f4fb30b707\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a9dca1add8d64e389eb508d00ebe09df : FAILURE in 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"108178a593d86bbc7c7287b851194044bd43d6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:51:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840dfe34227c42989bf127e23e48a80e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eff39cd29f14057be7d9c3141ce9f48 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf72590ad98d4c41922819a56c5b03eb : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cafc5b4ff5f54dcca681fd03b4d8577f : SUCCESS in 7m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d17178b2935468a8685d7f5f6491bc3 : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6adb5a55167849afad8b47a5390afba1 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e12ba1fa6d0b4a7d9feccf4cc3399750 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/718f459ca49648cfb1342e4a161c8764 : SUCCESS in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7ca70b9950b548a5b43746fdb23195e4 : SUCCESS in 12m 33s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/24b488cd93b34a8ebaa3707322d684be : SUCCESS in 6m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b53f44e6ddc74667825a8f31adbe7300 : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/68e4d924855f441c958681582d16db3c : SUCCESS in 7m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ae9dd2762a624b7a9f8036fb7ecfe56b : SUCCESS in 10m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2bb64bcff4904114b6d0de04c0cd20c8 : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ecd47578116c45308d5bf6ae56a45601 : SUCCESS in 4m 42s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9d7e3c4d9eb24057a071a3f9dc771776 : SUCCESS in 9m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/922de67f34284ae7a471cc8afdbd65cf : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3fef667af06e4ed69f4bc301dd387187 : SUCCESS in 7m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/593dd90f1dcf41c18289c08a612b7f90 : SUCCESS in 11m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6e09b8bc603743598641ad0c796e0292 : SUCCESS in 13m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/77f7ec776cb6407d9f83cb191791631b : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/252bc64b12d8474bb768720b1c7c31db : SUCCESS in 11m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ad515e173485439cb2a20bbe711e64eb : SUCCESS in 4m 33s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/db6fe35d35144d9c8e76b272d3dfb113 : SUCCESS in 10m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9b3a81f2f7d646dc81d73a677b789a78 : SUCCESS in 28m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/805057f3b5964fa68a1f1f58c3252a39 : SUCCESS in 10m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5b50e3258ad44790a568319b89b2d03c : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46d841fb56224c459cdd7cd1a31c9fa3 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7da5992f63764adba2d9164921093362 : SUCCESS in 20m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4efd2f13ef17475cb6a5e444bf04216d : SUCCESS in 4m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c59769c6fe9f43799e8477bed2d85bce : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f6debd6b17140099acc2aa8a15c9958 : SUCCESS in 6m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/96f7a4415878403fb1d3cc5f1e00e5fa : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/849cada213c54f469d8e6ddcd5e2ef3a : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b077d955f6184d1484fe4bcb8d972ca5 : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c4e1d7466b8da61213c26dde0bdd089a50b9e115","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-05 13:22:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe7bbee2f2ee4f6dee78a67d3928b88a61bfe583","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-05 16:53:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24c3ea47cd3af35f5afbb3d6f96f0716c7afa447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 16:54:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8d32078fecdb42aca25cb1f4021ff43f2b2c7ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:45:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3528f6b58d464d96927c15336744463c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c93e0ddcd94c70a1c3d8d3f6774382 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515a36266a8346239a0ba1e8625e656a : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/185045ea70a340de831ad470f92450a1 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65034d5cd65841958226a18d458afd46 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4bed2d6db0d4ee2a55bd2e57bf639d3 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d37ca66eac3a4aa4b9996664644da575 : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59b112f046a9402688433a2a7e2480ed : SUCCESS in 7m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9bd5254fd6814a009692761ebce753bc : SUCCESS in 10m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/023efdcb9995422f8d8cc1baead897d8 : FAILURE in 2m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b7a91666ecf5488791a0af36e8b6858e : SUCCESS in 4m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7771cb60e6f44079a088b021a7403b09 : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1256c49bf535460e9759bf2b63ccf55d : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b054e03d595d4f60b854d3287facd04a : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d790b4422b34b79970bf8e874e0be77 : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/eead5c2a0e6344be9abf27bc8befb367 : SUCCESS in 7m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4df97853f1634736a6b6838ec08ad4b1 : SUCCESS in 11m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/69f3a8190c6c4889b4a10f07bb0eb8da : SUCCESS in 3m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c5838384cd73405b80881a75a5f32b92 : SUCCESS in 8m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/554a41db64404fb4b59ef7d12144e935 : SUCCESS in 4m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5839f555183348ef9060451b84b1b85a : SUCCESS in 27m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b08cde09fb1f4ebcbde2cf83f961c426 : SUCCESS in 14m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/973737d6208343c38f000abaec1961d8 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43e8b210da3b45c9a16301deca9b14f2 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d85c66606d834adb839a9a5c36a6c899 : SUCCESS in 25m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6d68c43b52bb4c7391e2ecfaf2d2f3ba : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4603f1500aea43cabb7e961994fec7fb : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b78c639c05740dbaf6a09d6ec03e96c : SUCCESS in 6m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/180ef03e488748019a77df088a83f209 : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/377972b194604e54824e23d3e999eff8 : SUCCESS in 48m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17ae024e819c4135a71314183c576371 : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":1},{"id":"74af2894c4bdc847e650f5208ee874b2e4926829","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-06 03:18:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ddba423a62f6459445d9c5a16a3a84dd77c2793","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 03:36:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d7e41f87bd4ba5b603aea6714820aa\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4f1538e186b34399991ed36b7ae5d7b7 : FAILURE in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6befd4e82675cb0e257ac96f7a73015b488cb429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 04:58:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e80ae2d9f7246f598d2f4950c7454aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb6619e6011440f995dfde04fdb9625 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc853afd9b3044258321112a0ccd8472 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b92bb48f1904600b4c62f3b221a3f28 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3add4f38dbfc4f7fa7a33c1c72fb1166 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2dd0984bda5d4f6da7018de349f52819 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0f60dab8e5a4c2db88e141e8761f299 : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fd2f24be54c6442791fd9e2f38543315 : SUCCESS in 9m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a78928b9350f4d4780ba1a302e77cede : SUCCESS in 12m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7f71aef371df4250ac52fa4b0583604c : SUCCESS in 9m 39s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9961f9d20979429398e3bbfc190b854e : SUCCESS in 5m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5aba3f3851c8453fa803cccab3bdaeef : SUCCESS in 5m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/500a7bfc2b48456887d8935845e659ed : FAILURE in 4m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/670882e04a824088b70a243072dc2c76 : SUCCESS in 5m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6ebf9877b1b84caca5694178ef43f196 : SUCCESS in 6m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7bf95d017201490e936c704e0f572ee9 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c462bf0752804336b0b1c3d3a16ccc85 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb844597ae6f44bfb431e58b599c7bae : SUCCESS in 10m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e83aea1526e043bcbd818247bda2898b : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/33ac4860a6904d92babff3abd138aa37 : SUCCESS in 12m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/370b565aa06e4f669e0ecb9d32a009e2 : SUCCESS in 4m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3f96b955e0f54695a35eeafbc7f640ae : SUCCESS in 12m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9d20a7ce57e14a47980cf730eb561ea2 : SUCCESS in 7m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/751299fd858047579d7eefbf2756eaf7 : SUCCESS in 8m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7edd2971a4e1412984ff9ee42e8ddcb2 : SUCCESS in 24m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/100d9a4ee64445fd98433318d0bd22eb : SUCCESS in 14m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3cc8d9233024457cab0e80878ca6a340 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0598c332898417fad8b82e8b8e1d15b : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7d07ef113f84ab2a874be136d464a52 : SUCCESS in 26m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/31aa0edd6aae43c8bd591cb715b3529e : SUCCESS in 4m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f4e6aa9b35c045228442e81a3c548349 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7ba1295c1184c839142ad798b8fecbb : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4c4784db47a34964910ff4243bce4b83 : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c319b7bbfdcf4ae4bc4eb53bb324dfa2 : SUCCESS in 57m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abceee9cc72d47c29bacfe3e8289adee : SUCCESS in 1h 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"94524d04c723e4e895343585ffbcf5bdb274f014","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-09 22:55:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6997f3447e86236e156be401b37c9f7da66dcace","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:30:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909906f545134cf6b113838734e6976d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d236d11f38d4247b4c59a6e7973578a : FAILURE in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c53defcf175d8c76e52a7e51c92cd68dde117d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:39:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7975d7a4386c4cf68c8d6d849f33f802\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4945ddf21a44b793c8e8e3997075e6 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fd7cf6e5424116a57b5723e793721a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a29415c2709c4edaacf94596b2546bbd : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecf6b29fc9354089aeeb2481f3461f37 : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a04c671161e94b90a90dcdb7eae19824 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11b4c7f5ca6247fc9ab65a686336f726 : SUCCESS in 2m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0363d263569428599a48a91f86778b5 : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1c54f6d9fc3f49489509a097231d66cc : SUCCESS in 11m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/283f9e2572724635a9d8392ebee1581d : SUCCESS in 12m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f3dc526d6a154af48771dba5a0397968 : SUCCESS in 4m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/69aa9d1159674aa8ab42e9561335d943 : SUCCESS in 6m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d18edbbd461c4a57b5d32d96b631d5bb : SUCCESS in 9m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8babbaeeac3147a0bc15dcb84cc720b3 : SUCCESS in 4m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0a83b542c0694ddaa4b9931f48628966 : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4ef1d342b1064df2877c654b51a57970 : SUCCESS in 8m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c91e4a9fe9ee44db9e6b4e6b25095962 : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/efc0b8ec97dc4d42ac89782fcbbd5c69 : SUCCESS in 7m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2575400df4a14587a253c8f5a1c36121 : SUCCESS in 11m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a689551f570c4be99e343e3cd4363a85 : SUCCESS in 10m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ed82b84a85f64945a73d6a6ab7c12ce4 : SUCCESS in 3m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/63398120fa254775b01eefef52ab0dfc : FAILURE in 18m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d5025f9ad17b4d62847a258e1dec150f : SUCCESS in 5m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6d9d99667d8845668e399801653df71a : SUCCESS in 7m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/740ceef17130419d804e24cc30f6e940 : SUCCESS in 29m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ccf9a63d97584a6aa2dfee885019117d : SUCCESS in 11m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d7c3971175324f579b7ac4b802fbfdb8 : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ce580cba25c34b4e8ef446dce1382248 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/05dfc2c42c4942b893e692b8f92c1f2f : SUCCESS in 25m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/35c71846676d45caba6f39e04e25965f : SUCCESS in 4m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d949121de9a041308ffc9d7500646b03 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0aa03802e646460ca142e9d0d2eb6e84 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/72db29c14f904380abb7d231a12851b3 : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/418f6bf08f4e415b82041fbde8062110 : SUCCESS in 38m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab7a4f934d6e4c52a8fe5dc0b02b8c57 : SUCCESS in 1h 41m 56s","accounts_in_message":[],"_revision_number":1},{"id":"18b9ecdd7934a06ef189a9daff7a7d4af0ccbcee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-10 05:46:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"886b0cf15f8e8689c351f1768ac8a23bdb92b497","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 06:41:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e1c410e13364ec7a23998a14d357c68\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/680e8f239a414f48848f8c8c52579121 : FAILURE in 13m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2f9abc1528c0e4b1df8bdc69741fa38be229577e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 07:47:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b567ea92b842b089c4264ffe662d14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23bb4f672a845f585d4c96f7c46a817 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066e0108d53c4bb6ae48daedf13e5a9c : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/529d76e6f0464d01aefb88feb2566f47 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78a4b619346b4d33a42e826d24cce132 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/535cc11cbb7c44ac8360d066b0091454 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00da784f49eb4c688b178201c2ee8247 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b99cffd4b204c6fb62662e379a137ac : SUCCESS in 7m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a8261f39b763445f9234fac89894ca88 : SUCCESS in 10m 06s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6e6d713875bd43b78d605cb7176a1984 : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d43cbaa0ec2d46b78096c363d141eda8 : SUCCESS in 5m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7269be0acc854559ad52200de93af946 : SUCCESS in 4m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b80c3b70cb134c2daa57a86400c15148 : SUCCESS in 9m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0a13cf3219b34852a8a5243a90eba034 : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a5d260f64c1f4314bbb242815c2e7594 : SUCCESS in 4m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1c0c9bba45864570ac08cefcea7a8f1f : SUCCESS in 6m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/06e1b66f6bbc4aecab6eb82e8e52bf4c : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d0355b61c61497ea66a1df8045208bf : SUCCESS in 22m 43s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1b0ef1070f9448568a90495f8352f15d : SUCCESS in 6m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b932ee8d03c34464bf0d6cd6d1acb058 : SUCCESS in 10m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/af2ee8dd0b4d41fc9d0f93630ad448f7 : SUCCESS in 3m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ee603f51613c417da22e037b1ece0b29 : SUCCESS in 8m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ad7c817f89204c9ea26752b1274e82aa : SUCCESS in 4m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2ce2d93a70c94d029bd089555dc30d36 : SUCCESS in 9m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/afda487826174348b96d010a9ea5c359 : SUCCESS in 20m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5b3b61cf95cc4ea39f8ddd2029f368fe : SUCCESS in 12m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3721b91d0c89438facbc393495331378 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f6ffb19c1cf4364b35bbba3185961cd : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82e90c4b83b646e2a7a7b36785d23ea1 : SUCCESS in 18m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f2e7d54ea8684ece95e8e33019ccd9e9 : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4c5ff59978284eacb713c447b20ded7e : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e646fa1368d48778ac9a8b7dbbd3fe3 : SUCCESS in 6m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c4874b43d1274e04a9ce1ea7f81de5ca : SUCCESS in 9m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce8809a5b28f4eefbe210cc28cb0ca17 : SUCCESS in 49m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e98c8b3e99647628140cd35228acbaf : SUCCESS in 1h 52m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1c05676dd97d3bf0924ced3e830b5a7b03b49b46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 07:48:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37e2001878c61414eb08a8b645ae5101a41ae23c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 09:42:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/584321c9db55422da0e71143e6576c09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fc4bd42e377445db91198b4193331cb : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d83b8dc9eeb4c9b9e1e56fbb1ceaef5 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0df88e0c5c2e4f61a98f1f0a04c59577 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/268d5bd3fe3e4a9b9d0a6328216f2b4e : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b113257e8564aba83439456b20394eb : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8725e2a55ff840ba83d1356b2400c954 : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4dd6e3aec7644a38982b57b50eb33979 : SUCCESS in 9m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/00815955a5af4395b2834801f8990432 : SUCCESS in 12m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e2daaadd1ff24106a3333dbe20132892 : SUCCESS in 9m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a298b3103e244bb0a335b569e5d9eed9 : SUCCESS in 4m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/01e210e5e98a464f9ba043db840f41cc : SUCCESS in 6m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1195c597935c4ae8b3d10052e1c6b84c : SUCCESS in 8m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8afa831fb7e2425ba05fc7986ed3bdb6 : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/701f42f6a9ec4d619fa0de5d77ec1f9c : SUCCESS in 13m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b6c5474827bd433fbdc0bab51f398838 : SUCCESS in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/35f06923769b407980a1c1d124503353 : SUCCESS in 15m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/08c8d424fddb470bb61dc03d5e684cc2 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/239be8418252476f87165e2b5311fef3 : SUCCESS in 15m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bc637b6e9b3a4d6c90ca654697dcfb29 : SUCCESS in 5m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e87faa21ba824287880da2b3c94908b5 : SUCCESS in 36m 08s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e3998c76291e4bd8b982eb8784d97558 : SUCCESS in 15m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4206034f095946a8be6f932fa3f86b34 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f7ac9742927415ab8bac9afb8b121b3 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6cf99f82b6d34b11b8d5f5f755662eb0 : SUCCESS in 24m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cd8b9e58be3c45e297809b2c457bdeca : SUCCESS in 6m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/53fa1291fe9b4d1081cc04ce75e77e06 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45112031ca2a42efae27e8c419987650 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/101548495e1e441382e7e5f3d5d5de35 : SUCCESS in 14m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d88bb46cdfe4af2beb242b1b78b110d : SUCCESS in 58m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c5ff40e27414379b6e39b5c13c0a28a : SUCCESS in 1h 52m 23s","accounts_in_message":[],"_revision_number":1},{"id":"dd0852fdb08cd938e17ef04c4a5be5d8ca5bb636","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 09:42:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"48952472101af7fba492cd6e74cc7123e67b88af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 09:43:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2ca5d55403c423e980218c9c4e444ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6125de91891a4dfea001051713d95f9f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a6af3a72c499d1411992da05faf63a932de4174","revisions":{"8a6af3a72c499d1411992da05faf63a932de4174":{"kind":"REWORK","_number":1,"created":"2023-01-05 10:20:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/869338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/869338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/869338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/869338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/869338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/869338/1"}}},"commit":{"parents":[{"commit":"27174c1170d666bf35de05c56547d3e25551f3aa","subject":"update constraint for keystoneauth1 to new release 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27174c1170d666bf35de05c56547d3e25551f3aa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:20:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-05 10:20:51.000000000","tz":0},"subject":"update constraint for glance_store to new release 4.2.0","message":"update constraint for glance_store to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I387be3413714c436e6fb576f1a7108c51d82a74a\nmeta: release:Code-Review+1: Pranali Deore \u003cpdeore@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I89033024d6fcaaa45815732dc2684dbedaf8efed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a6af3a72c499d1411992da05faf63a932de4174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a6af3a72c499d1411992da05faf63a932de4174"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
