)]}'
{"id":"openstack%2Frequirements~869484","triplet_id":"openstack%2Frequirements~master~I8dab1534dfeeb8d2accc3369a20540b6771b058a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8dab1534dfeeb8d2accc3369a20540b6771b058a","subject":"update constraint for keystonemiddleware to new release 10.2.0","status":"MERGED","created":"2023-01-06 16:24:57.000000000","updated":"2023-01-07 03:04:49.000000000","submitted":"2023-01-07 03:03:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869484-new-release","meta_rev_id":"7dde064374480d0e84827bb691f078dab8738f0a","_number":869484,"virtual_id_number":869484,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-07 03:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-07 01:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-06 16:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-07 01:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-06 16:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-06 16:59:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-06 18:12:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-07 01:27:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6a48171c2ef0fa66fdf8908715ba0a6cbb35e3db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-06 16:24:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acdc5c09066171c74c3369bd1037189e1813b8a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 16:42:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b45ada2bb4f44c48054cd7de51e170c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f7c12330a1647b19f08a300335becff : FAILURE in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"616c9d3a3ca21bd2080086618a639b2e68701761","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-06 16:59:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67ec6b1287849b902d4e35f20a08f225f068c5d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 18:12:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3a93e8ff0d94d7d823acfe6ab27f68a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f89202e0cf4d859616525380f731e2 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b748906a9840ab83024deadc0bb502 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8692a4b1b5ef4adf95b18fbbdb7f212b : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d609385163b04dc3a91947bfe3fd2479 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b996778de4e8402a87e8b93d05820706 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/786db0f0435c479eb8a0eca8f4a70a1a : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05c40c1fb1c341f2baede876da223464 : SUCCESS in 8m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4eb3e338543f496282daed43ea516175 : SUCCESS in 14m 21s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/21f526e1ce0e4f619b02c70eb6ff860b : SUCCESS in 7m 17s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6acf4afd144641f59f8aac0bac2136c9 : SUCCESS in 5m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/050abc3a534344bba1a52b5aaa90ad24 : SUCCESS in 6m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5e52049dbb074a66a43691d17820d7bc : SUCCESS in 15m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2e827b3c093a477f894de1347fbf67e5 : SUCCESS in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a7f5fb7dcab34d488d51dfa855957511 : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6ec5a5ea906141339ad5c7098a0751d3 : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/54c8191b4a3c4163b7d2eccbb559431f : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba142243f9df475c956fefae116aef44 : SUCCESS in 7m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b718c82a2cdc42f8b9483f8846cd7e99 : SUCCESS in 8m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1ae680b2573a4bac90420af8ce5822ec : SUCCESS in 15m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5706dfb003224f29aa06ff7799eecd4e : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1e53f3307f9248bfaa4d4eb10797f5d7 : SUCCESS in 14m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2b50b1f0d91d4bb395380682877eb46e : SUCCESS in 7m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5849f16c71d741e98b204b1f31bea165 : SUCCESS in 8m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2084af8abfaf4bfe930522a995e37016 : SUCCESS in 26m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1e9a5dab7749408b8a19c67c9c6c2b2a : SUCCESS in 12m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2dcf2101a6f54461b800785cabbb86a0 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2104a67d64f4e66a8151596fbad1ef0 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a73aa49fbc24c21bfec7bb1802df3f8 : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5a3676de774b4f5b994f13c4439e12dd : SUCCESS in 4m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d279a813814040fe87344cd265bf6f17 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/476effe07b544349b9a2527f30a693b5 : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6afbd057a4f1459e9053757392b89635 : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c99722cccc842bcb9e2d6997be8b94b : SUCCESS in 1h 03m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/baa93250f041459da8d7a42ec8051d54 : SUCCESS in 1h 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"174516817d0a55a1fe5234fd2ae5fdcd11b1718c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-07 01:27:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3551d5d7007528dd12af8813de9a94f91beba183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:28:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66de8b3613c9b9b557f922a0b910a6034615091c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 03:03:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ebf29487264de08303b9b03a56d2e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43603e3856c4c8fb6dd671cea444e49 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac8638d65e764c9e93f307fe3bb787aa : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7eceb75c1614cc8b48eaad42b7e6e7b : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/642592943e5040e98ad2669dbe0d9b18 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c870dc2273f1420587cff44bc000814c : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7583547498548eca02843d1caff9907 : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82b0ebafef944e3bac3e965fd0e6043a : SUCCESS in 10m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6bb00f588ef046789c6286852c202c56 : SUCCESS in 14m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/378c0e47283f4c6f9fbcf2d3ab084023 : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/db248846cb1344ab852b90ec3ca7ce1d : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/249b3ce71d8a4a22a22ccd6d004fcb78 : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e4b4e4753aa8441cb1e601f0a61df6cf : SUCCESS in 10m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b9e89ceb1a42412099b4cfedcde124d8 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50ab6742731d475fa13cc79433e96303 : SUCCESS in 8m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/68e01c8508c1490ba9a3c0e9777b0658 : SUCCESS in 11m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/306aada9330c4d5e9d33c40f5f2dfad2 : SUCCESS in 10m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6a1db037564f4507bd84824dab1cdce1 : SUCCESS in 4m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/86540b879b954420a1d6835f6d743323 : SUCCESS in 13m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/52b19bf1bf784d1abe17123de2f4fd08 : SUCCESS in 4m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ff0d42f1bb2e46648681bcf2eb7b30b4 : SUCCESS in 31m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9a123bfe122f46bfb56b50df6260c233 : SUCCESS in 11m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/757953cb54dc425e896a0a04c11e8faa : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8549851979e54e7e885ddd4523a428c0 : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b48e5d403d9442a8de7eeb63ae17b76 : SUCCESS in 20m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/36c5a76fd2b448b49183e6b8ac8d94bd : SUCCESS in 4m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/73609104a0d849ae83bc57baa2defa71 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04fb4f64487147dd8f4bd24ea7a8adb2 : SUCCESS in 6m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3a62176a362f4b69a439a3e07128ee7e : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d29e7e398bcf43c1b786522e9aeb4a48 : SUCCESS in 59m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e8869ab3ba148f4bd94af70e29c003e : SUCCESS in 1h 29m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1471ca8896705063362fb273f4e4948d9ad9571a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 03:03:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7dde064374480d0e84827bb691f078dab8738f0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 03:04:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85be714a6e7343b4a99340ac56881b2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda32b71c7c34311a9b3f41308cc9839 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9a56ee55e929e7c562d2c28555625c9f1c99d04","revisions":{"e9a56ee55e929e7c562d2c28555625c9f1c99d04":{"kind":"REWORK","_number":1,"created":"2023-01-06 16:24:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/869484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/869484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/869484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/869484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/869484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/869484/1"}}},"commit":{"parents":[{"commit":"27174c1170d666bf35de05c56547d3e25551f3aa","subject":"update constraint for keystoneauth1 to new release 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27174c1170d666bf35de05c56547d3e25551f3aa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-06 16:24:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-06 16:24:54.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 10.2.0","message":"update constraint for keystonemiddleware to new release 10.2.0\n\nmeta: version: 10.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I677106b5215275925b135e9093fed5f83a1413b3\nmeta: release:Code-Review+1: Grzegorz Grasza \u003cxek@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: David Wilde \u003cdwilde@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8dab1534dfeeb8d2accc3369a20540b6771b058a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9a56ee55e929e7c562d2c28555625c9f1c99d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9a56ee55e929e7c562d2c28555625c9f1c99d04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
