)]}'
{"id":"openstack%2Frequirements~869803","triplet_id":"openstack%2Frequirements~master~I2d998da6396a332f0be91bb6eadddb60073707ef","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2d998da6396a332f0be91bb6eadddb60073707ef","subject":"update constraint for pbr to new release 5.11.1","status":"MERGED","created":"2023-01-11 10:09:35.000000000","updated":"2023-01-12 18:11:45.000000000","submitted":"2023-01-12 18:10:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869803-new-release","meta_rev_id":"9fed75971276679f88b5b8ad406da6dfc9627e5c","_number":869803,"virtual_id_number":869803,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 18:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-12 16:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-12 10:45:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2023-01-12 08:08:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-12 16:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-11 10:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-11 11:52:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-11 14:14:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-01-12 08:08:32.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2023-01-12 10:45:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-12 16:26:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"206912aec743a0bfe084e28ac5c36803af60a20f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-11 10:09:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9afeaa0d746b239ab3d37f96cac35e2ef05230e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 10:23:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4bbb6ad261544f2826d7500872e69e8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/13e20c9ec59743edb21d33f36c32b63b : FAILURE in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e22500275395625f670004a03df6344660ce64d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 11:52:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddbdacb72bc44d959cd1427c47e1981b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289f47705ab74229a75bdffc7ae8a1da : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abbf3d073a548a3b285587f39caa28e : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b3ee956af014922baa2e5b6523d07ae : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bb08e92731d47a8985481b3a6b522e2 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b72e2b5939b4fd090ed5b4dacd1e693 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b55ca488e40949fe9d19fb0b1d97a4b6 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/335e5b8852ca42aa94efd4e74cc5cd32 : SUCCESS in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3f1303eb223a433fa90dacaed2932a31 : SUCCESS in 11m 28s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9844b6c1dcb04d9c944b7dc599870266 : FAILURE in 2m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6ef69301261c4bea85436028801e08c0 : SUCCESS in 6m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d17578cb99004c3181a3a26aad13be85 : SUCCESS in 4m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f41e29012fee47a080a558c12717c1c0 : SUCCESS in 10m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0234ab5f87bf4e05b834e3776ae26ad4 : SUCCESS in 5m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3933ccbd2d8e43e5b5e08997d5cc3a43 : SUCCESS in 5m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f3d1b86117454692aa89f54d35c58871 : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d267ca8215064c43816d3ff6eb46f206 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26513a08bc764355a8f76e83c4e789e2 : SUCCESS in 6m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b2a46435e7b445e7b0b92c9bbd24f1b5 : SUCCESS in 8m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bbe01004c73245ca9f9fa24104b8e23c : FAILURE in 3m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d03e923ceccb4fcaa244b8b2b01c0444 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4ae411d950354731adf48a55bc131337 : FAILURE in 3m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4e60135ed8e948cdacc4efcb900f1ff4 : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ae338e32aed44b0689a304e559f907e6 : SUCCESS in 9m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e38948c17a4e48489da9aac5e46fd5ea : FAILURE in 3m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/663c3f84db4648708c16b8b05e8c8e9d : SUCCESS in 11m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6a34cc7e2893489b91f7dcd62e8d90b6 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e0d85bebb3248c3beae8bc6dbbf1964 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ac5a62ae8964af1ad637986e201dd3e : SUCCESS in 20m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b1beec9164414f95883e0af28b7cc287 : SUCCESS in 4m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a4bcb31c39a04d0fb6e3cb889793fe05 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73ffa077c5594b47a35ddbbdf38837c7 : SUCCESS in 6m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/09267ffc4d07428e8b51f3a4153aad79 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acba62c5c43b48f5bb0da966b0ba7661 : SUCCESS in 43m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34174e3a3f354a71af4a690b90dcfe64 : SUCCESS in 1h 36m 51s","accounts_in_message":[],"_revision_number":1},{"id":"bd785aadef0b424d7ede37937e6f0a703dfb43c6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-11 14:14:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f104f0d2ade3847f4baaabe0dd15588553badc59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 14:28:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a30a503ef3eb491191fd1100184acc7f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/18e48bb66f5c49a084b7bad9c15ac503 : FAILURE in 11m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6bf8632c7cfbc06ab6eb317f3b8e5e0331f90aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 16:15:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e24a7ef91b54cdd9ef1d4f9cf6d517f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07751edc4c8b4ddabf1f2f372f42090e : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a23fd43ddca4e4aa41163699293fb00 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd415d4c96941bf8a7135727c035618 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d37fe3b09c2f4daca5fcaba116f53739 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a4fb11bc8f54c349b52b7d0efcb0b0f : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dbed1713db1f4e4eac3a53f263a21622 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/caa9e02d89d44fcda1d2f6324e573cd0 : SUCCESS in 8m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/70f169b007314eed8ed187dc0302de2a : SUCCESS in 12m 40s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a224d39ca7d34f78a166cc1d653dbf7a : SUCCESS in 6m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ffc20d033ec44d1fa57a06706531e048 : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0e92e691fdf64ae79812702c6b96ecdf : SUCCESS in 5m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4f476631e1e74a31b10fac75c697ceb7 : SUCCESS in 10m 01s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4a5f1ec0e33a422495e4a3d63a28e53f : SUCCESS in 4m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c54f323b8b0946ee8564bd8ba8987119 : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/86f4d053fd22442ba65a22810012150b : SUCCESS in 7m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0084035bc3da4794a984b649eb61199f : SUCCESS in 8m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07ff4f56b340441c8225a668166cfa80 : SUCCESS in 9m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b898b9b150004a0bb31de87528884805 : FAILURE in 7m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3c3c6f3254274018b129cf33d2797576 : SUCCESS in 11m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fdc8cde977d94376a5b498e9ce01d02b : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bb61c6c0148d464eb51b6faaf72d515a : SUCCESS in 13m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3fc36b74504f4869b4077b6c0d3a0ab9 : SUCCESS in 4m 44s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/45338439484847b192f8a7b303b0b1d6 : SUCCESS in 11m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/227773f225744537a3cbd9190fa0df72 : SUCCESS in 31m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6379286a8d9947429e78691bf36b64db : SUCCESS in 13m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9c4bf0fb6f4545949a41171855035681 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f318644394dd40c59cf88554df4382ae : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/637690971d684edaa034c9291b22d79f : SUCCESS in 27m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cb6ae5e4b0b44e11b806c0b2b7dcf082 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c4268b7d8e40419e9dc8426aa7c64958 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dea9559372064895b7807ea5ebd3f573 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/33e42803b9354255ad7e8308f15a2961 : SUCCESS in 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4faed0d96ce240c0a6dc443b8cad0d9a : SUCCESS in 1h 07m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba87ffc5859b47679fdf23b138f26b29 : SUCCESS in 1h 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"de45602aa36edc939784d77809479db74bc15cb5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-01-12 08:08:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"789573206a46d79b4bea6e97ce52e3c0293aca52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 08:23:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce97c2ac2a749869ac67e54c92fcb45\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bb55bde4a916449bb4eff03ff5ab2d9b : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"319ff7a35a9e3d0ba0d1661a1ee5c59a5fe77cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 09:37:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c6f94963d646da9c98c086589f56e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ae88fefd004db6850b7e033376c5e3 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c50879fba243fd9a4b6967d8cc580a : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e94021dc878410e884993e41112772c : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0f88111be234d31b7f99c37c8ec0631 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/399a494e37d44d8c96aac7040c5c0e9a : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef679c42e1ee43caab960c0c93e820b3 : SUCCESS in 2m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af1fd52fad0245d2af5e1417ab53a58f : SUCCESS in 9m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5f51f7b380ce404babae2ddb4ffb370e : SUCCESS in 13m 34s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/79e31333d0714db8b41d443043286177 : SUCCESS in 7m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0e7a673a2419484fa7cd1a590d67a8a8 : SUCCESS in 5m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/74839fd7547b49c8b96f30e70d32daec : SUCCESS in 4m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c4dc9d81f29d499ea383b4e16ce05e40 : SUCCESS in 9m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c2eedd794dd94b6cadc07f61927cce7a : SUCCESS in 7m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/28f9b058844247ecb0f91533dd6e4e82 : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/61c231f592f94e89a73eff340dabe7f0 : SUCCESS in 8m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5d533e7971af4df786cb6acf1ac27edb : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd971af1c6ed4972bd3e136ef49e776d : SUCCESS in 6m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e9c9826e4eee45ae99dcff8bab467976 : SUCCESS in 10m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a60725fe3aed41a0bb74818cbf3db87a : SUCCESS in 15m 21s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d0359b301cf0490cb85c3870e3541ccf : SUCCESS in 4m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/096abf2835084a9ca2d8194fa9bc5357 : SUCCESS in 10m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c19f67922cc845de9c09d0ca701b3fff : SUCCESS in 3m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9ec33e1a6cda47b0aa33da02dea43b8c : SUCCESS in 9m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bed86f3245eb4da0a1cd264fc5bd307c : SUCCESS in 27m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e84c1225f8424318ad730dce0d823c4b : SUCCESS in 13m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/97c273ce56794f27a15b20568cf7846a : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95d2505cb40644648cb26f966fdfcf5b : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89d5014c94f14a29a7b1df14ae701859 : SUCCESS in 21m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9f28f29558844207af2c3dd9a7ee51dc : SUCCESS in 6m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/20de94f19fe541caa47220ff3b5a15f7 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4212e2cd01c4921879dd41f40957ea0 : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/965e9c54d42346e5a46b7505a0b9a155 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a784697fefc84db7a2b12492aa761c15 : SUCCESS in 1h 03m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35397007384c4b19b0e8153d03331ef0 : SUCCESS in 1h 24m 51s","accounts_in_message":[],"_revision_number":1},{"id":"539b8a16fb5fccf265dffe6e5da4f08051c525ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-12 10:45:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f172b917313226da57270880ce4b013c07ada6be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-12 16:26:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ddfde7c741f37416275e0b73627686e7f6e4737a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:27:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b3bac68aeff3cd6b9f7c2e42287879b3a8de8a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 18:10:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c3973e9c1694f20b1e5facaa78dee74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276aa6c9287a4f618c6bfc6370b6538c : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad85b2633317409f9b5dafd49f160d38 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f482b00d81dc48c7ba11fc633692c784 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59a86592ff58477eba3ec0f371577e1d : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62ef5328af0b4be3abe44d728a29aa18 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/937ced38d381458eb25a8fe8ebbaec3b : SUCCESS in 2m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/789acaaf04bb4cc590bbca7bfd3c463d : SUCCESS in 11m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b8e357f1bec947c0b6320b8d2acde529 : SUCCESS in 9m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bfc15103fe0945deb43bc805912613cb : SUCCESS in 10m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c1ae021233d345b9a1496c7040d8a63b : SUCCESS in 3m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fc4bfb5d3c0f47cf9dabffd808ccb708 : SUCCESS in 6m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b98d3abb72254c0e85469e7bea80a55c : SUCCESS in 8m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f2499af939f14cd09479d2fe8ec046fb : SUCCESS in 7m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6b33e712a45424dad584dad22749211 : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/acdd7da837644096b87767ed14a5efbc : SUCCESS in 13m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ea1fc503aae94fd9bbfe9e63e4d141c4 : SUCCESS in 12m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bccbe995272b491789a6036f30156ecd : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6ecfe891c5fb4e46a7ab12beaba08a39 : SUCCESS in 9m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/97a38f0b30a24421a95325cebd3fa276 : SUCCESS in 4m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a1fa529236c448a4a7e97f4f94173660 : SUCCESS in 35m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ee24d2ec41f440dbb87d4ad36570c744 : SUCCESS in 11m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0d7718d557f24b6f83ee6412488ca508 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc781bbd63d54c83ada4cf90e65bc310 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c220b4c0e177456b8acc910572733dac : SUCCESS in 22m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9f11cee3d49b40258fe8aca65fc6c5d0 : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0fe4cd2c51fa465abe30ef3978672116 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fabfd93e6ca9421192b7966ca1126ab3 : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c8195792754d4292a99f859b7ad59650 : SUCCESS in 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/750d435ee19449859cd4c9716c8de6dc : SUCCESS in 1h 00m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcbd8fe028d64c8c8016d70cba0a18e2 : SUCCESS in 1h 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5666a3e0623b66d2c8004f50567ddcc9262acf7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 18:10:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9fed75971276679f88b5b8ad406da6dfc9627e5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 18:11:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5395a9dc72284743929b258c4ca259c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415ae5368f044475835772b580c29e15 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9816eeb72a01e38a2cafe48e32ad277e00435a80","revisions":{"9816eeb72a01e38a2cafe48e32ad277e00435a80":{"kind":"REWORK","_number":1,"created":"2023-01-11 10:09:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/869803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/869803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/869803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/869803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/869803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/869803/1"}}},"commit":{"parents":[{"commit":"2aaf64dd91c63aa55f4cbe8c037a6f545e91b302","subject":"Merge \"Bump GitPython to 3.1.30\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2aaf64dd91c63aa55f4cbe8c037a6f545e91b302"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-11 10:09:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-11 10:09:34.000000000","tz":0},"subject":"update constraint for pbr to new release 5.11.1","message":"update constraint for pbr to new release 5.11.1\n\nmeta: version: 5.11.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Commit: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Change-Id: Ia5ac5672c294e3689b35cd569827e3f8f6a03b5e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2d998da6396a332f0be91bb6eadddb60073707ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9816eeb72a01e38a2cafe48e32ad277e00435a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9816eeb72a01e38a2cafe48e32ad277e00435a80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
