)]}'
{"id":"openstack%2Frequirements~870911","triplet_id":"openstack%2Frequirements~stable%2Fxena~I852497bee4e9ab210c0055aa9e892c1c23c9efc9","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I852497bee4e9ab210c0055aa9e892c1c23c9efc9","subject":"update constraint for horizon to new release 20.1.3","status":"MERGED","created":"2023-01-18 09:50:15.000000000","updated":"2023-01-18 19:43:16.000000000","submitted":"2023-01-18 19:43:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870911-new-release","meta_rev_id":"f44cf8517fd9ba1c31ab91a9733ca0ffba317d88","_number":870911,"virtual_id_number":870911,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 19:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-18 16:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-18 10:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-18 16:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 10:11:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-18 10:28:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-18 11:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-18 16:09:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0a8d0a13739f8115a840ebcbe73552c3ece2b7a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-18 09:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4177512e80d2ab6660c130ddd38c45ad74e0281d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 10:11:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/361a0bb9b3284f6cbf600baaca3cf559\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e0ae02cabe194cb98356dcfa13a359c3 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ee499ae7b2ab778b5aa949c2a6cf0f6199a0bb93","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-18 10:28:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4562e089bd4c93b49421c42286a606eb8cb2e71e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:42:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb3ba8dd76404c3693e820ce79e424d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6077e43b883b4a14be06ec7abd27c6b3 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ed3ce9097b94e3495a4d750b8db88e7 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a893cab65f64ec4a9d4937805f497b9 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5b4e5511b064963899b79f2ce92e31f : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fc9a4b7b04142739f73565436abfa28 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b62e98e2df644aeca6e5d704b94b2388 : SUCCESS in 3m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/32f5f4f8cc094b7f861b3d3d98136563 : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4dea6e72ca6a4c89b52815772d0f36cc : SUCCESS in 7m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8d52174fe46c4fb5ac62d9a64c74386a : SUCCESS in 4m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3ea7f1b31a024769b5a3372e5e50573e : SUCCESS in 9m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9b3570491ffd4f1bb2d04199413c8f71 : SUCCESS in 4m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3fc1ae7fcb174889ad2cea811f01ee91 : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e486021dd24e47d89ebdb547271841ca : SUCCESS in 7m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/37063be8dc014f179c116deaea3e2eb5 : SUCCESS in 8m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7fbaa54a4767423aa54b83d213ab4ad7 : SUCCESS in 17m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c18f57de3f35484eab5b4cda61346d04 : SUCCESS in 6m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e30e8ab4deb74eab8271bfc08738ef50 : SUCCESS in 3m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7673993a4723473fb1c036a0026c21a9 : SUCCESS in 23m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0a38f62897b742db9919ac1bacd639db : SUCCESS in 10m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/999ef2a23e3141feb3cdc3bd744cc63b : SUCCESS in 4m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/463296a2d4a04706aad4f0a084ba6c0c : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86aae3ac89694fd9ad3127deb221d949 : SUCCESS in 20m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3507043938ae4114b7f0ecca2da0e883 : SUCCESS in 4m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2813a8dae9344ba1aa201750d242a662 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/306314cf7efa4cdab5402b9aaa657524 : SUCCESS in 5m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5cdaf701223c4f8da7b695624b1f5bb8 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e14cd1a1e8c341fbae47cc0489043140 : SUCCESS in 54m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe56e2bc6ebc4fadbccee73f071c2a74 : SUCCESS in 1h 45m 33s","accounts_in_message":[],"_revision_number":1},{"id":"31ce52079549d9e88883e1c59ee19112abbff362","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-18 16:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e1da63fc52a37500361bda62772b79a4a9470c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:10:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf9cefde9edc860082e2fdbdd4c489391a2d9018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:43:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/806d1d58af0d450b86be246f879d0365\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a85e0d58e642f5bc8f7ff32447b126 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44ae6b1fc7ac46de8d845de8ae105dc9 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a18a9d5f79e48ccbc9d037fc3106e9e : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd89fddbdb6b4244894f32f525ec6adf : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f87ba80eed0478994cac85003f3749a : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/560ba8cc5ed14e2096c2ab81b09c0d77 : SUCCESS in 7m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/219f7ff11beb4f08b185bcafb350c14a : SUCCESS in 9m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c1d6733a5cb54cd498f8b2755caf9698 : SUCCESS in 11m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/874583dacd7a4afdb1526ab357b3c545 : SUCCESS in 5m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/74bc9bae7b064c4994fc492d25569442 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f757f4399fc640cc9e0893f7d205ede1 : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f07955a9dad0419ba18314ba73275e39 : SUCCESS in 10m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cb1a6d79c0f945ec9f77a5ccaee97c2a : SUCCESS in 14m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a9feae52453946f3bf0d6d2a4b555021 : SUCCESS in 9m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/135c46472dbe44db8162d6bf74277663 : SUCCESS in 4m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d3fa58e6d3794822b7d6aa875b69cdee : SUCCESS in 23m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/188fab2e3db64f3cb06072fcb8b2edb3 : SUCCESS in 12m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/427eb41c44c54be394162099b78ca9e9 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b2fa41e024044fd9a5b960e8bd3f141a : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d15ab8f1f8745c79fa13d5162aaa3af : SUCCESS in 21m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f9e3287a1471456c88c8463072855cf3 : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/be795a25c0ad417c8eebfedf4d0a1289 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ce3f7c405924a02baffee564983f91b : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ede35be9bc7c48c59bc87a8a0217f9b1 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a4c6fd6548140eba8f0420edc1f5b45 : SUCCESS in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3de97b2f8c96408195d17135fe93ac2e : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f44cf8517fd9ba1c31ab91a9733ca0ffba317d88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:43:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73f59eb357942fbc465785bfb3e71c8d469bbbca","revisions":{"73f59eb357942fbc465785bfb3e71c8d469bbbca":{"kind":"REWORK","_number":1,"created":"2023-01-18 09:50:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/870911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/870911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/870911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/870911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/870911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/870911/1"}}},"commit":{"parents":[{"commit":"ce4590a60086450e4b66c6b5d810c0bbb8258518","subject":"Use correct nodeset for requirement-checks job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce4590a60086450e4b66c6b5d810c0bbb8258518"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-18 09:50:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-18 09:50:14.000000000","tz":0},"subject":"update constraint for horizon to new release 20.1.3","message":"update constraint for horizon to new release 20.1.3\n\nmeta: version: 20.1.3\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I0cac0efc0a0458f23eca1fad2ad9114745ac18fb\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I852497bee4e9ab210c0055aa9e892c1c23c9efc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73f59eb357942fbc465785bfb3e71c8d469bbbca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73f59eb357942fbc465785bfb3e71c8d469bbbca"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
