)]}'
{"id":"openstack%2Frequirements~870927","triplet_id":"openstack%2Frequirements~stable%2Fxena~I3f0536ce13fde079df49067bec3d486424bb43d3","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I3f0536ce13fde079df49067bec3d486424bb43d3","subject":"update constraint for neutron-lib to new release 2.15.3","status":"MERGED","created":"2023-01-18 11:11:42.000000000","updated":"2023-01-18 19:51:40.000000000","submitted":"2023-01-18 19:51:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870927-new-release","meta_rev_id":"71dbe957cd3c4dde3b767f866db48cf982b53db7","_number":870927,"virtual_id_number":870927,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 19:51:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-18 16:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-18 16:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 12:01:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-18 12:54:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-18 13:20:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-01-18 16:09:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b6596cf0f9d886721f26309b6e26bceeeff0f8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-18 11:11:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c85310ff9e872a4afdda28e47a9e81a7c14c4412","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 12:01:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92114a24c95840c388ad77d3269a2fa4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8922f1e0b1e34bebb3ed40534f55b30b : FAILURE in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3befed7b93d63cc94c4c0b4b8eb3a1f560f46f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 12:54:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da7cbff3dfcc4b6abd8e1d3e6a7c80b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a485713cbd4241299f45c33ce8cd17b5 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa8118a0ad6640658281c44aa3334202 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae80c8e509184c72b668c31e1ed21adb : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fca350e2b96042ab976578ab14a6fbe9 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82861141f60b4ebb8155d1a74f41693f : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c1d0940e34c4fc496a08f799a546784 : SUCCESS in 2m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2df3ddb80cdc4efb8236e2fd4c69899e : SUCCESS in 8m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b06e3ddc47c47e6b0abaf72dd2fa581 : SUCCESS in 8m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ea955854da0478ebe3cc5de6b7500e1 : SUCCESS in 4m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/49563a4b3b1b49d896d914cb2e2079aa : SUCCESS in 8m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/76c3b2a09a024238a44701931b958de9 : SUCCESS in 4m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6fc291f3915949debae6b0631cb59523 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc407e7ff6284655a871a3143945a87c : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69ce1f464c5d4430bde2b14106d26143 : SUCCESS in 7m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/13fd3a62122140a3be6fc2863bf3add0 : SUCCESS in 14m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/147c5e0c05a142a48791ea3fa1f562c3 : SUCCESS in 10m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b0c54126523f4df7b15ab98bede45fd3 : SUCCESS in 4m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5a308f9e485341c7972b6d9446005c8f : SUCCESS in 25m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/07cea64d5ec548219de4c9a061daf293 : SUCCESS in 11m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d8abade7adb44b7e819ffa1439f1c20e : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f635814af7c7411eb63b422644873296 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01a27df59a8a47b8b7091aa0be0799e3 : SUCCESS in 20m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0eefa517c0f34770ad780dcc5ca292df : SUCCESS in 6m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d1026e205173476180c126a5f23da41f : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9387c4767f8c4955929302c83153b9e1 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0a101c764b9a41e7b098830a46e656dd : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0713ca9b7d9440bfb23a776f30d58767 : SUCCESS in 52m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/396d4abfa08c435cbd9bf085c5737d9c : FAILURE in 1h 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"be23b5884c52ba64dc2576b7e080595ef017fa83","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-18 13:20:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9aaca751b9d4114a2196ab26e2cec9642629877c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 14:46:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f8844b86b7462ea0c39a24abb5fe24\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f64fde9f3b4b2ca63d925d28b6bfe2 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05416642916a4df9a1d109e4daa9e701 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0427914521b54e028b27d1832e2e4eca : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a29a270b90c44dbbf41f69ac03a7235 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8253286ab1634d46a63e9a10e5140e23 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d70cb1eafa9c4cf9ad62a9ed3a478a52 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/033aadc802844c088a7c9b89e3b82cd1 : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4583a90df80c4956bed002363c732c87 : SUCCESS in 8m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e8a35b9a389f4c48882558cdb7c9780d : SUCCESS in 4m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/105594ae14c74f5cbd31f4be54fad40f : SUCCESS in 9m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ec3ceb60ae064629bb227db1003f9eda : SUCCESS in 4m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/957d891e23604c66bb3dd4255b82ad5e : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6209fdda9994d0382c206ce73fb7cb4 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9e375161bd0047999bab6d7b024b0b3b : SUCCESS in 7m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8fd9b376862246dfab188802a5de4512 : SUCCESS in 16m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0d89a0edbc5d4cbba327c639d671d9f6 : SUCCESS in 9m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/87db16e01d9f49d79d7073d86eeaa465 : SUCCESS in 4m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/aba3513fa83a4f2e9130445d7b9e528d : SUCCESS in 32m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/04eaef5fa610412abb27a4ff43008e99 : SUCCESS in 15m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0d5b7be2847946489dc219c10d9224f4 : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09883aacd5944392b8c1bc7190266bad : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/140fd92afb8d4fbfa26972d23405c46e : SUCCESS in 25m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/87ed7ea74da84564b31196184b6d9c2a : SUCCESS in 5m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dd6a2bd1c5924400bb3414d47ae1f20f : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d74143186c7a469ea2a2b92ce2fdb324 : SUCCESS in 5m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/730a86e4ce654a7abc5d37480934a6c1 : SUCCESS in 21m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/388810928ee24393b2ca438209694df9 : SUCCESS in 48m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d069ee632262474b89102d2a4e0da7c4 : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7052c822d1f3d26a53b7db7d297a24b7461bb8ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:54:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15555d4058b490e99bf29624cc23b0d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/69c244186d024b3fa6b07a074c5c6c42 : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"492343f48d24586448e8f79aa2361a041a2c6fcb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-18 16:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c38448e49607294e208758e15e412579490dae8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:10:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ccc4e4ac2e400dba04dcfa66a08cf13fb4d4064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:51:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f27363c641b437fb36e5c61b0be4631\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd888356da542328f332c14aea1d88b : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/deb2230790c548ad89948f68a4675cd0 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a3480fa2f3f4f40b893f8f479c5b1a5 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed2b705676d0494692610e1d4ad4d3aa : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/382b7a7a6d23466a8fd4b26a15299505 : SUCCESS in 2m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/edc49fe465614aaba5c7651a74f60052 : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63578d0dbbf44b0ab8aebbafaf8da537 : SUCCESS in 7m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e2eb2745dd8246e98cda5d40b4ff4d6d : SUCCESS in 8m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7d530825737f45b084435d4099659b7a : SUCCESS in 4m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cacbc3a6dce34b47ba5bdfacebb7505b : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4c27e8a750a4993bb4ddc6d76df2efd : SUCCESS in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8c7997a021fe46a9ac3692e30ddec8d9 : SUCCESS in 7m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cb2b396ec8c945a19abd0c28505d0081 : SUCCESS in 38m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9f78d5e682f340a3a1f715fe66cadbc3 : SUCCESS in 8m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5f04ce9a88904c838b776325cbec85fb : SUCCESS in 5m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/51fe5cc58b3e4fb78986930e6326a5f5 : SUCCESS in 23m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/16a76e7303464c2285e59b133e57a312 : SUCCESS in 12m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/69367496ec784b2eacfe4f43178ab46f : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51df155b33f74145bb3af57e13f8677d : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f10d61d6832c4d2a9dba78b994760f55 : SUCCESS in 17m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dc64bf3ae24d42968a234e32aa9a2039 : SUCCESS in 5m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/14b76086539f41d0b6c030bbcd115ef2 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50a5b3e40f9e40b49d68e583077d2ee7 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3391f7c52ec642a1adac6b4595dc640d : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d3342a3b06041a59cea71dbd25fcc0f : SUCCESS in 43m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96d5b3330fad493e971ac093f1a004da : SUCCESS in 1h 27m 57s","accounts_in_message":[],"_revision_number":1},{"id":"71dbe957cd3c4dde3b767f866db48cf982b53db7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:51:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee6d534ff331e42a92069bdfce6fbb46f69b38d0","revisions":{"ee6d534ff331e42a92069bdfce6fbb46f69b38d0":{"kind":"REWORK","_number":1,"created":"2023-01-18 11:11:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/870927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/870927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/870927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/870927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/870927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/870927/1"}}},"commit":{"parents":[{"commit":"ce4590a60086450e4b66c6b5d810c0bbb8258518","subject":"Use correct nodeset for requirement-checks job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce4590a60086450e4b66c6b5d810c0bbb8258518"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-18 11:11:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-18 11:11:38.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.15.3","message":"update constraint for neutron-lib to new release 2.15.3\n\nmeta: version: 2.15.3\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I7f6b02532c004b612b0978ead0ce6d782dfb5a35\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3f0536ce13fde079df49067bec3d486424bb43d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee6d534ff331e42a92069bdfce6fbb46f69b38d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee6d534ff331e42a92069bdfce6fbb46f69b38d0"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
