)]}'
{"id":"openstack%2Frequirements~870988","triplet_id":"openstack%2Frequirements~master~Iaa9d4b464553fbeb63656d91aee7bb8088ec4de0","project":"openstack/requirements","branch":"master","topic":"bump-eventlet","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-03-26 09:18:14.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Iaa9d4b464553fbeb63656d91aee7bb8088ec4de0","subject":"Update eventlet to 0.33.3","status":"ABANDONED","created":"2023-01-18 17:23:33.000000000","updated":"2023-05-21 08:22:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"951955b563cca70591e6222961fbf1e0457e3ef7","_number":870988,"virtual_id_number":870988,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-26 10:32:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 19:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 05:36:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"e4412ad42d0812a6d16ecf4d976470d567398fb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-01-18 17:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d01bb4c1727dabecaf8529a50ab9ab458d3b8f88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:32:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8968c8bf1be4caa86d4d6ef0fb23cd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77e32241e5b49dda56e700aeb0bca68 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90028a4ea5094d7388bc11e6d328378a : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/418b010f1945499c915d3e08c35d2e16 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aa9ef7151f94633a7a892b94ab1478d : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/368ec59fc957485ea58809e1f0059cf0 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ae0aa375b294eb8832411725e1a3909 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/36ba7c0501854c34b7fb761083981494 : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/24a6a7580ff14c2f9b9da314468a8eb3 : SUCCESS in 12m 45s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4d308da25cde43cfa854376b38b2849b : SUCCESS in 7m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7016d62fea8f4183a25d1ac82564a5b0 : SUCCESS in 5m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/076f58b68d56426c9ac293dc20d9f403 : SUCCESS in 5m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5a94a16a87d94ed48355cbd215a629e0 : SUCCESS in 9m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c020412db0614b92a27dc8673a74c0b2 : SUCCESS in 5m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/97ee6a6265004c5e8930cfdef796395d : SUCCESS in 8m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d68c60c1648645e5aacc97c274f9eed4 : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/24de37d985f34c8488368a7767ea2b3b : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d10c9851b484b5bb8b62d6f36cb712b : SUCCESS in 6m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/46759d5c0dc24bf78d864f9c83c98096 : SUCCESS in 9m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c4124672f73144378ab2a727f9b41ab2 : SUCCESS in 9m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b411672a41964895896e50c50c173cf4 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f1893935dd4049b88e1d212786aa9022 : SUCCESS in 9m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/35d76ffd3b584748bc443a0bd59795ea : SUCCESS in 4m 36s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/344c9438fbe64863bb7f2e3e673d1bc3 : SUCCESS in 9m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c1095e60dee440d59ee5aa2157c64f9f : SUCCESS in 29m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ab34ad4cb97444f4b8061675f9b6a739 : SUCCESS in 13m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c617f7da7c2c422f8597652bbde890ef : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/03d0eb103d834c9c84f1b91e1b475dac : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cb759118fe448a7a550be905bb59510 : FAILURE in 32m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fe68e2513f424a2f9057441f8283ef49 : SUCCESS in 5m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fcfb5bdd4fe445d8979dd613803f40f1 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e10c9b40a3e4ea8a62d6585efe76efb : SUCCESS in 6m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f4429cc4d30641948285d8680984bc53 : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b900d9c14c284e3daf30592d744b7a35 : SUCCESS in 1h 00m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6231aad33c14411389390f1023dc2855 : SUCCESS in 1h 47m 23s","accounts_in_message":[],"_revision_number":1},{"id":"af507a068076265872dc8ca042d19d55ba929075","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 22:47:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a22026f3fd1a4ec7adee6b85235b1205\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/88db1e730fb24ce1a9910a8aba72f668 : FAILURE in 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2eef38084e9b79c3adafde92b9ec2e1ef0c15d89","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-25 05:36:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fdb440c2cf752fa06c942a19939d7f389d416009","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 05:41:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84d8864523e846b28052c880fa05c61a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f01ff7306fbe492b9ebb8ca2d7aa7de0 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d38d45aebde7d66cff42e4df58de6692b326347b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 07:59:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34c87b4a04584f27985d2247373e9f18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22ecdd17fa44014ad6676a91ead5a73 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b615f18c05471981341f2b33570651 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de8cb1a1ba6a4e15ab25059243476623 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c67683c1d394e41830fff5b142a6ee3 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d318e00d444e4e4193ed717016bcd05c : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8cce709ac1ef4fd08e1f889ba78f097b : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1f43f6e898b4b088ac60b7bcffcff47 : SUCCESS in 11m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/210571b87304406dba31bead8f6f38eb : SUCCESS in 15m 10s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e2116ebbb1f644848b818f12567bbfc2 : SUCCESS in 7m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b242b81a440b48eeadd55ec044fed576 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/975ee844fd1a41ac872a93ae73f476d9 : SUCCESS in 7m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c491b43e69634917963fc4cc501a1a25 : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7319373cf6894a2c9d455872f4031a1f : FAILURE in 4m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9c23253722654b4cba091bbe3e472c57 : SUCCESS in 5m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9ab75a2106604325996b2eb83f1e2466 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/50ffb8539aa64e6c85ede74ff7280bc8 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d6f19dbf3f74082a8842b3a198c3032 : SUCCESS in 8m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fbca8a19948a41e1bc6fc5c6200cf063 : SUCCESS in 9m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/65197fb165c94b40b1ad24d241468585 : SUCCESS in 16m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bd13390f4d694f8f8670849846190189 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7cf68e44f6ea42708bf5425774896dd1 : SUCCESS in 16m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f57800e1d9ff4bfa806b5fe7a34a03f8 : SUCCESS in 5m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a792f8af330e466d8a0ecab23d6deba5 : SUCCESS in 8m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/51fa9d054c7243dba59c195cc5f0b989 : SUCCESS in 33m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5b5eff9e21754811b4493a4070eab5dd : SUCCESS in 12m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bbfd0ab6de09424da43e375dea2a3833 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/112556553f624d3c99c5ba1ef6e8a308 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5510219723b04030816666c690d35b59 : SUCCESS in 25m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/47f7ade283f044fa9badef5ff9993db6 : SUCCESS in 4m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a8c757e75a174727bb20d6b281c147ad : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f11eac7602af471bb582ba69ef2f381c : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8d2ee0398d0249478a0b39c9bb2385ec : SUCCESS in 26m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d6215817bc44c3ebf09a5101230a383 : SUCCESS in 45m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f01edd3f70d84d2fad61ae3db300f4d0 : SUCCESS in 1h 52m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3304b0cbdb0e1c5228667d120af1c13cbd9e8cfe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-25 22:37:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c95bc7051aa5a0e34d81e6a2d89a6e6c6cbe4753","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 22:51:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea5f31028c64564abf111e24dd87ca2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/317929c7a7564adea4471f1ee3a855f0 : FAILURE in 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"db68120667b888f6f422c7b2b4b35ddc8f2c9403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 00:04:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2a494a22f3c42cfae5a405699ec164c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8998eae8647249ab95b20638efe23e34 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58196845927a4894bdf06f2048e51195 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fcecc8ab1fa4288a013c67723c1fedd : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ad024c4e8c48d1b8f34368dcfbb347 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f599cdce525541b4a68ea8d60d219dc0 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5eda385dbb04a72ad3d077d7b070f67 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9515599b71f64bdc9ea66c1cefd0ee68 : SUCCESS in 11m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a9411f67ef4b4de189de55ceab723879 : SUCCESS in 12m 51s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/86db00c514ac4a398e473f8ac31508c5 : FAILURE in 4m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8dc756f6f4ef407499ecfa33f7bf4a63 : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d377e8c297bd4205a7f7f929382ae33e : SUCCESS in 6m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c6b09961041d485f9ff3774d71c36398 : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4dc15d88bb5f4876b3f1b4541dea5142 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7c8d2351e0e048b7ac074c44d28607fc : SUCCESS in 4m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8e8048b12a4549cab29ca138000d330d : SUCCESS in 8m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/026d0c1a44e746f5be4869d7641324f4 : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5670d3984684409fa9cd2a28fd029c4f : SUCCESS in 8m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a1f4868ece164cfa9196826c9ca0e9c5 : SUCCESS in 8m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8ff47752721549dc99ab8706ea937a67 : SUCCESS in 11m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2d72ce34284e432592cc811068de46f5 : SUCCESS in 4m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/889529468f754d24a53ab80b0897757c : SUCCESS in 8m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b15496f1abe04e719d6aa04a711ed98b : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/639b16a457b548f5b033801b4b093eaf : SUCCESS in 9m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/95f2be70ba3a4f0b9a0c8a81dfa140d2 : SUCCESS in 29m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/80506f7f80aa40d88513577cd18207cd : SUCCESS in 12m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/29d60911019844ab948a2a524b5be1c1 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69bcac58b46d48339ce590c1ca33e108 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29e1cac2300f4fed8bfc216ed0c59f28 : SUCCESS in 22m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ae45669ad1634541bf26c03ca6ec3a67 : SUCCESS in 5m 05s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d6646e91e3ab49e1953e8f79eac2b37e : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/133a534d953241958502fe9f22858ea4 : SUCCESS in 5m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/66a0b43d2fa7477d93ef2005d7386e21 : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86cee090a19d4323a20bfcc8dd6fb155 : SUCCESS in 51m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7c6db43ea054842a4d5ec34cab327fc : SUCCESS in 1h 24m 16s","accounts_in_message":[],"_revision_number":1},{"id":"987f8b2cc94dc1a2996c12319e2743f73986a17a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-26 09:18:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7a82b7e113d34efbd50c8864af03d7b7b70c31f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-26 09:23:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f2556e9d95644c98ead1a8171954f69\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/252cb63a61154b5eb71fbd3748026f0a : FAILURE in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"91ca4246f3eeae59cbb766c82b71b2570eb6a8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-26 10:32:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5443712ae4cb44cca5dbfff5447a4f2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3043da713b974027bbaa9c9e3f4dac0c : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cda60bd419d4613a6d56fc0880a4fe3 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2ed2bdd47e34fc88588c78574321bc0 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12b9ee38e1e64204ba1a113aad357147 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13e0e676e5ae4aa8a80d817620f941b4 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55c5e8b80ca64af49355e6c571f0febe : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2454e4ab595a406599958e7c2b47d65a : SUCCESS in 7m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/434ddfbd97aa41e99d66d11a80afdefe : SUCCESS in 13m 56s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/712acc60f25b488f9b45df369f7999e9 : SUCCESS in 6m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c3acbe10072842fe9a51d6f5ff850b79 : SUCCESS in 6m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/966590a038584f65891de3f28472d1dd : SUCCESS in 6m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4df2e4fcfbda4e1a997ec4feced39c97 : SUCCESS in 8m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7360ea23d644c0a962f5c5d8fcb33d8 : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/00629ae90d9447758ac48648ad494296 : SUCCESS in 6m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a108c3803df44c55bad146186adb626c : SUCCESS in 8m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/532a159e28ca4994bbf266f4d8c8e7cb : SUCCESS in 7m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a76a4969953e4dfb83bdfb6c7d7611c4 : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4e0d3118d02749ab9921b2bb1da5b425 : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e12490a3f77244ff816bbf3fabf4d1b0 : SUCCESS in 12m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8cfadcdaadaf41909f86768d910dabe2 : SUCCESS in 4m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/54bb09af9da84ab99719071fa230968f : SUCCESS in 11m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/81ad062b772048998aab926b9f81cbae : SUCCESS in 3m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b50da7a64b4048cdb8e0cc563b44d635 : SUCCESS in 9m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/23830f05e72244a4bfcdfff2684d69ad : SUCCESS in 23m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6c9ea65c6ead445183cc7d370a0a74f7 : SUCCESS in 13m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/80b847839d7e4c1290d90ba0fdb6a372 : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/af0ce32fb61d41eeadde5626dc75c29f : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/428335cccefa4c59ab0a5fdd9b87c0e1 : SUCCESS in 19m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4142b6e68bbe4b11aa105da8274a359a : SUCCESS in 6m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/53dca3f843aa435fb920a4ffee3decd6 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92ebd92b51cb4af191105e3a1abc6a0b : SUCCESS in 6m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2675c801c56e4b67911b30c14d13c1b8 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f1b558edd814aeb814c488b97a0677b : SUCCESS in 40m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e026a643a07b435fba380b35167c91d1 : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"951955b563cca70591e6222961fbf1e0457e3ef7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-21 08:22:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"478f8e92ccf2b84c65cb62caaf0e578904bba41c","revisions":{"478f8e92ccf2b84c65cb62caaf0e578904bba41c":{"kind":"REWORK","_number":1,"created":"2023-01-18 17:23:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/88/870988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/870988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/870988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/870988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/870988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/870988/1"}}},"commit":{"parents":[{"commit":"9816eeb72a01e38a2cafe48e32ad277e00435a80","subject":"update constraint for pbr to new release 5.11.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9816eeb72a01e38a2cafe48e32ad277e00435a80"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-01-18 17:21:51.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-01-18 17:21:51.000000000","tz":60},"subject":"Update eventlet to 0.33.3","message":"Update eventlet to 0.33.3\n\nThe new release of eventlet fixes a bug with latest dnspython[0]. Let\u0027s\nupdate this first before dnspython itself, since designate still needs\nother fixes applied.\n\n[0] https://github.com/eventlet/eventlet/issues/781\n\nChange-Id: Iaa9d4b464553fbeb63656d91aee7bb8088ec4de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/478f8e92ccf2b84c65cb62caaf0e578904bba41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/478f8e92ccf2b84c65cb62caaf0e578904bba41c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
