)]}'
{"id":"openstack%2Frequirements~871001","triplet_id":"openstack%2Frequirements~master~I952dae2967ee328fcbf2720f04791086f4d327c8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I952dae2967ee328fcbf2720f04791086f4d327c8","subject":"update constraint for oslo.db to new release 12.3.1","status":"MERGED","created":"2023-01-18 21:24:37.000000000","updated":"2023-01-19 20:11:56.000000000","submitted":"2023-01-19 20:10:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871001-new-release","meta_rev_id":"ffef9d3114aad6386477be9bb51437ab66c5598a","_number":871001,"virtual_id_number":871001,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 20:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-19 17:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-19 17:41:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 23:14:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-19 15:55:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-01-19 17:41:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5071bb157b05cf36713b4309768e7200212798ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-18 21:24:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66f6e627eca688129eb314716e5f2ef55e6d61fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 23:14:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/689ba92b41f04831a1ce91a34f367f0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414cf2f6628d4220a1fc5a52c28ecc80 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25deb5364604cfeab9375cc13ff69fc : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b2ce423265e467f9b35240db8c90325 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72f7906b88e8486db5e8a6e06de96f5f : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7579c0035dd456ba3ac666bc375d173 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/47f1ddb4fb044416b388349fb160f395 : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e7f5821ecdd54a3e9334afe1da607f9a : SUCCESS in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69def63788114581ad121943a33e4094 : SUCCESS in 13m 50s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5c20a59400b048f19255aad1d7bb83ec : SUCCESS in 13m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8f9fcb4a502d49839fcf70ead556d5bd : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bc8e63951e9d4406ae82fc4ee065bf52 : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/eabf99b464f947409605b700190e0c16 : SUCCESS in 13m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ca98fea4cadb47ba9a706fa3ca2a98ad : SUCCESS in 4m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/91b39a6214aa48cb84329b978d4bbf5b : SUCCESS in 5m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ebabbedcb3234a2b9ab8b6efbc801e1f : SUCCESS in 10m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8f0442c74cd1442ab923b7f85ddc229f : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f803afb371a4b7da9b23322cacb3dba : SUCCESS in 7m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ddf15a3391ea45e182313d87e647d9bf : SUCCESS in 10m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/73fd505d9bf94ea6be3e18cd350ff01f : SUCCESS in 12m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/715e77f897054bcda33efeab72f6d85d : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d03b7f4695c24baea4d8a9a4125f2974 : FAILURE in 17m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/24ad53ab82a44f5589ac98a15df877f5 : SUCCESS in 4m 30s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a5d410287a404398a2adeb39ee67d299 : SUCCESS in 9m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/523ae49ca49f45ffb340c97d7340d73f : SUCCESS in 22m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/52ab627086ca444591fdcd9cc33d1d05 : SUCCESS in 22m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b44ea7617f0644119be32ce1632e902f : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/82b6c567745c4a1991ae8e6a74fa75a9 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8fa5c43ece7468bb0636ecf18e27402 : FAILURE in 23m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fca86c039b794793a766c91b9478608a : SUCCESS in 6m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ccdea242f293418dbc9d4f9c8ad6354d : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07239482446a4d88b4b794a8940a4369 : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/43cf1d0da7504b50b7839ba7e378d1fa : SUCCESS in 19m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7d7b5879cbf469ba6a17fee8283dbc2 : SUCCESS in 57m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8b488a965bc4c34a8c7ad55be05d5f9 : SUCCESS in 1h 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b98f883e441e065deb2c2baab7928b5f4689980f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 23:17:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdc2af4d52ac4485a64ad22394e3a26a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ae40be378b954443a71ccf2168b01415 : FAILURE in 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f9ba24b23195ab4d99583aafa374a4e99a6335f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-19 15:55:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"350d8864ef99e93d2b1b07dcdfc377002ef67cd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 17:24:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a6e0770f8464454aa9c94f13d73a4d4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b5595f5f813348ac833d0eee190a9a09 : FAILURE in 18m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0e4bec2feebfa8bfad3f11240d75d1227df6d922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 17:40:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a04a80effe734bb2a523bd2996054378\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026def3a4c224633a4976d0ffe867dfe : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fea2a990198456e8c05f33658994d61 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ddecb32cbd643c58b66fe4440390d68 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/721dc7ebb2b24e8084a6eb79670ec205 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/427ca7fb355b4740b6ed0178e688800b : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/610076e8630a4813a8c7537dbcb03163 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a17cd6cd871342328544461f49b3dbd1 : SUCCESS in 8m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/af500a1c52ed4cf4b908cbd941eaea92 : SUCCESS in 11m 45s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d1e4732b0b5c4b7c91f8c744bf28541a : SUCCESS in 12m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/20e49d1a028f4be19333d71898bfd0fa : SUCCESS in 5m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/32da98743c6744fbb5363bb944d920db : SUCCESS in 5m 31s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/dc6043f96814420bb575155cc7d0c5c2 : SUCCESS in 10m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/53f1c85d00664762ab712f95d1299091 : SUCCESS in 5m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/00d77e17b301444c8cd36a9ec55878d2 : SUCCESS in 5m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a3de630800bc483285b9ec686c2ec279 : SUCCESS in 8m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ba38d8b93eec429db96682f91fd10915 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ae120fa4edf4891b362aa05fa968613 : SUCCESS in 9m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/96348ef5322e4f91b9424803c09036b9 : SUCCESS in 7m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/65ce8210ff7f42c08cd77073324f9d9c : SUCCESS in 10m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2a26cf7c8a0a4743b14e1c8290c61e8c : SUCCESS in 3m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0ff5208a1da748d5a8598eb8a5bc91dc : SUCCESS in 10m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9e42c766e1c94ba98da0edb2d81fc2d5 : SUCCESS in 5m 21s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/371a0b821cdd48f1bd07f6a7816e07db : SUCCESS in 10m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d24ee693024d4d819ef2dd6fe81b2fce : SUCCESS in 28m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/99e4d2a5c2b4422fa27802572fdb77a2 : SUCCESS in 13m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/332188ce649347d3bebe2eb65e7ce952 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25d3a53aabc649f28813a3d754cd9834 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2e08393dc8d4214a2e48f03aafee4e6 : SUCCESS in 21m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8e02239c723f4ec58c274e94c37ba68b : SUCCESS in 4m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c9310cb893ef4b2cb7a46c8579d72f60 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66de81392898434daa804d2a0de903e7 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8893589402de4014a42ea0c13bcbb34a : SUCCESS in 17m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa100f2a288049a484158563d9afe78a : SUCCESS in 1h 04m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02062560a36c478fae373b645882286e : SUCCESS in 1h 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4c0125ed8f294377c0e19e5706c89c1e5a9223c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-19 17:41:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33cf4ef0de2c852304747082395ef53ccdee8b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 17:42:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09d25e5ff5ef8361260d809aa158660e4b7912cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 20:10:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26033733fae4c3790da84fcb3356329\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050d8b7719194462b66b50dd9053e7b5 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214b2474553d4400967840bdec9af1df : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60b229a976a948c5a40691685beb6f07 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aaba1189e7584963b2c3d7ba6f41b6c1 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34fc67d4a83848168bdbc99a0f4dfd05 : SUCCESS in 2m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8f8fea2ecb541ac80ab9e5d51251234 : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b4c14560c094e6d91bf9074d833d3d3 : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/87c96fc21ddb417b928bb7fd7d781f17 : SUCCESS in 12m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5e6b263324c74bda9cfab3fa0089889b : SUCCESS in 9m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/aa976c2e2d1a42c69e4b4d44f61946df : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fe50056029d247d7a3a22deed435675c : SUCCESS in 5m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0d1546eccc164d0b8ab4aa8cd7efa416 : SUCCESS in 7m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c999a85eb7c24b30b1d838eade323907 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d7c771e6d154b018729df5a6c7f66d7 : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/51fbe3df3b654d2abcad40d5f5f95303 : SUCCESS in 10m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/86c8b130fd484dbd9fd8bbe777d5ff15 : SUCCESS in 10m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/278afb9fd75e47438db3453fffb6f069 : SUCCESS in 4m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d4cfda01c89f4cf290a01f048c0ae9cd : SUCCESS in 9m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a64b3b080c514f239bc49c42f8f63e16 : SUCCESS in 4m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3e568819b1604e43a82e0e98622d1f3c : SUCCESS in 23m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4b63052448e943dc8d05344a65a2cb7e : SUCCESS in 9m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dcaa90b97c5e42c89c6d62cad64b6382 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/362a637041dd4b9797d7a6f0b54771e3 : SUCCESS in 2m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f996c0ae2cac430088a8f512e3f9e2b6 : SUCCESS in 21m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b4002e75a80a4d03a54a8ab347df6f03 : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/11869904f57d4f5b817f7879f222377c : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f4da5553ac2467fa7b4ba1f2c35af4a : SUCCESS in 5m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/018c6f5c2cd843f782dff8a03a78dd1f : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47601dd5a4f24077af775873ca286f93 : SUCCESS in 54m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/721035b6b9e54f8a8497244155101f3c : SUCCESS in 1h 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f2a62bee78eb8cba05a6413f5d6fcbd37b2f4035","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 20:10:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ffef9d3114aad6386477be9bb51437ab66c5598a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 20:11:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11fdf5be967a47f7b0751bcf4c119a2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099ddc1f37ce4a139ccaf5d2fe589b9d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b","revisions":{"3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b":{"kind":"REWORK","_number":1,"created":"2023-01-18 21:24:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/871001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/871001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/871001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/871001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/871001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/871001/1"}}},"commit":{"parents":[{"commit":"9816eeb72a01e38a2cafe48e32ad277e00435a80","subject":"update constraint for pbr to new release 5.11.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9816eeb72a01e38a2cafe48e32ad277e00435a80"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-18 21:24:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-18 21:24:36.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.3.1","message":"update constraint for oslo.db to new release 12.3.1\n\nmeta: version: 12.3.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Id49b06e92e47926827384578e0af00f5020a0df6\nmeta: release:Code-Review+1: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I952dae2967ee328fcbf2720f04791086f4d327c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
