)]}'
{"id":"openstack%2Frequirements~871065","triplet_id":"openstack%2Frequirements~stable%2Fxena~If13f34c23fa828ec8d6fa1039d246c2fc13c61cc","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"If13f34c23fa828ec8d6fa1039d246c2fc13c61cc","subject":"update constraint for os-ken to new release 2.1.1","status":"MERGED","created":"2023-01-19 09:59:04.000000000","updated":"2023-01-19 17:44:30.000000000","submitted":"2023-01-19 17:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871065-new-release","meta_rev_id":"cc5b6f98cb21b7c01a37b152360b28e8ff73fe85","_number":871065,"virtual_id_number":871065,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 17:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-19 15:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-19 10:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-19 15:55:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 10:07:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-19 10:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 12:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-19 15:55:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"07c3945e75cc04ab70666ffd87523969e47a74a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-19 09:59:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a1e3689958331c0c55a1ce9a78b96aabd345a14","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-19 10:07:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcf05cdf5cef9be20123294c295ae0910d00b3c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:30:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07c465dd454e442da80ab86766cc370a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/80e59cecadb64dd8b31144dedde82af3 : FAILURE in 19m 35s","accounts_in_message":[],"_revision_number":1},{"id":"11ad3f69cc46aba4ed8ab76193ff8e0e9a405215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 12:00:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/747cc6b4dc014dc08e98f2775814d9a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f21766bae1364a7cabef461bb74b0319 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0d0fe7f367c4e48848bd1cea21be06c : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/870ae39f014a40af93a3334775da0568 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70d6799b077749d3be776e54e1599012 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8f0bb69657e4258b88edba9d641ae96 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a830a447f05d462ca3e0275960c5e570 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f4b7e827948e4d30a61427da27e13c8d : SUCCESS in 8m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79d69936d9754ee39d119e4d56fe3c2c : SUCCESS in 8m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9ec485028fa7471fb975ca23dbb88329 : SUCCESS in 5m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5926dcf84e4643a6b86d92340d3d2be9 : SUCCESS in 8m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/389473d5f90e48d8a1f6ce3ea5131721 : SUCCESS in 5m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/68bf5c137cd8430ab60c4a5a1b28c08c : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e654765e57849b7adb7ab0b2c6193b2 : SUCCESS in 6m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a0dbae891c04b36a2cada58e4299c69 : SUCCESS in 8m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/00b438cdc1b2431e91bb665328ca78e7 : SUCCESS in 14m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bfa1d4868e1b4d8eb3ead1d39700fb3c : SUCCESS in 8m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9221b4c2ab294d839c75aac910322520 : SUCCESS in 4m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/13292d5bfce74ee6bef01db7723c3c1a : SUCCESS in 24m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a2c9ed8068d444918a2149de973f2739 : SUCCESS in 11m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8e94c8f41f6e400aaae02115d10758ed : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/527ad108a9874794a34cb7cb86b6e6e2 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c332dae08c00463fbf04a0824ba1a1de : SUCCESS in 21m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7ceb3e91bc8c4386aa5db8380484131c : SUCCESS in 5m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9ee44aa076994cb5a048e6f6fb3be802 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/011a38a463c14c85a15bd473a2a9f976 : SUCCESS in 5m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/04c548e80e864cc1a7e0edd97f94cdd2 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/801cc34af3d64ddfaa5325fe2b6e9d3c : SUCCESS in 40m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9d0e5871aae4fc686f874d240f0bd04 : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b3267c875d93ff9decd6da00f5362da4ce212173","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-19 15:55:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f42b7fe161735e0b4e6629e8d0604a70d0ae162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:55:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b75a14b3066fcd94ecc8514e3f8a1c3bfc0fd8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 17:44:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b08a0da80e42fe8727dfcbb1e6459b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73df36bb81e4989824837e26123f222 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7adcd0b9fd8f40a693167feec60b08a9 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3d5f7cd890d4515a28d04f39bf19a56 : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32be499813f546dc89f0a340cb9272c8 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3e783afba7c4fd694e24a0ca78c421e : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c455e0252a4448caa4ea7522251a45f : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bebe115ac2114b7d95379b8b1c072e0d : SUCCESS in 10m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5622f2d327434b869ef771aca51d62a1 : SUCCESS in 9m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/448e5a9d7c3c417893316cb6e04eef23 : SUCCESS in 5m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/16f081bf0bab40868455447daa6afe97 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e234264c48c845659918400ea33910e6 : SUCCESS in 31m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/781a2ff3e94040948f10b4705c8309c4 : SUCCESS in 6m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/add1cde4e14b4318a094ddad971e6237 : SUCCESS in 11m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8ef4b86669b6426e95f4817e9bfb4bd9 : SUCCESS in 7m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/391368ac20ce4ff48bafebd123c1bbbb : SUCCESS in 4m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/de3c4ecebce5490e9d93791e017c9b23 : SUCCESS in 23m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/24585a9de4284935bb17b45a7cf555a5 : SUCCESS in 11m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a968e3c9a3694fc2a238fc20ff80897b : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4eabaadbfde242d9ab248e3a2db683cf : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/36a1bd27e57c4c248f982a9666e51196 : SUCCESS in 18m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1ab078ca8e824852a0f202171ab204dd : SUCCESS in 5m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d7b84df8255441aaa52f79e4274ab793 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d430f412531e423db1dfe636f24652af : SUCCESS in 5m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b5107c9b634f4a2297165c13f75d4e51 : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb6afa8ac9834b609a18adea12aae11c : SUCCESS in 53m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f13c61676a5c40c7b2eeff24625d213e : SUCCESS in 1h 29m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cc5b6f98cb21b7c01a37b152360b28e8ff73fe85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 17:44:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d28d724039e6fe2380fa8a7bbbe637cef3e32fd3","revisions":{"d28d724039e6fe2380fa8a7bbbe637cef3e32fd3":{"kind":"REWORK","_number":1,"created":"2023-01-19 09:59:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/871065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/871065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/871065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/871065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/871065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/871065/1"}}},"commit":{"parents":[{"commit":"b428cc20779e324c310e016a7932c8f3dad7aca8","subject":"Merge \"update constraint for ceilometer to new release 17.0.2\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b428cc20779e324c310e016a7932c8f3dad7aca8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-19 09:59:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-19 09:59:02.000000000","tz":0},"subject":"update constraint for os-ken to new release 2.1.1","message":"update constraint for os-ken to new release 2.1.1\n\nmeta: version: 2.1.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I40baa43a2f5b3fccb6fe65352c20b445aa746035\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If13f34c23fa828ec8d6fa1039d246c2fc13c61cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d28d724039e6fe2380fa8a7bbbe637cef3e32fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d28d724039e6fe2380fa8a7bbbe637cef3e32fd3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
