)]}'
{"id":"openstack%2Frequirements~871454","triplet_id":"openstack%2Frequirements~master~Ia7693835a755366934f99942b1ef9ff01e2a7eb7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia7693835a755366934f99942b1ef9ff01e2a7eb7","subject":"update constraint for python-zaqarclient to new release 2.5.0","status":"MERGED","created":"2023-01-23 09:58:52.000000000","updated":"2023-01-23 13:39:07.000000000","submitted":"2023-01-23 13:37:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871454-new-release","meta_rev_id":"cfad8e643e634ac8d4d59ef9459535da1a7c4078","_number":871454,"virtual_id_number":871454,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-23 13:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-01-23 10:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-01-23 10:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-01-23 10:01:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 10:00:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-23 10:01:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-01-23 10:16:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-23 11:34:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7cbe6cf4bbb630f54eeaae9295ee07e68dd354e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-23 09:58:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe96e62453e1d21d41c8b7290392cca9059815d4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-23 10:00:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4d1eaec2ece33ab2b804d20e8a503bf8f7cbb42","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-01-23 10:01:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e58cdd80c664454b5647450a7103349c9447bf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 10:16:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0923246615944ef1a72073eca20ab09f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af064d5706e94a638e2d609767124e16 : FAILURE in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"14b84ebec93fc37e7d924b52fd146e2b2a0394f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 11:34:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/222d19899a474e929df81ff06bf7c267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc1d1058465499c95190a71f33c03a7 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a8890536504bdd9a185d9b837db36a : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2cb5aecca68462abe07452a7f84bed4 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c765666fbb14820b9e31cb2b216c2f6 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b810e656fee433fb8eda8d2847085b4 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53bbcb95f3bb459ebf2c526a68d5020d : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a667eb4a881e41beaa7e7390f7c41888 : SUCCESS in 8m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2f2ba7738c647bd85a89724c1e08d96 : SUCCESS in 12m 51s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/941d1a8800bd42a1b96ecf5727dcd126 : SUCCESS in 6m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7e29afb8b237413492e9f73f0c5fb66b : SUCCESS in 5m 07s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a24ec610d7f047cdb4661147090957f1 : SUCCESS in 5m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/decb885e729f4075b8c6d1306b76297e : SUCCESS in 10m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2568941cb6a34dd1a929000912c38dc6 : SUCCESS in 4m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/01bfdb2d1199428993f8ac87a88e1521 : SUCCESS in 4m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/45951d24b5ad40aea110d50ea618a492 : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e8b56e26423944d7bc90c7e1a50e5860 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9cd1e7f1ff34f7592e3c47edf00f3f0 : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ecb4dee416cc44928ce39d161ed3ef86 : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e2f36537e16a49178b455140596903a5 : SUCCESS in 11m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b1829ec26dc048cf8dbfa3aac96bdc1f : SUCCESS in 3m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/217731b753a741a782f9a2c47977d02d : SUCCESS in 8m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1917f67711654283897db33995302274 : SUCCESS in 4m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3033e289f6f84b71beeafc7b0988ad75 : SUCCESS in 9m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/41b784d5573242839448913652ac6930 : SUCCESS in 25m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f724a46216484d748d8e691450785a60 : SUCCESS in 11m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/34fda1a0a4824594b4783f6b2aea404c : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9044b9744b8341d9b778731d8e743a3b : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ddb126eb4e3a47b5b812c35ccd700b98 : SUCCESS in 21m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1184adb826194ecea6d8525235a24d86 : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ff24cba6fee346dd83611560a69fe751 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f74739442a9f45e1be4bd23943e5a272 : SUCCESS in 6m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/17c2ee4371534a088aa43e1b8d9eb4e2 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1981c97b3bac4de3bb6a282ca2802ca5 : SUCCESS in 53m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c85bd627f32b409cad59400bc2d0873e : SUCCESS in 1h 33m 20s","accounts_in_message":[],"_revision_number":1},{"id":"34c184575e73445e4f6d704d402240242c1225a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 11:35:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7eba9af757fad1dc9bf83d9d0708e3a3f4a4b6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 13:37:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c016cda468447e95b5c333e6318959\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d2768df97f647ceace4e7f145961279 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc577dc672434b46826f9762555cdd6a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5af4bbcb17d49449c744fffc8abb6ad : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d15cd19b6edb483286459a2b30523760 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6a7690b193c478bbb3526b5b3a4700d : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b0122317b224cc487054220d0f7cfe2 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1b23495a8a5e4bc0a4e2e22c98928c9f : SUCCESS in 8m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5fd7365b2dfc42f0ad504cf54a810567 : SUCCESS in 13m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/46bb330208314068b2ae842898ae9e0b : SUCCESS in 10m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e93388835d43427f83cdcfc594fbd8d8 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6a03fcefcc5e400cba2005f4b9bca101 : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/48f0c68d0f7e4962860526f45f136cd3 : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/894e5ace13cd4a369f6171e5b8f58a9c : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f772af8141104f1da6ce1186e2f4177c : SUCCESS in 7m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8cbfe6dae85247b5b70ac37d7013d58a : SUCCESS in 8m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/21972c991d8648b8afbd45973290c8a7 : SUCCESS in 11m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/73568f47a46143a4b613191408941a07 : SUCCESS in 5m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/44e21a33e900453f81ab826ec227d00c : SUCCESS in 9m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/41515b4dc4cf45759b3d89567bbc9644 : SUCCESS in 4m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/73087e4b8b734f1a881b849d213f110c : SUCCESS in 31m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e5eca098f9fd46fc8245a84ca13cf6a4 : SUCCESS in 12m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/85bff91dc3ae4cd697891f66fc43d481 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4bf57d16614f4330a6cfb57eb638f748 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9de247fdbdad4081bbf960066e798ee8 : SUCCESS in 30m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b71ba8bbdfe54ea284c6595f5c35dba3 : SUCCESS in 5m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/be380cfca1dc4307ad6a0fb8ddd90cd0 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b457bc086664becb21c90fed91586e7 : SUCCESS in 8m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d551a4312b9d4b88ae54ab593ffabdf7 : SUCCESS in 7m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/584388184cf4400eb8aa82ae548c76f3 : SUCCESS in 49m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b79eee4b07004ffdbb5938d86708d4a7 : SUCCESS in 1h 55m 17s","accounts_in_message":[],"_revision_number":1},{"id":"454096bfcecd9f7128be7f5218a3612c0c4d705b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 13:37:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cfad8e643e634ac8d4d59ef9459535da1a7c4078","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 13:39:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf02abc5d4349e8beeb5eb5d32de89d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da7fe9d9b3f453eb635fb0f7bf8c667 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ddee9b28949d5b3fb2179513184acc1e8d838b5","revisions":{"0ddee9b28949d5b3fb2179513184acc1e8d838b5":{"kind":"REWORK","_number":1,"created":"2023-01-23 09:58:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/871454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/871454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/871454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/871454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/871454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/871454/1"}}},"commit":{"parents":[{"commit":"3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b","subject":"update constraint for oslo.db to new release 12.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-23 09:58:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-23 09:58:51.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 2.5.0","message":"update constraint for python-zaqarclient to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9b41ea50a05315d8035a9eb465c3e01c004281f5\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia7693835a755366934f99942b1ef9ff01e2a7eb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ddee9b28949d5b3fb2179513184acc1e8d838b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ddee9b28949d5b3fb2179513184acc1e8d838b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[]}
