)]}'
{"id":"openstack%2Frequirements~871463","triplet_id":"openstack%2Frequirements~master~I276a5f0c769db71e7258a0abfe56f2ee0d145025","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I276a5f0c769db71e7258a0abfe56f2ee0d145025","subject":"update constraint for mistral-lib to new release 2.8.0","status":"MERGED","created":"2023-01-23 10:47:38.000000000","updated":"2023-01-23 17:46:22.000000000","submitted":"2023-01-23 17:45:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871463-new-release","meta_rev_id":"8be33ea449375fd9aad3e63af9bbb40d88f24696","_number":871463,"virtual_id_number":871463,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-23 17:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-01-23 15:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-01-23 14:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-01-23 15:40:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 10:59:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-23 12:21:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 14:21:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-23 15:40:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8dfd09d916aaaa9c722c96722053db4a601d539c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-23 10:47:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbea3b87040eb7a37397eb927c9b0582bf7d3b40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 10:59:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa3bb4c105e432cac49cfd70793a9e6\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4fafc978016d448eb2544e2e43ade18b : FAILURE in 8m 56s","accounts_in_message":[],"_revision_number":1},{"id":"75dca8ed9c4787190c8a2b6e78613f762cc5cc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 12:21:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110c91ecca924457b0e9cd9e80107d00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd5cc2d96b54f79a6fbf53f74283791 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877cabdd0b084f4fa1450a258467f738 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71bd2b792b054d009bdaceb2e6a79eb1 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35353e25bcca4f3ba1f2e6531dfd6d00 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a314bbbdbb64258ac2c9a0bd0307af6 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/881184f9b62447db8e2d5a1d74054c5f : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c3f2cc4a5a4149c899bf790a2315bebf : SUCCESS in 9m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/30d3e9f864d348f8b82017c06bfd78ad : SUCCESS in 11m 04s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e93b94a6e68b473490998d791d932d41 : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c6956afb0db540c59dc56d770e6d5d96 : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2045046664a84600aee882e30caa418c : SUCCESS in 5m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/834b4cce27784e58b001f507e69a5ee1 : SUCCESS in 10m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a3331e007b37404ab759cfb36ba0a41c : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f780746afa154b418bd4047c6145e4a3 : SUCCESS in 4m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0fda38a633bb4ff4ab55fe988c669d16 : SUCCESS in 7m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c587f9a18401434b812e0780433a308f : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/763334178d754df9b38231f57c87373d : SUCCESS in 7m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3872fa819c0e42ebacfacc0d5c75d38b : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/518c2a6f0a514a6e8a9b1a4f8411e518 : SUCCESS in 10m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c44c180236ae4be689704ee703479c3d : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ea6455e35b9a423bbbedbfd2628a9b3d : SUCCESS in 8m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/64d5716a5a4a4cd5b56285bb6ba99d12 : SUCCESS in 4m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3acee069e98b4e038eedb590bec0beaf : SUCCESS in 8m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1bb2f958ef134496849666d78d763d34 : SUCCESS in 25m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dc3167f0fffe4546b56b7218ec0d490e : SUCCESS in 11m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e9bb051ff03b4ffe9ea2375f0f31b38e : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a8873839faa4e2db25f5cb0c6680b1b : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbc172d63ad34daa84a0e77fb1fa9412 : SUCCESS in 20m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/abfd740b282d4bb59022187c0b6d4777 : SUCCESS in 4m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e16cf462d0cd4c8197c058deb7a2fc49 : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9be2a79b2de0400f8bf82c587a15573d : SUCCESS in 6m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/989e6c8a272641c38b68c2378409eb31 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d45d8a29a12a47fba199e058283355f0 : SUCCESS in 56m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfe6d1b49c684e1396613c3be9db6f80 : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2421d37098bbf9ae116831f3d3077023b8aa4ae2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-23 14:21:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"286352364e02c346a10c6a2081e4fda5484e11b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-01-23 15:40:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2138c3c9975a7969a58311451f1c93ed4414a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 15:41:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcf654c7c8c3714d0fe0e98ad8cccff6e0c035f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:45:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b7bab9b8b784d418fb8e40763744201\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11256405b4b4810b0f3dc15e9c9ac91 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8718c8b6eb084270a3f5f4531915e3f6 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d67d5695b6914be089807d21e9c7d4a3 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c280a2146b54c0bbf360ca40ac83f2d : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e10164a186d44263862869fe20ed8292 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a76e593aa3e458f94c9568cd80778f4 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42308d45f69e4eb5bec07c95e91a5bd6 : SUCCESS in 10m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4c18630b3a4c4015abfcd353a575448d : SUCCESS in 12m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1b0f0a2838f449febf9e4a9ac641b65d : SUCCESS in 11m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8e0a4752e8194aafaae9433276249339 : SUCCESS in 6m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/71d8ac45764e4c6893100c6307884bda : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/432306eb0da744fab30f006c1938fac3 : SUCCESS in 8m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/55c3d4cac6284e01be789002a8bb5d7b : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f870f9e508c94d68a1637270631ebd72 : SUCCESS in 9m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8f2faf5147db4f9190a5927e65ccd4c4 : SUCCESS in 9m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ee53785d61324be99ab8a54c7a668e85 : SUCCESS in 12m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0647f3e1b93a4e2296d0993201a385e1 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/26204ec046d645febb0df7abf318a979 : SUCCESS in 10m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2353d57cce1f47e986595c73d60d5748 : SUCCESS in 4m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9823f4e77d6b456f86644864ddfcdea8 : SUCCESS in 26m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2ef95aa269b0424c9a3932274364fd34 : SUCCESS in 11m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dc9b0bc287d44c97b3259a9dca1d8db5 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f9092d2b7e994975b4177d8c11e7bf9f : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef034b3fcdf74fee963e705e7105aea4 : SUCCESS in 22m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e6961b4eb69544348ebbda918c2111ef : SUCCESS in 5m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9b6debabefd24259bc858734f0dd2842 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84e7700899ae4e0d8d4770e4e42b8e27 : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2efcfa85f2394ffda6cc207cc9dcfea3 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c13327e205a040b2b02745a0b44f3569 : SUCCESS in 1h 08m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e5e32274d864b6c833bfe85bf650399 : SUCCESS in 1h 57m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c33837759c635ce930fd3b4362730f4a80638310","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:45:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8be33ea449375fd9aad3e63af9bbb40d88f24696","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:46:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce86ae440f934b909c2b8f6ead7803e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b421db5838a344719a00ca7bc1696fa6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e45b1a9e315f81f1f7cb843d3607e97ade08bbd","revisions":{"8e45b1a9e315f81f1f7cb843d3607e97ade08bbd":{"kind":"REWORK","_number":1,"created":"2023-01-23 10:47:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/871463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/871463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/871463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/871463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/871463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/871463/1"}}},"commit":{"parents":[{"commit":"3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b","subject":"update constraint for oslo.db to new release 12.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3248010fdffd2c29cd4a4ce0ee948dfec7a4c75b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-23 10:47:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-23 10:47:37.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 2.8.0","message":"update constraint for mistral-lib to new release 2.8.0\n\nmeta: version: 2.8.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie323726a13173d1347fa0056381caf67364a8936\nmeta: release:Code-Review+1: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I276a5f0c769db71e7258a0abfe56f2ee0d145025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e45b1a9e315f81f1f7cb843d3607e97ade08bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e45b1a9e315f81f1f7cb843d3607e97ade08bbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
