)]}'
{"id":"openstack%2Frequirements~872282","triplet_id":"openstack%2Frequirements~master~Ic432da2938401185c6dc0f302139c39950cefcaf","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-02-18 15:56:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic432da2938401185c6dc0f302139c39950cefcaf","subject":"update constraint for openstacksdk to new release 1.0.0","status":"ABANDONED","created":"2023-01-31 14:42:30.000000000","updated":"2023-02-18 15:56:45.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4368f2da71d9e70e1f216856ffad39fbf84bc110","_number":872282,"virtual_id_number":872282,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-17 04:26:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-02-17 02:41:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2023-02-17 02:41:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-02-17 02:41:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 14:47:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-31 15:37:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-01-31 16:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-01 10:48:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-01 15:37:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-08 14:16:34.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-02-15 10:59:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c600c1111fafdba8b268da2ebdea0af9d0c6b048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-31 14:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84115b7591499738c3d3444aab152183f0e1735c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 14:47:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8f41ced0f74f4eaadd754c8201f30f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1726aacc6f34464d8e832bfab617564b : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eed5c50519f52dfc01cdff5fb15e59277883ae57","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-01-31 15:37:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d6f18d445dd03e746f41eb73e728867a3d92ede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 16:42:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8d84a1cfa814547b010bd4cf2383950\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0503340cda2a45079674764be55f3cd3 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b989535dfab4118a3eaaddd1edfd234 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a37d8093e36d431893fccc51a1bd69e9 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83f30800bd754bc18be8aacb6943e51e : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e35adb4c0fb4864b1cd60525252a52e : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee738f6305594a21ad6dbd1be36f4e49 : SUCCESS in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0e7f2352f8194cb9994131dc99a44dc6 : SUCCESS in 12m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4418fb83aff4491091a9e4e3b08838a0 : SUCCESS in 13m 05s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ef7a51eabda64e82a1e2ac3e60357253 : SUCCESS in 18m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/dc430083729742aebe123c3677e0408d : SUCCESS in 5m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/aa32db9fa71e4a93988886ad0c11de64 : SUCCESS in 5m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b4561159ce4b476880780366ed451040 : SUCCESS in 16m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b5c5ba5eb5a54c5e99d349959a44f63b : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/be5b96dd583d48fea655443c5f5f15e2 : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b79043d9b57a433cb71006d42e853f19 : SUCCESS in 9m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a2a7370a3c49426e9de8434e3a4bfb37 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/113a537bf0654bbc8c2b7ce1c7dcc5a1 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d78d011405b24a21aac05aac5e04771b : SUCCESS in 13m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/81f6573cc56a46ad92ae5ec201c863c9 : SUCCESS in 10m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3c1a089161a34169b15da8e5c99df9a8 : FAILURE in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/04cc6e2cd6f64969831186f784fce8b1 : SUCCESS in 9m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/75e004e178f14882bfded5ae9d04c34b : SUCCESS in 5m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/320827ea50fe40ea921e9dde83ad8157 : SUCCESS in 10m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bafe926132f24ae4920412eaac0c41a9 : SUCCESS in 23m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/93a087c26b414a6ca1af9a280d809ec9 : SUCCESS in 25m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/56ed99a60ca348e09c06a85cbae3524f : SUCCESS in 6m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f0583b5e42846629bed5698e00e61c0 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/509c6f1ec0bf41bbb07c735baaa1d07c : SUCCESS in 33m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/968db8888b7045049ad7f7dddbf6f6fe : SUCCESS in 8m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6dc526b3b3ea49e7a99d95a1f009c4ec : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13573ca0e0ae4e64849de53f408a2e98 : FAILURE in 6m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/784f6ecdd0484ad7a8ac067ad6480c8c : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85d35930a7654ec986ada278759ee01c : SUCCESS in 1h 03m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1466f01838364848af872ae84d653b88 : SUCCESS in 1h 53m 31s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"b3707251a42e31806886b8aaa18ac72a9acf04ac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-01 10:48:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0add0cb28b38cb93e679fd5977b9b2f8ac22af71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 10:55:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ef97a9a9bc642a49ad2c08a18ce5e0a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3ffae261b022480298671808fa23f17c : FAILURE in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d003243d56064030cc315c909f0b9b89f72b1e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:38:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c2ff6dac7284ce69e730ea51ef619ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d002f5784fb480090311c330e0dafd9 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002f9d5ca2774b1fa19ef6460d391ef2 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d5c9154d19740838a55263531810657 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25833888900f46bd968e2a4f607980c4 : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2851dc0ec41e42f7ade1d4219a7cbf05 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/135c2bdb5e184a9db914eac965fad4d0 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8189b5cd549b420c88d6a085badcbfd1 : SUCCESS in 8m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c16ac79f71d41b38e2aee0fcd2a2bf2 : SUCCESS in 14m 01s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7c15c23492f94b658a2cdcef92de2afc : FAILURE in 4m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/62788d1de38d4c96973bb57895f22f20 : SUCCESS in 6m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1a25081606624aa1a9fbcc142d193e44 : SUCCESS in 5m 31s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1050a50ac14e484faa9321e29f727c3b : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0ec506ccf28048728b63841349485188 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5abb86c8e6644ea78dd56f2f22bd62f7 : SUCCESS in 5m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ccb6495e28b04c0cb6f573aa46241e2e : SUCCESS in 7m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e05f20fe9f6e4997b5a8a49f13f18e3e : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbaee9a0d4d44ee9aacce9b2688c95af : SUCCESS in 6m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/07e24e17940340308ddb6fda5513b58f : SUCCESS in 7m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/99f59a55e59e44eea1a273b575b6f1eb : SUCCESS in 10m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/db4f7db4a54044b1b219a135e015e389 : FAILURE in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fc291e6da9a445288532cc5a770b54b0 : SUCCESS in 9m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c9914c09ec2544728ee7fb54452fdad1 : SUCCESS in 4m 34s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/539731d3f0134a5f8fa7dae8bbf0a1f2 : SUCCESS in 10m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a95d7913033643c6ae26f084ee94afce : SUCCESS in 26m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e73bb127f1f24199b4a8b206bdd7b3a5 : SUCCESS in 11m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4ece8e9f45684b699be141efacf51926 : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f874f52a89240ea9b464865b282afef : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce6c1736320c4d248377c144b7c3da97 : SUCCESS in 24m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3e91395a792249998eccbd4221aadaa9 : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/137d406104d9467b8225be8bd4958d16 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38fa530be0054a0b9aaeb13e01462e87 : FAILURE in 4m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bef257b0ee224745b161acf10ab8c764 : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30bd12e3501c43fa82f8f6a78101f73d : SUCCESS in 54m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68c7a6b4ee14402f9c7c3ffb008918b5 : SUCCESS in 1h 38m 59s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"b013cd668c82e63786fe5850094778ce2c85cf5c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-01 15:37:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10803039db6303493a02a13f22c884d1cb59514c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:44:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a97a609c0a6541c58ae0699d0d70c7c3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/31321e76319e460f8694db9c3e77140b : FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"613b6dd224c0e9ec30b6ec9774be51497cc05331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 18:09:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cb7dfb07bd240978939474dca4d7cfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a993b44efe9a4297b1074c97aeb0d337 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c39325dd41e42adb19ce19ce2d34f20 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22bf3ddc3e604692b690bf0fa9493f33 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dd71c4bad8c4fff847bbe999dfe6daf : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/320d5c6d9f7740f9a51851cd12e4f5c4 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d49f9ea9f004916a6472b2e0f66320c : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7151b1bd2e82407eb44e4ae12163852d : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6f144f605ca44d4bb5deb64ab36de200 : SUCCESS in 12m 44s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e016ad03adb648fa90c0d4c15a0ce476 : SUCCESS in 9m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/53815f3a1e3c4fd9b223f8e7a434258f : SUCCESS in 6m 28s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4aed41aa9fc84824b155ed0ccd3d3ed8 : SUCCESS in 7m 57s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aaf5894a4a224d6cbe08e660bdf7a379 : SUCCESS in 16m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2559a4d931a44ae6a8d228998808d289 : SUCCESS in 4m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/58a5a7a361414f4a929e770e8f71a123 : SUCCESS in 5m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f751091b8d784661b610bd42fcb53a5d : SUCCESS in 7m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8d14a2fb3d1f43b8afd7fb9287b6f1bb : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9738fcc627ae4c66a3d85d6bb9cfbb0b : SUCCESS in 8m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cd7ef1acac404327af2d87c63bead86f : SUCCESS in 11m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c8577cf4a6af4da68a3560682136ce48 : SUCCESS in 10m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/19ff6c01c71a4e3c95e8157f242508c5 : FAILURE in 5m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3306276e92104530afa6b1af34c2c219 : SUCCESS in 9m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5662b135110b4fef893bd153bb1b5517 : SUCCESS in 6m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/514c140d5bc04dc482d36caf991c7252 : SUCCESS in 10m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fa92372ad065426495f855677c31d260 : SUCCESS in 25m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e83b954f05db47baa5083e7bb9deed74 : SUCCESS in 11m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9b1b888bca694cd6a3091db0993c8577 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75c9a2114fba45c5a2c4df6793af16f4 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f42f82cdb814b22baf903e10ec1bca4 : SUCCESS in 22m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a4ab39d4500d4210883d7ad9ab0813e3 : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2a935047723443e08d63d7fd1b550320 : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0749c6dc2cf14709a3fc16a0abb9bf88 : FAILURE in 4m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9875c73edf894ff59025c5571ade1902 : SUCCESS in 17m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fac1f3ab092c42f58fbbeaa0e4781554 : SUCCESS in 58m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41d76360dd1c43cbaced76d9d0f5e3c5 : SUCCESS in 1h 32m 58s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"2c3f9a4724468b3e0972c662def51dca6b96d0d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-02 02:31:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"47afeef99975b54c119179b7785498f0ccaebaf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 02:42:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b63491c7814230aef163110319d1be\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/50db3f336e0c437289a75f57aaddadfd : FAILURE in 8m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2784e6c61d8b0dc66a4cb9f568394f19cb827d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 04:41:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7c0f3c2f7264f6d993d1b71e0148431\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02267de5f09462abd1523d7ae7a9c02 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921d0331bd1543e4816b49a838267328 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c50beb2cc7464c10b782784f11c2ac8d : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d594f460fe954bfda59af1293e79ebb7 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/928c233ed16a4cd5b80ebe3d4723c70f : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26ab44af08f848c19e15186a4afb0ba9 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4fbda685977440fdb2415a3b1863c8e6 : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0db2b19fd8f64781a69a52acbb4b4c4f : SUCCESS in 12m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2bc5792e0f0845038218b4710c1f0d22 : SUCCESS in 7m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fa24a68b20bf4ac692a10d3827d40071 : SUCCESS in 5m 42s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/343c26a4260d4446b14e7a7b33d7ebaf : SUCCESS in 5m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6a0ce2aa3c45422cbf23688ae327fde4 : SUCCESS in 11m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6756625f6d7c44c4890f40e6feba5627 : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e501384765884a07ba47dc456fc38ad6 : SUCCESS in 5m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c579822b096c425e8d6fbc3058a5ce16 : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b63f38e5cdf544c69779f7de68e6db1b : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/597a00b08ed14efca36c6adef56dd7e5 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c71178ebd501412b88c27ce2a30aff8c : SUCCESS in 10m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b24330fdaaa148b9b0d01b73915c1fb8 : SUCCESS in 12m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/61f5885a489047699fe8903790e710de : FAILURE in 3m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/76cb3f68e5e6414b86e80f35ba5b9305 : SUCCESS in 9m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/af87074d39c54723be2d330d369b35bf : SUCCESS in 4m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bc9ecbdbe55848a8a4f5e26e9c04a003 : SUCCESS in 9m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5b4d3043ed2b49599459cf55f2deb886 : SUCCESS in 26m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/14de4cc4b00f44639b0641abf69f0a50 : SUCCESS in 14m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d5707b7ff7bb42f68e9984527c149b6a : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c0da5de88ce4f62bdcd3a836d5bd636 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/531702d70d10402cab472bc394716d37 : SUCCESS in 21m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/30daf23c0bd949f38075514e404e738c : SUCCESS in 6m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9b2ae7abeeba43d7a7861146fe016cb9 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aaa2d0a9d7e04e2da56efbb6165a3bac : FAILURE in 5m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ea2ad80d8b7d424d9f74d4fff64c8fe7 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27de56a9aeaf446a8dd466665c948662 : SUCCESS in 52m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfc7f962502c4567b44731fa8fb40aab : SUCCESS in 1h 43m 45s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"fc8ad1c2471ed2d2d70f8814eda8c0acf0bacd29","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-02 10:26:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20eef6d90146f10e2a60d5a30450a3a957b6571f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-08 14:16:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ac0c9f4855f4bd26b3192c36dacd5c506d95aba","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-09 10:24:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a696067d7985bee92e70b2b0deb3a39e6a6090b0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-02-15 10:59:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a8d0db79ac047ff03d3b02e5a6333dffbe0f714","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"69f62dc526948203f4930836c4fa33be8a63190f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:47:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a39c70deee749d09aa4fbd03b777c30\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d6c56b5534354b40a366c9d0a08dc597 : FAILURE in 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0ceabe9ab87db951a40a71e45d9114a1168e25cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 04:47:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78fa5a9e6bc943cca158e628d84030b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aac6740c4ee4f5db259fcdfa4f56290 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e3d54b0e914050b790a9b6dacf5831 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d5c13a82692410990e6026ba1933be1 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63825b39b52640f0a9095c436219c358 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e5668f0a9fa44209ca6018cd4948f36 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29f7b770bb21443c9b427fed9dcd6e67 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ddff28d53eb74d48a6e724564fa9bc9c : SUCCESS in 8m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d41fb63211d045bfbe9499c57e4e0250 : SUCCESS in 11m 29s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a59f99fab79648d3b303cf1b99677e90 : SUCCESS in 8m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b40474e163a24eb7a9fa6715eb88a30e : SUCCESS in 4m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/795ff2edcd654855b67f9cb4fe95f7eb : SUCCESS in 6m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e2f7688883be40a298fabd95ee1ed804 : SUCCESS in 10m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cfb68019f91b41b4a2c88c2e4d61e850 : FAILURE in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b1ff24aebf64453889eb69b60785191f : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e6ec00dfa7ab424c9e98469d58b82b3e : SUCCESS in 7m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/81782e9c670a474cb9c1329c586e5047 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a69a0b9394eb44df92feafd9cc263850 : SUCCESS in 7m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e2defe384a414991893593949d6313f0 : SUCCESS in 9m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ebc80237816e40e1920d854b388b6d49 : SUCCESS in 10m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d7e61684e233455d9818778dc47820ba : FAILURE in 3m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/30c7e28015894dcd991bbc8fbd6622b5 : SUCCESS in 9m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/332a20a27930475e9eccdbfe3984ca0d : SUCCESS in 4m 44s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/dccb7c134a9d46eaa3459ad6736e9205 : SUCCESS in 8m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6337310f664b49eeb67a1e9e02689709 : SUCCESS in 23m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a4b108ab0a1c4673a3aa7b7ebfe183fb : SUCCESS in 10m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e9d159f56be64054b6a645b1b6a8eae3 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60a20dc729b34e69bb167876242f87a5 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4cf1967b864e4f0ca89cd36f909a4d18 : SUCCESS in 20m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b80c992d1aab42209637bb0ab2f35c76 : SUCCESS in 8m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/876b7db5d0f047fca72c3b44bb6740e3 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8f10e0bc05642bbb3d32970a9c43ac2 : FAILURE in 5m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3aed95bd82934adeb3a345c66e015688 : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95615d63a4604591b7a2f518bcc91736 : SUCCESS in 58m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/daa9f2a72edf48148c1a579b5b801dd0 : SUCCESS in 1h 15m 48s\n\nWarning:\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/cli/plugin-commands/octavia.rst in the commit\u0027\n  Comments left for invalid file doc/source/cli/plugin-commands/octavia.rst","accounts_in_message":[],"_revision_number":1},{"id":"bc2dee533725ad70b0a7a4b671093738665795e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:41:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"876b9ca5cc34ee9c1f0b350d4efc94e9e5a847ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:53:10.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a138be48bf5e4b05ba7f6f29b1d96f6f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4e6ddb6481ec427a9afd52c9a893ae73 : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"94769c37821446a48c7979b9a02f2dbd4bbff2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:26:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20b3659760404e05a044907225935909\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e0359c60074469b09e22048d64595a : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d772deb12a74085ac41980a6f540c2e : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ff30e8f9444e278dba366e66ca2b81 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40b5d129994649e9bc91265a9707970e : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62893166fd21453aa002bb5010b868db : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6ea648ebb81476d94cd63b426019614 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af0321589c1e4092afc08774fa2c4dd5 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/922263509a69451b961df1ba97f96a2e : SUCCESS in 12m 21s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3e145d8e9e24427396037316a78fc79e : SUCCESS in 6m 40s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cab63c7ed2b34d84a79b125e1d8f39a5 : SUCCESS in 6m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b5608a3a98104843910282ca8626b828 : SUCCESS in 6m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bf956daa6c074aadbea77e69eb25cea7 : SUCCESS in 11m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ab540f71583048c3a3bcb5a8955b6a00 : SUCCESS in 3m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/438a277fdbe948c5a81ddf2f5597bd47 : SUCCESS in 6m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8e666feea0584947990249ea5096c39d : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/29d1d9a443ea4b9a96234b42135f9437 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48b90b7149cc436d87abe7665654054e : SUCCESS in 8m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d0ff654e2b7e4aa2a11d63aaa49ed335 : SUCCESS in 8m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0f56790ff9e444c88bacbbf700160a9c : SUCCESS in 11m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3e5d824ad9cb4581b5eae36f089c5ede : FAILURE in 3m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c57cdbaccb7f4c309ac46641e94f04ec : SUCCESS in 9m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/991cad18c39b4a72baa2ceadc20b4879 : SUCCESS in 4m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/293cee6498564d6abab52c80bb9fca7a : SUCCESS in 9m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b8bea1cc35fd4e25966e6d09f09b359f : SUCCESS in 25m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cd73cea1d9d74dbcb8d93443d09b8760 : SUCCESS in 11m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b4bc714a65e74468866685a7d94d9ea1 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47d9affb3e7b4b7ab43fff304c08b4b8 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18fd3d43e4cc4cc5b133a1aceacb79b8 : SUCCESS in 24m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/db3de248ead441f39df6f0387306c349 : SUCCESS in 4m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d0fdea4f2d1047a3b210a56c79ac1fac : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4750a3ec49a4a4fbeee8082f7581ee7 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b884419bc4044d7ca2ccd5e48e05d79c : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78d5e9a194084bb6a1eb2f6e8ae3adc0 : SUCCESS in 53m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a2bb605292c4e21a998ef5584d752a5 : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4368f2da71d9e70e1f216856ffad39fbf84bc110","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-18 15:56:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3b2618ffae783b7f7eaf576f38e91577159cd15f","revisions":{"a957f9aeb31832d48655110cb82d5cb0c52683cd":{"kind":"REWORK","_number":1,"created":"2023-01-31 14:42:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/872282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/872282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/872282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/872282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/872282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/872282/1"}}},"commit":{"parents":[{"commit":"3341a625879d5f2a91f9033dc52d4f5185bb5ab5","subject":"Merge \"Remove anyjson\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3341a625879d5f2a91f9033dc52d4f5185bb5ab5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-31 14:42:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-31 14:42:29.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 1.0.0","message":"update constraint for openstacksdk to new release 1.0.0\n\nmeta: version: 1.0.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: I7e978d9f48c8b3ee8788953a73c2fc7736d035be\nmeta: release:Code-Review+1: Shnaidman Sagi (Sergey) \u003csshnaidm@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nChange-Id: Ic432da2938401185c6dc0f302139c39950cefcaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a957f9aeb31832d48655110cb82d5cb0c52683cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a957f9aeb31832d48655110cb82d5cb0c52683cd"}]},"branch":"refs/heads/master"},"3b2618ffae783b7f7eaf576f38e91577159cd15f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 02:41:40.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/82/872282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/872282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/872282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/872282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/872282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/872282/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-31 14:42:29.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:41:25.000000000","tz":-360},"subject":"update constraint for openstacksdk to new release 1.0.0","message":"update constraint for openstacksdk to new release 1.0.0\n\nmeta: version: 1.0.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: I7e978d9f48c8b3ee8788953a73c2fc7736d035be\nmeta: release:Code-Review+1: Shnaidman Sagi (Sergey) \u003csshnaidm@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nChange-Id: Ic432da2938401185c6dc0f302139c39950cefcaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b2618ffae783b7f7eaf576f38e91577159cd15f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b2618ffae783b7f7eaf576f38e91577159cd15f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
