)]}'
{"id":"openstack%2Frequirements~872322","triplet_id":"openstack%2Frequirements~stable%2Fxena~I69f67e7c296e5f4bc86c7ad1a00a3bc5f73c6e12","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I69f67e7c296e5f4bc86c7ad1a00a3bc5f73c6e12","subject":"update constraint for python-openstackclient to new release 5.6.1","status":"ABANDONED","created":"2023-01-31 15:33:30.000000000","updated":"2023-02-01 15:37:38.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"99071943f2d66b4f62223cdb01ad4307e251e260","_number":872322,"virtual_id_number":872322,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-31 16:59:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"disliked":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2023-01-31 15:37:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-02-01 10:49:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":-1,"date":"2023-02-01 14:56:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 15:37:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-01-31 15:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-31 16:59:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-01 10:49:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-01 14:56:38.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"f6245067806f4ad6a5ac612948778652c6948d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-01-31 15:33:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99c8c5e259313a2d3fa9ec33221a2506e48da0a3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-01-31 15:37:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b02c8bfc752ab04dc9161d1c5ef0ae2e44ad2357","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:59:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7972eb2cc1d749a0bb2708a6364b4c73\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/297623564006479899d97abec33e059c : FAILURE in 23m 23s","accounts_in_message":[],"_revision_number":1},{"id":"349af5930174c8dbd6cabbf34565d3311655dc1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 16:59:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f7de5c5a1a346eb97a64978dce258ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/710977084f2540a59cdbfe88d73825c8 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ddc838fdc5d437c913f82ee24144d42 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4d37474f7a4670a9d149a3b0664bbb : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74532d59cf2648ebbb0bb54aea50e1ff : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c6ca533dd194f9c8ff2709e775b2306 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/151a3e60883b4d49b2a1b9ff803f19f3 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc2d788d3dd640fabf706f4e8e763f96 : SUCCESS in 11m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0bf216f622e42ac85ce3814439c1014 : SUCCESS in 10m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6449f3cdbcc0400497e4630894c78fb4 : SUCCESS in 4m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d37284b484bd4fc096ec45e484d0cb27 : SUCCESS in 8m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/57eb29b02ffb40ab89903a61188b45b2 : SUCCESS in 4m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/43d4735f11c043fca05e6048998e0dcd : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bff5f7effbcb41b9af21b5bb3c84c6d2 : SUCCESS in 6m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cde6fb22bd024ec5aeb3cf809f36a312 : SUCCESS in 7m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9d2f587cccb6460ab09f48f60aa74bc0 : SUCCESS in 15m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c08413e25a3f4cb9b45d4f3015399106 : SUCCESS in 7m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4928e387ef074f75afe60484cd4f0a7c : SUCCESS in 4m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/848bad17897c4930b34527b7a5128985 : SUCCESS in 26m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2fec02da0c9d44cc9fe75d6f2e94186d : SUCCESS in 10m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cda3be65004a4057bf65921820614311 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3650bf9a2f724382ab251ab737a18dd9 : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/934983aa7f104c12a2abcd4b1935c83e : SUCCESS in 17m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/817e2c71a64b45fd8dc6d0a4983646c1 : SUCCESS in 5m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0a54bb00e12b43e6aa618bdacb84358c : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d132eeb356794ed59ba1ec0bc73d098e : SUCCESS in 6m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9019f0764d2645579b97ff08a1dd1ca8 : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb8e39a98d59427cbefe9e5e2ebd28c9 : POST_FAILURE in 17m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/619d124b8c9245e49e3199f313817f7c : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9b1b2db84926339bd7d50351b8af9951c7026df0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-01 10:49:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f32da7b1236f32ad5a11171cf4e3417589ea7655","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 10:59:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1917d6b7e7ba41ba8fe0b6a318ed9637\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1cacf14ac2f8401aaf7f4e26a65fdf19 : FAILURE in 8m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1f4d8a9f7f1eba425ea23a52056b669c43ae71ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:22:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf02134707b440eb874c657ef09097c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce887b3a9e14ee7bfa7f53f6f437980 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99fe0b23ead462486cb58f5639e56e1 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84823d488bc24797a6fe9bddf5469ccb : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65767c6e40cb4808b4d53ed394508810 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08f088138ad841e2b2a045a8006c3fd3 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bf868c6b4054ae9a09290c1484141ad : SUCCESS in 2m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8232be9192fe4e41bc0cbaf569c48022 : SUCCESS in 14m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8389c1cdbe544868b55477290b88166f : SUCCESS in 10m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a875bea6bcb54d5c9b82a9f9a9cf27d5 : SUCCESS in 4m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a575e49237f146e8bc1554ab27f60dbf : SUCCESS in 9m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/74b759e4ff324731af81f39d820ccd2e : SUCCESS in 4m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/61aac03cdfba426e961fc2499735e1f9 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49e9d1a5073c4f9f8b73e1b5a0af0b30 : SUCCESS in 6m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e533349905474230b2f736d99c47ccfd : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/509d26a183a445c68ca9c626fa1e235f : SUCCESS in 17m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/abf3550f93fd4c26bc5ac6adec3ec404 : SUCCESS in 8m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7be257081d004e65b582440e8b9d5d6f : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c3023273436b422db7f48b42cbd54815 : SUCCESS in 32m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cbd2aa22e3914093af41422d765ef31f : SUCCESS in 10m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/57be27e781c945df868f694864c68935 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0325bafdf8754ad8b621ba7c9c5046a9 : SUCCESS in 2m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f439096e0bf94a9782b4e8aff71d3a62 : SUCCESS in 18m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5fd9b0319e014a53bfb9b25426bbf4cd : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4165c92990d44baeab21dce342a4db89 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a57212c7955422fa1383ad3efe23851 : SUCCESS in 6m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3a706daa9fe04ccd9e2b041bd879aa4b : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf33a1dbd48b42daa9f6463941a8d795 : POST_FAILURE in 27m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d43b93adff994fbf920c8f1e976f6775 : SUCCESS in 1h 29m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1656eb29f5c5b219ceb7659fa3f0813e3e3a026a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-02-01 14:56:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99071943f2d66b4f62223cdb01ad4307e251e260","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-01 15:37:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96e962312c40fc20709d6acec9a2bde712bad5e9","revisions":{"96e962312c40fc20709d6acec9a2bde712bad5e9":{"kind":"REWORK","_number":1,"created":"2023-01-31 15:33:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/872322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/872322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/872322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/872322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/872322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/872322/1"}}},"commit":{"parents":[{"commit":"cad8e89bc30b6e21f46ddfc9a7a800a7c6cee3a1","subject":"Merge \"update constraint for ovsdbapp to new release 1.12.3\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cad8e89bc30b6e21f46ddfc9a7a800a7c6cee3a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-31 15:33:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-01-31 15:33:26.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 5.6.1","message":"update constraint for python-openstackclient to new release 5.6.1\n\nmeta: version: 5.6.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib2ded7cf51494efe25a63a7c652e5efe13022f41\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I69f67e7c296e5f4bc86c7ad1a00a3bc5f73c6e12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96e962312c40fc20709d6acec9a2bde712bad5e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96e962312c40fc20709d6acec9a2bde712bad5e9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
