)]}'
{"id":"openstack%2Frequirements~872399","triplet_id":"openstack%2Frequirements~master~Ia8fc80ca6e827ea4b1fa56152ddea2eab0fd6389","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-02-01 12:37:06.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Ia8fc80ca6e827ea4b1fa56152ddea2eab0fd6389","subject":"update constraint for oslo.messaging to new release 14.2.0","status":"MERGED","created":"2023-02-01 11:08:19.000000000","updated":"2023-02-02 00:42:13.000000000","submitted":"2023-02-02 00:41:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872399-new-release","meta_rev_id":"125c6243f0bca5bdcc325e000d80d549ae62fa08","_number":872399,"virtual_id_number":872399,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-02 00:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-01 15:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-02-01 14:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-01 15:37:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 11:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 12:36:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-02-01 12:55:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-01 14:13:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-01 15:37:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6f2f4acdee6c7fca8e285d68d836c7323704945e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-01 11:08:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baddf320359e2fa3a80fbbe31a18f41557e328d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:14:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55ecc4553b1742f0ba1ec84e47902653\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1da018478a6c46a78aa25f6ed095a228 : FAILURE in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d73e698ddbf5ab8ad43332dcd492b80e869e37e1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-01 12:37:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbcb01c1c6c5cde8ee555aa1217c44b75ed03a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:55:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4201272ea6a94924a429dfbe4ad24e22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5941a7ce77614a569436f03064f124f0 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a290022cdc3e4afea0425c53326dbd96 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83a5e8785e514f5a9101686c2e9cf4b0 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/153c6a1c8146425a8cff041a5ad3a161 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0081fbb371343f198cf3c8443d150d2 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6100e1f708ad4ecfb404dee3bfe2180c : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ac5ed6d6a94149a5b334c43626a30aef : SUCCESS in 13m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dfb32b31e85b4f98893ddc67913b5e94 : SUCCESS in 17m 50s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/107a203a10cc4ad9a700c760f92c4607 : SUCCESS in 7m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2f1f05de6bce4b729ebecfffd1146a35 : SUCCESS in 5m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/baaceaa527074418a9f35402178d1b1f : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/68922a89f92c40989616d87697f768a3 : SUCCESS in 9m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4b29e538574649a8add039ac40270091 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/58b9204b812c48a2bda475508649ff3a : SUCCESS in 4m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e6b8b880951e436aa682b870be2ff581 : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ac8a18b825ae40e3af456c24f2a1a9e4 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a884676766cf40a3b2d1a06d0901e4c7 : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/43d88e78574842cf9536050094cb2cf3 : SUCCESS in 6m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4d69ff56b41b470aac6a1254884a78cf : SUCCESS in 11m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d83a681c3d5c4a08b303539cb45a78a1 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/deb989753a744a96a2cc731705a3759c : SUCCESS in 7m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bac40ba8f89e4f0ebd47b8c47680f3ed : SUCCESS in 4m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bdf4e9ddeac84a7586b54c34fd9bc2f7 : SUCCESS in 7m 40s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8ab47e827dc245a8b53ca07d12dd0f7a : SUCCESS in 24m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ea4c2482395842b88a99c806c7087458 : SUCCESS in 11m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/60e4f57518874b19b3592b8afbc0404c : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a60e86d388b4227b5ac73fb17b374c9 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23051f820ef0476eabd206ae42c5f1a2 : SUCCESS in 22m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8e86c984f1fe43d387b881f13cbb0e6b : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0de8e6af27194f178283dd84b1b85190 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53dcdc813b24478aa251b540e2926eea : SUCCESS in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/66088d1547bd4315915fd8cb350207f8 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dad2885ac624d059178ca6c67fb30de : SUCCESS in 54m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4921c094307148489ad03d451ae1a748 : SUCCESS in 1h 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2ad72b8155d7cf5588adcaef1a600d4ca9a96de1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-01 14:13:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f944ec53341586b859983bd1d19017de68c10b6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-01 15:37:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cd2b3be91876e5ad8a437dc244b6981543f380f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:38:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95e6210f9aeb6efadd5607d052454e79a50828ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:41:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bea7dac1468148b882cd89e5ac4740ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314b96b3ddaf46faa4704944f4a4aaed : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0624d7baa1944f0695527087acd591dd : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43e128f1b6f44944927524b9d1257e40 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b157ae28e7374df3a06f9617902317a4 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7111b49dc7f4919a044b7d863bbc9ab : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13f892cdf76244d59e3d7ded60665ac8 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d9f6e1373a51481faf57fe353617b49c : SUCCESS in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d4f2b08611c243d8a23a00d89c0b137c : SUCCESS in 10m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/31d7ca486ca840d988d7aaab480d1081 : SUCCESS in 11m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/509fd972a1c34a99bad8d7946ae6910f : SUCCESS in 5m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b3e2af0b84124152ba06e6a900c25ce0 : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a362210ae9fd4385bb63fc52b747ff24 : SUCCESS in 7m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/34af9f6cabdd40d39b0657e58c65aa93 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c33fa15728d44102a418ad6098fa6457 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5cb8540732d44fd4a72f7c0c058f29ea : SUCCESS in 7m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/af826517a67440aca32df9c76ce06973 : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9745e1ce627f4057af5355e28c377fff : SUCCESS in 3m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9e8ed07d50d74c6986e46be35b53e087 : SUCCESS in 10m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f9b91d99c68548149eb3eb4ecf3f7a6d : SUCCESS in 4m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8c76dd1e6857403eb41dff93d4adcd67 : SUCCESS in 21m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e50e680f517846799df41b5556249e42 : SUCCESS in 12m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/93e437daab1c4013b6cbcacf40a180a9 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4fbba8fd564d45a2bc180598afacad1e : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0ceeef2ab36415e8644117fac280079 : SUCCESS in 21m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4c9f9cce2d9a45748a4ef35c5f17734b : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cc589c2c52a54b71b7cbe4c58647073d : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c073bb13fb74373be078958e8813113 : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ca7989f64d3640839e22ca83ca364673 : SUCCESS in 9m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3e0f9c2806d4d03a71d73edad9a76f7 : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07d4e27760c644939a8b70b320b7ab22 : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b690b32026267139e1fa436dcf0088ca6859a050","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:41:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"125c6243f0bca5bdcc325e000d80d549ae62fa08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:42:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4318de5407684b3baa82771571a6dc1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001ecc9d94f04c3f903a540a18fc3395 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","revisions":{"0b5f5d0273975dfa808e7f921dff14ed2887b9f6":{"kind":"REWORK","_number":1,"created":"2023-02-01 11:08:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/872399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/872399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/872399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/872399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/872399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/872399/1"}}},"commit":{"parents":[{"commit":"2fe10843ec35474dc5977fb00bce1011bcad1d7a","subject":"Merge \"Revert \"Temporary pin tox\u003c4 until projects are fixed for tox4\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fe10843ec35474dc5977fb00bce1011bcad1d7a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-01 11:08:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-01 11:08:18.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.2.0","message":"update constraint for oslo.messaging to new release 14.2.0\n\nmeta: version: 14.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nmeta: release:Commit: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nmeta: release:Change-Id: I28b7ee5db02b8b1b332feee97d5d22c195e05727\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia8fc80ca6e827ea4b1fa56152ddea2eab0fd6389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
