)]}'
{"id":"openstack%2Frequirements~873343","triplet_id":"openstack%2Frequirements~master~Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79","project":"openstack/requirements","branch":"master","topic":"upgrade-pyzmq","attention_set":{},"removed_from_attention_set":{"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2023-05-21 08:21:39.000000000","reason":"Change was abandoned"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-03-30 11:28:24.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"25277":{"account":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"last_update":"2023-03-31 09:55:13.000000000","reason":"\u003cGERRIT_ACCOUNT_25277\u003e replied on the change","reason_account":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"}}},"hashtags":[],"change_id":"Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79","subject":"Downgrade pyzmq to 19.0.0","status":"ABANDONED","created":"2023-02-09 23:24:11.000000000","updated":"2023-05-21 08:21:39.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8cded71a697bed369e0e902e4dde3b02accc284","_number":873343,"virtual_id_number":873343,"owner":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-30 22:03:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-09 23:29:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 01:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 20:10:00.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2023-03-30 09:47:20.000000000","updated_by":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"reviewer":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"state":"CC"},{"updated":"2023-03-30 14:06:07.000000000","updated_by":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"reviewer":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"state":"REVIEWER"},{"updated":"2023-03-30 14:07:42.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-02 11:48:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"44974fca88ac53c22fd76413bb127e6e28e9ebc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-02-09 23:24:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8acc6f4c09f98bc025016a87d95cd25b08860374","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 23:29:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2d0a7a903f47cca89eac39100e96bc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7e53fb3539ed4a2ba47e3e5bccbb52db : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"91e7361169ccfc58b9f9cdfac439f371a13ad3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 01:30:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae7388fae0164016ad5ff5f0c38ed751\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe28478473d4a55a9c69c5fb2264fa1 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e581cefcef61485e9594133a016c913b : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f385d47f8eb4553b884fad0b3f0aa62 : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6c503e235404188aac872f5fe70ee7e : FAILURE in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/482be66946f144ffb089406483d0b40e : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/612545e1c17442a5bab70d6aa9de6f56 : SUCCESS in 4m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/386c86375df84f3db838c153ee89034d : FAILURE in 5m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9ea64e0b754c45658c39dd25c1ce8bef : FAILURE in 5m 44s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/64720bffe7c0451784e259d85448ac8a : SUCCESS in 7m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/340ba30ee4d64ee8a46292ebb2819aba : SUCCESS in 5m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7205b9e488ed40409e65bc1c5cd7e5d1 : SUCCESS in 7m 31s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/18a79a30fc324d16860d95f7a1b5c829 : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5982dba951c740a6b0c506c233551cf2 : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d293568dd55f45ba9eae096460ac1bcb : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a1491f88cd384f8da41408c35772209c : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fd773b5600074adfacc5c11e9c57d46b : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61ddad39823049a19ee2e8c8e10841d0 : SUCCESS in 7m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0267a2b676754b57ad2df71cf5a313b9 : SUCCESS in 10m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8021c427e5144ed09260780dd2517f0b : SUCCESS in 10m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/54f7114e4194425da77ebac3dd86abca : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9667650f3b624511bb6068ecd6cd45f5 : SUCCESS in 8m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/acb5ff1000944de98427466f736ef7be : SUCCESS in 4m 48s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/334daabda3014481aabea7449fb111a2 : SUCCESS in 8m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a353f32ae0b141c49e4aaf497c8ede8e : SUCCESS in 26m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c33b4313230147b3b36e259707288431 : SUCCESS in 11m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/feeebc65103240f6aa16547d1319cfaf : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01f866228f124269a072016d8d88bcc3 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37d31df27a684a5e812d11d26e7631f4 : SUCCESS in 23m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/dbab06aa54e540de80f3c31d43815955 : SUCCESS in 4m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f44ad00f372a4d6cb5cc5bddfbec3536 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96034ca484324c709f38151b003f0859 : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/21d3efe441bf42f0b02833727f022060 : SUCCESS in 8m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2a02426331e4abc90ce181be8095010 : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/476a5d285aa74d78bbb248dfe72bf2d8 : TIMED_OUT in 2h 04m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9f5e885dad37fcd8a51c93b70e3f3860212db78d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-02-10 10:51:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30f5f71c7b423ae199539ab9a51660e87a008e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 11:05:35.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20758a00392b4bca93faae5a711f8f89\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a212a1f7c1ff4da08d63b02fb75e952b : FAILURE in 12m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0d12b053524f2559fbf861d4d9dceaa800e1369a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 13:01:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bedff69b4c3f41318180dba19d500515\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665870f7fed44c039d271b95173d35a5 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f042fd17814092b1fe168d585df8f5 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afdc7257420944859081572b627ac6d8 : FAILURE in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44f0cee03e754ad0aedc6da9ed3b8f71 : FAILURE in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54862693c05c4e3196c166422b004e8b : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20856549c1a94b53a48fddeae625227a : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdc61eac41ef43cc9c60fa242831fc8c : FAILURE in 5m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7fdcc46e4e57483dab1a22056c79f5b5 : FAILURE in 5m 49s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/56bf4b35ab42494d983cd2a952fd5796 : SUCCESS in 6m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5bdcd751bd884755af3ee76abce1f58e : SUCCESS in 6m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/20a1509577fb482d84ebd3c5dc28170e : SUCCESS in 6m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cc7184537c55422f892e249f73cbe768 : SUCCESS in 10m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f321b7e84ed848558fe80205c923cae8 : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f3e186e6643a47f3a1c373f177e5c191 : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/81b24d8475b64a74a5727bf7fda371d9 : SUCCESS in 7m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1ab58f1ac1de4d39967b27981cdfeda1 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f588ee2ac9d04e3686eac132431b6a8d : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/062f3d0b84644ef1b327da419fef1459 : SUCCESS in 9m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fe43084ddbc84632a2c1acb8eeb38650 : SUCCESS in 11m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/00191982b81a4d0996cfa352ef57e9f0 : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/97cf30f0ccad43b7ac6443dace8e1fb8 : SUCCESS in 9m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c9f4b4b07f324d94984880999f873352 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/13ca6731c12b4f2686a355109a87b05e : SUCCESS in 9m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/05e71ee541444703a3bc69d88869cf64 : SUCCESS in 26m 35s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/036206d28de6456fa344633c997a3fd6 : SUCCESS in 13m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4fac75b6d520427cbb0d32c2e018fce3 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b57170c77d043ddba1db9caae922c3e : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29084268f4ed47eb90053e468285bbc3 : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7477aa59183e40189e6499446a1590a0 : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/21363b9d4d1a45909e54791c40a8d833 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c984d497693481ebb19005ef114f83e : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bfa9de4466dc43eda5d7acc152e8254a : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/212734baac7c45ec9d57c73dfef12d50 : SUCCESS in 1h 03m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0f2bb6d038b48e18da6e914130fc062 : TIMED_OUT in 2h 04m 35s","accounts_in_message":[],"_revision_number":2},{"id":"fab5012c0519494469f30ed822f938dd01f2d9c8","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-02-15 13:07:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f680da2b3ad8a156c10fff9937b88c42e23d354","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 13:12:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6083672271a448a4b3f09958cffa6946\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/75646f6fdcf149dcb60a919a70881c7e : FAILURE in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bfd72595151723bcfb934f0f0b1390dddeacb1f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:48:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffc8c9c090744877a940d471b553b72e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3149642a02ba4d708815ed22461eabc6 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ac887cfab6409d98535fbe8e0e4c65 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/842dfaab155c4140aa037870cb8f3aec : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96fa72d89a834cdd8ed9ca3a3ddde81b : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0edb8e9d00944ecbaf08add6658be899 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1db685f915c4e34af4c7fd206929e8e : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82b314f550324e16af14d03967c40b1e : FAILURE in 4m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/18b049ee5b9d42b1b5693adc81f33024 : FAILURE in 5m 43s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/01392eaede7d4f0496bd65e99893201b : SUCCESS in 8m 03s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a57fa18f178a498b96c97460e9c6cf2b : SUCCESS in 6m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/aea6b9f074104cf5bf588183cc4c560d : SUCCESS in 6m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3506193f9f1f4577992eedf450245144 : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/099732e1473a429e9045a9a85e738fe1 : SUCCESS in 3m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/37856e39c77048159a5d930fd097d3f6 : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c2b302a49fae4f63afa506343ccf3032 : SUCCESS in 7m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3828be5bfc1a46dda61add741de76a05 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b5e9cb76c104c8a98caf01b32e9eb16 : SUCCESS in 6m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/25f075d95a554d75b59330269273fd39 : SUCCESS in 9m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/48299a08aa0843dd95cc408386d1d157 : SUCCESS in 13m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dc20951981c14de391385f82644ae7b6 : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/47ea23444fa441619b920322850c3b3b : SUCCESS in 10m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/93ead2dfbe404501b5f8d9091ae8df8b : SUCCESS in 5m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/68796d38448c4c12bc220eb3d2101441 : SUCCESS in 10m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0e7ce6a14d1046d691cf9167064a9b38 : SUCCESS in 25m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1f1def5690564ae68ee92c9839ad724b : SUCCESS in 13m 14s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1acd2d08dd1643839ef594f65c63ed36 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd9379efb5124d2c971dcf5d49381b76 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e73f1b110fc4681b44030400bfaa32b : SUCCESS in 24m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4ec8f32f5d234b55ba28faf559de87e5 : SUCCESS in 4m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6a3f4efd8fc241e1b68d102ac20050c6 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/47a9c61cdbfe42208a23b2e198fb8770 : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3bc8093b67cc49bf9b6175fd777ed12b : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef46ad2d072548ea9c0e4b841c1a4966 : SUCCESS in 56m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e71457a035f9416384ab82dcace3ae34 : SUCCESS in 1h 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cbc074bdfce5874034530a06bb56ceb37d16a237","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 20:10:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73ec48ca2e5f3c982bccbfcadddd7e886a622380","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-03-20 21:14:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cdc38f881959a89ed07e8f5ac5435d637f83c61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-03-20 21:16:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b228a4e5998413ce21195d55ebb9d6da64d73db1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 21:21:30.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b545015d623e46debc9c0a5c94a83eff\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/44889065a423487281580e3ce589eb94 : FAILURE in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b6ccdc54e8fb6cc523a07fe1abe337a394fb658f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 22:44:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed83917850d04c16b1bb198e3acf657e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01b9a052e3c4574b355be119f7f12a2 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29248a88414d4f4b91a1a39aba92ba41 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc4f6084eab4cb68eee885ed3c29b40 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0c9a39e8c52432a976a1d525186e67b : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36bc79507d2a455b9eb070a8e7af5576 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2afb7848a8bc4dc79e7853dc0cf82806 : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0309eadef3014f269a8cebd97b1c6652 : FAILURE in 7m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fee64d87799f44b8b2b05fea629f1ddf : FAILURE in 7m 41s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/71b9a52f24684254b56c7ea66860cf6d : SUCCESS in 7m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/16fb6eefb12a490581a9d9418627a983 : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6ad315e70ff34b04b4ca73dc379d2972 : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/87c7d9cb38a243c5ba554134fefb6b5b : SUCCESS in 11m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5886f031ceb74d0f8be852b5906ab185 : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/15bfb98b89704acb843e89aea9417821 : SUCCESS in 5m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1514d6db120e469aa1e8cbfb3c7b0a36 : SUCCESS in 7m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/78a1f9e851fb4e7287bdab6f93d4258e : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc502e43e101454ba38982ad686a9e9c : SUCCESS in 7m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/321fc618007d49c4b23cd9b623bd7ab0 : SUCCESS in 10m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/311a29c4819d4717b2994efa46a03a39 : SUCCESS in 15m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a36ff5bbcd6e40a9b24a12e834dae4a9 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6591bca255314c2ca24b2e96cfdb2656 : SUCCESS in 9m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d6c5a30861ea48d3be00bf1c26969bcc : SUCCESS in 4m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3e1b4d7e2bc941ec9f2548ca1b733ced : SUCCESS in 11m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0382e11640df4a37b0b2504280980c28 : SUCCESS in 26m 03s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/78bc02136056494ab45070ed7a68b55d : SUCCESS in 13m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/adbe5cb648f241188010de6563967ee0 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/00312b624c7f405da1eb8a92a91b19bf : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07dc0004a26741178f5e02d49138b8c0 : SUCCESS in 29m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/291217511aa8420188b0c839a8a64ce7 : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8fe4e9c7c5394c4fbe4e6d47ea98712a : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc177602fbb343bd92f6ad8d072db8b3 : SUCCESS in 6m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9e8c691b8671490fbaba3aa949fca824 : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8631028e28044c489e2b1d420178262 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f47fb3ef823047ebb949bc1b056d2330 : SUCCESS in 1h 22m 21s","accounts_in_message":[],"_revision_number":3},{"id":"a98f4900bc839ae8b6fa500d3a346bb9b358fb86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-03-21 10:02:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a21baeb42040b7d340360436d3e29919862543e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 10:09:09.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e456a92b1c7487aba60e3b26952bae3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e1ee26996fd0437989b0a154781c7dc5 : FAILURE in 5m 17s","accounts_in_message":[],"_revision_number":4},{"id":"55c816fc53ba3048772e59cdf4f133cdcb66ff69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 11:24:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba41244ce7bf42e1ae4bb53613b57ab3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6c6db4a7424527b8da5795217f015c : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9f65af7910480fb6212c7b75794798 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/979efc71c59043e6bdc27ef643e573f4 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/892fc8b18b3b41e494d36d9c2ccd9108 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7107420d151f4d61855a457f93503757 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cd7f388cf8745bf926397c95617f7ce : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/885af0bb3a3f45b5ae2c92028b4b8801 : FAILURE in 6m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1b146f3fb4f9420a89e6525ce1aa4e7a : FAILURE in 7m 50s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a436022d0e2b43129ecfeed8d4dc7e95 : SUCCESS in 12m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8ff5d9bef92b4ec3a55eb5c522074fd1 : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d5a2b154cbc045b688b5f3d61581f4ff : SUCCESS in 5m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/710e35fa360c472ea236e14d5b18fc6c : SUCCESS in 11m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ede7c93691ef48999974843e5ad3bad9 : SUCCESS in 4m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/49b12b489a84480cbec29a2f0a6acd24 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ea5fd4a9f9484091b5e5dad3e386fbcd : SUCCESS in 7m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b1a9fe184d4e492384b2837232de0af6 : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2052585c24f941b8abac2f7e3fcde45e : SUCCESS in 8m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ae947e070793484f8c78d684e4bb8d58 : SUCCESS in 8m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e6c38250cfcd4e2aa28fa8d755df0784 : SUCCESS in 19m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/678642fc58c0452fa138609f66245e0a : SUCCESS in 3m 20s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/20b6ac364a67479e93ade3cb19655375 : SUCCESS in 9m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/40613237b9724b898f62323c6de7dd97 : SUCCESS in 5m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/660236c2bd1f40c4839f16825da0914a : SUCCESS in 9m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/174aff3405814b9f8e6b15434097ed06 : SUCCESS in 24m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/656a5e8a59a14d2c80400182d56fa8a9 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8e07f081734a4d87bb45a68a53fc1476 : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78a3d341c58d41c8bcd688000e426b93 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d79c927145b843dca0b6117c5e977e72 : SUCCESS in 22m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/00d4ba558dee475b9914a51e4316cddc : SUCCESS in 6m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/458c85816fdb49ab91d2bbf2cb0e9d96 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ead9780a9144b9cb7b2aa144b0ac8ca : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eafe23cff26a4ad0aebc66f09775bacf : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/261c54e45f49473dad21891097f41245 : SUCCESS in 58m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63ec3df637e5469793d5d21a43463b16 : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":4},{"id":"ef5ed4dcd6314a7b582d0ec986d266448b651c8e","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2023-03-30 09:47:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"42c502645a45e78701263212a77474750d558639","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-03-30 11:14:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea2cb8e4c7d51c295111197c41457f6a5573a735","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-30 11:28:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3458ea5d3cbb9162b0f8e0127977714ae6756722","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2023-03-30 14:06:07.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1871764bc9eaa57f816f2a881a2acc75c3df8c69","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2023-03-30 14:07:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"23fb43a9713896bd5c058a322e601dd82987dc01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 14:07:39.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 873343,4","accounts_in_message":[],"_revision_number":4},{"id":"4cb05e33f51edd0a5ab1fbb8ea1602b592cce704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 14:07:39.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 873343,4","accounts_in_message":[],"_revision_number":4},{"id":"81324da859cf4af05d365ccd9c65129e12133956","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-30 14:07:42.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 873343,4","accounts_in_message":[],"_revision_number":4},{"id":"f599561f5afbb6f75bbe2cf219ea918796ce2751","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2023-03-30 14:09:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"22460cc9980455c47fe03280271b2c5b9883ebe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-03-30 20:31:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2d75dc8bd7b4077ecf59579b857ba5804106006c","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-03-30 20:31:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"295da601e0effbf90074ede40bdc71910af62c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-03-30 20:32:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"124a75dfb5c894bd7e230b4147c896645ff4e483","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 20:55:26.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/356cd5802f944336815fc60df1e305fb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/27de48a13db4495588fa4c1215b01f99 : FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":6},{"id":"14f8ac21c1251486e959dec5c0530c45d424ebc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 22:03:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5e435a2f7994d68aba26d8a0911508c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1931ebde32fe4082bfae61c2b395273e : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f48df1a01640efaa7f8a9a80bf91d6 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dec791dbd924af49f2ac1451dca872f : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b21855381ebc4e87a9507f914100319a : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5b96902b8204981a38bf7ca9bbfe45f : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dba385f4131d4675b9fbc8b43a245c08 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4365b9e103c24784a378155b4a7f97c3 : FAILURE in 6m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4d7cf7aefda44650ac0e171abebafab9 : FAILURE in 6m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2d0583cb417b4ca3b422d0ee2e0fbc18 : FAILURE in 7m 32s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5b2e7504abf14d2cb9250afb656de214 : SUCCESS in 7m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/132f8bcb7930489692c708fb611bb704 : SUCCESS in 4m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/82c6252ed35a4593a4cd2c9c2d3feabe : SUCCESS in 5m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6dfe389f95374723b7332970595b6768 : SUCCESS in 11m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6f1c943c50ea45beaa0b3eea8a39db55 : SUCCESS in 4m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a815ead78c8040d78038c0d9ea6b98a1 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/643b970445c04c9b954631e0bcca4c01 : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a898c0a1afde4816bd8ba2d7d5e23847 : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87acf1775ffe4370950a446c120424ee : SUCCESS in 6m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8eae5a7936bb414e9790b756d8d37855 : SUCCESS in 9m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8a4f993535114c889fe91ca01f9d09be : SUCCESS in 11m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b07cf46e6d084245aab6e9f9f152bba3 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0ee4f88566594854bc0670247a8fd487 : SUCCESS in 10m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3654ae56400f4cb88948127ec747dfe8 : SUCCESS in 4m 51s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1a6e02c82d334dea84af3714d82cd8c3 : SUCCESS in 10m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e612b1a6e8044dc3a83dd48a6ac0886b : SUCCESS in 25m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/16b1871848814e69a2305465c32199a7 : SUCCESS in 16m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9569b11fda4d49ea8ad94f0f675e5e37 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/66f16c6f438e4b14b337630107bbb878 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93bff5939bb54866b55ed46f86d0db52 : SUCCESS in 21m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/895e2972d4134fb1a326d5102bb848e5 : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8f9fef1e80534f31a9a063d1e9ba6d78 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a72ebdec9c7146f990a26ee9262ab3b2 : SUCCESS in 6m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4177431ca2d2431e851bd78b10718f80 : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f21a477982c047f7a119ae1bd444b50b : SUCCESS in 45m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57f1c0c404944d499c314c7303f4fc8d : SUCCESS in 1h 22m 47s","accounts_in_message":[],"_revision_number":6},{"id":"129866cbe11b6a9b75c403b911903f3c219126ec","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2023-03-31 09:55:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f1d5abacd05ec5185fdb49db91ea49b057473757","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-02 11:48:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d8cded71a697bed369e0e902e4dde3b02accc284","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-21 08:21:39.000000000","message":"Abandoned\n\nthis should be obsolete with the latest salt update","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2473053c350d5a9034a6ed21d7ce91391b6937fc","revisions":{"9c31f1a4deb86afd7bd291ebb5bafa350678206e":{"kind":"REWORK","_number":1,"created":"2023-02-09 23:24:11.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/43/873343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/873343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/873343/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-02-09 23:23:08.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-02-09 23:23:15.000000000","tz":-180},"subject":"Upgrade pyzmq to 25.0.0","message":"Upgrade pyzmq to 25.0.0\n\n``pyzmq 20.0.0`` has an issue in its packages, which makes it\nimpossible to install via ``pip``.\n\nTo reproduce the problem one can do the following (via install\nfrom sources):\n```\nwget https://files.pythonhosted.org/packages/15/81/009a2cd08bc5de6ed092c94508b45d6dbc6f6d0bbeddbe14d36b51f28216/pyzmq-20.0.0.tar.gz\n\ntar -xf pyzmq-20.0.0.tar.gz\n\ncd pyzmq-20.0.0.tar.gz\n\npip install .\n\n```\n\nThe error that will happen is the following:\n```\nProcessing /pyzmq-20.0.0\n  Preparing metadata (setup.py) ... error\n  error: subprocess-exited-with-error\n\n  × python setup.py egg_info did not run successfully.\n  │ exit code: 1\n  ╰─\u003e [43 lines of output]\n\n              It is possible a package already installed in your system\n              contains an version that is invalid according to PEP 440.\n              You can try `pip install --use-pep517` as a workaround for this problem,\n              or rely on a new virtual environment.\n\n              If the problem refers to a package that is not installed yet,\n              please contact that package\u0027s maintainers or distributors.\n\n      Traceback (most recent call last):\n        File \"\u003cstring\u003e\", line 2, in \u003cmodule\u003e\n        File \"\u003cpip-setuptools-caller\u003e\", line 34, in \u003cmodule\u003e\n        File \"/pyzmq-20.0.0/setup.py\", line 1378, in \u003cmodule\u003e\n          setup(**setup_args)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 107, in setup\n          _install_setup_requires(attrs)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 80, in _install_setup_requires\n          _fetch_build_eggs(dist)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 85, in _fetch_build_eggs\n          dist.fetch_build_eggs(dist.setup_requires)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/dist.py\", line 879, in fetch_build_eggs\n          resolved_dists \u003d pkg_resources.working_set.resolve(\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 812, in resolve\n          if not req_extras.markers_pass(req, extras):\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1016, in markers_pass\n          return not req.marker or any(extra_evals)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1013, in \u003cgenexpr\u003e\n          req.marker.evaluate({\u0027extra\u0027: extra})\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 245, in evaluate\n          return _evaluate_markers(self._markers, current_environment)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 151, in _evaluate_markers\n          groups[-1].append(_eval_op(lhs_value, op, rhs_value))\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 109, in _eval_op\n          return spec.contains(lhs, prereleases\u003dTrue)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 565, in contains\n          normalized_item \u003d _coerce_version(item)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 36, in _coerce_version\n          version \u003d Version(version)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/version.py\", line 197, in __init__\n          raise InvalidVersion(f\"Invalid version: \u0027{version}\u0027\")\n      pkg_resources.extern.packaging.version.InvalidVersion: Invalid version: \u0027cpython\u0027\n      [end of output]\n\n  note: This error originates from a subprocess, and is likely not a problem with pip.\nerror: metadata-generation-failed\n\n× Encountered error while generating package metadata.\n╰─\u003e See above for output.\n\nnote: This is an issue with the package mentioned above, not pip.\nhint: See above for details.\n\n```\n\nThe same error happens when installing via ``pip`` and downloading from the\nrepository with ``pip install pyzmq\u003d\u003d\u003d20.0.0``.\n\nThe error does not happen with either pyzmq\u003c20 or pyzmq\u003e20. This error\nbreaks Kolla build of containers, such as Neutron.\n\nIdeally we need to propagate this change to stable branches as well.\n\nChange-Id: Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c31f1a4deb86afd7bd291ebb5bafa350678206e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c31f1a4deb86afd7bd291ebb5bafa350678206e"}]},"branch":"refs/heads/master"},"4b50e836ffc2b07fae001da30e17524547f19d42":{"kind":"REWORK","_number":2,"created":"2023-02-10 10:51:12.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/43/873343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/873343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/873343/2"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-02-09 23:23:08.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-02-10 10:50:58.000000000","tz":-180},"subject":"Downgrade pyzmq to 19.0.0","message":"Downgrade pyzmq to 19.0.0\n\n``pyzmq 20.0.0`` has an issue in its packages, which makes it\nimpossible to install via ``pip``.\n\nTo reproduce the problem one can do the following (via install\nfrom sources):\n```\nwget https://files.pythonhosted.org/packages/15/81/009a2cd08bc5de6ed092c94508b45d6dbc6f6d0bbeddbe14d36b51f28216/pyzmq-20.0.0.tar.gz\n\ntar -xf pyzmq-20.0.0.tar.gz\n\ncd pyzmq-20.0.0.tar.gz\n\npip install .\n\n```\n\nThe error that will happen is the following:\n```\nProcessing /pyzmq-20.0.0\n  Preparing metadata (setup.py) ... error\n  error: subprocess-exited-with-error\n\n  × python setup.py egg_info did not run successfully.\n  │ exit code: 1\n  ╰─\u003e [43 lines of output]\n\n              It is possible a package already installed in your system\n              contains an version that is invalid according to PEP 440.\n              You can try `pip install --use-pep517` as a workaround for this problem,\n              or rely on a new virtual environment.\n\n              If the problem refers to a package that is not installed yet,\n              please contact that package\u0027s maintainers or distributors.\n\n      Traceback (most recent call last):\n        File \"\u003cstring\u003e\", line 2, in \u003cmodule\u003e\n        File \"\u003cpip-setuptools-caller\u003e\", line 34, in \u003cmodule\u003e\n        File \"/pyzmq-20.0.0/setup.py\", line 1378, in \u003cmodule\u003e\n          setup(**setup_args)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 107, in setup\n          _install_setup_requires(attrs)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 80, in _install_setup_requires\n          _fetch_build_eggs(dist)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 85, in _fetch_build_eggs\n          dist.fetch_build_eggs(dist.setup_requires)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/dist.py\", line 879, in fetch_build_eggs\n          resolved_dists \u003d pkg_resources.working_set.resolve(\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 812, in resolve\n          if not req_extras.markers_pass(req, extras):\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1016, in markers_pass\n          return not req.marker or any(extra_evals)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1013, in \u003cgenexpr\u003e\n          req.marker.evaluate({\u0027extra\u0027: extra})\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 245, in evaluate\n          return _evaluate_markers(self._markers, current_environment)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 151, in _evaluate_markers\n          groups[-1].append(_eval_op(lhs_value, op, rhs_value))\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 109, in _eval_op\n          return spec.contains(lhs, prereleases\u003dTrue)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 565, in contains\n          normalized_item \u003d _coerce_version(item)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 36, in _coerce_version\n          version \u003d Version(version)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/version.py\", line 197, in __init__\n          raise InvalidVersion(f\"Invalid version: \u0027{version}\u0027\")\n      pkg_resources.extern.packaging.version.InvalidVersion: Invalid version: \u0027cpython\u0027\n      [end of output]\n\n  note: This error originates from a subprocess, and is likely not a problem with pip.\nerror: metadata-generation-failed\n\n× Encountered error while generating package metadata.\n╰─\u003e See above for output.\n\nnote: This is an issue with the package mentioned above, not pip.\nhint: See above for details.\n\n```\n\nThe same error happens when installing via ``pip`` and downloading from the\nrepository with ``pip install pyzmq\u003d\u003d\u003d20.0.0``.\n\nThe error does not happen with either pyzmq\u003c20 or pyzmq\u003e20. This error\nbreaks Kolla build of containers, such as Neutron.\n\nOne of the libraries requires ``pyzmq\u003c\u003d20.0.0``; therefore, I am proposing the\nversion 19.0.0.\n\nIdeally we need to propagate this change to stable branches as well.\n\nChange-Id: Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b50e836ffc2b07fae001da30e17524547f19d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b50e836ffc2b07fae001da30e17524547f19d42"}]},"branch":"refs/heads/master"},"49c847e18099b5ae91e43e8a33c2511ba4df23eb":{"kind":"REWORK","_number":3,"created":"2023-03-20 21:16:13.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/43/873343/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/873343/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/873343/3"}}},"commit":{"parents":[{"commit":"e7d3312a31340af93dc2f35d03f622272007f69d","subject":"Merge \"update constraint for python-novaclient to new release 18.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7d3312a31340af93dc2f35d03f622272007f69d"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-02-09 23:23:08.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-03-20 21:15:57.000000000","tz":-180},"subject":"Downgrade pyzmq to 19.0.0","message":"Downgrade pyzmq to 19.0.0\n\n``pyzmq 20.0.0`` has an issue in its packages, which makes it\nimpossible to install via ``pip``.\n\nTo reproduce the problem one can do the following (via install\nfrom sources):\n```\nwget https://files.pythonhosted.org/packages/15/81/009a2cd08bc5de6ed092c94508b45d6dbc6f6d0bbeddbe14d36b51f28216/pyzmq-20.0.0.tar.gz\n\ntar -xf pyzmq-20.0.0.tar.gz\n\ncd pyzmq-20.0.0.tar.gz\n\npip install .\n\n```\n\nThe error that will happen is the following:\n```\nProcessing /pyzmq-20.0.0\n  Preparing metadata (setup.py) ... error\n  error: subprocess-exited-with-error\n\n  × python setup.py egg_info did not run successfully.\n  │ exit code: 1\n  ╰─\u003e [43 lines of output]\n\n              It is possible a package already installed in your system\n              contains an version that is invalid according to PEP 440.\n              You can try `pip install --use-pep517` as a workaround for this problem,\n              or rely on a new virtual environment.\n\n              If the problem refers to a package that is not installed yet,\n              please contact that package\u0027s maintainers or distributors.\n\n      Traceback (most recent call last):\n        File \"\u003cstring\u003e\", line 2, in \u003cmodule\u003e\n        File \"\u003cpip-setuptools-caller\u003e\", line 34, in \u003cmodule\u003e\n        File \"/pyzmq-20.0.0/setup.py\", line 1378, in \u003cmodule\u003e\n          setup(**setup_args)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 107, in setup\n          _install_setup_requires(attrs)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 80, in _install_setup_requires\n          _fetch_build_eggs(dist)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 85, in _fetch_build_eggs\n          dist.fetch_build_eggs(dist.setup_requires)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/dist.py\", line 879, in fetch_build_eggs\n          resolved_dists \u003d pkg_resources.working_set.resolve(\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 812, in resolve\n          if not req_extras.markers_pass(req, extras):\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1016, in markers_pass\n          return not req.marker or any(extra_evals)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1013, in \u003cgenexpr\u003e\n          req.marker.evaluate({\u0027extra\u0027: extra})\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 245, in evaluate\n          return _evaluate_markers(self._markers, current_environment)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 151, in _evaluate_markers\n          groups[-1].append(_eval_op(lhs_value, op, rhs_value))\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 109, in _eval_op\n          return spec.contains(lhs, prereleases\u003dTrue)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 565, in contains\n          normalized_item \u003d _coerce_version(item)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 36, in _coerce_version\n          version \u003d Version(version)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/version.py\", line 197, in __init__\n          raise InvalidVersion(f\"Invalid version: \u0027{version}\u0027\")\n      pkg_resources.extern.packaging.version.InvalidVersion: Invalid version: \u0027cpython\u0027\n      [end of output]\n\n  note: This error originates from a subprocess, and is likely not a problem with pip.\nerror: metadata-generation-failed\n\n× Encountered error while generating package metadata.\n╰─\u003e See above for output.\n\nnote: This is an issue with the package mentioned above, not pip.\nhint: See above for details.\n\n```\n\nThe same error happens when installing via ``pip`` and downloading from the\nrepository with ``pip install pyzmq\u003d\u003d\u003d20.0.0``.\n\nThe error does not happen with either pyzmq\u003c20 or pyzmq\u003e20. This error\nbreaks Kolla build of containers, such as Neutron.\n\nOne of the libraries requires ``pyzmq\u003c\u003d20.0.0``; therefore, I am proposing the\nversion 19.0.0.\n\nIdeally we need to propagate this change to stable branches as well.\n\nChange-Id: Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49c847e18099b5ae91e43e8a33c2511ba4df23eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49c847e18099b5ae91e43e8a33c2511ba4df23eb"}]},"branch":"refs/heads/master"},"078c2e528e66aada59436cc53ed1b5781c761927":{"kind":"REWORK","_number":4,"created":"2023-03-21 10:02:38.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/43/873343/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/873343/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/873343/4"}}},"commit":{"parents":[{"commit":"e7d3312a31340af93dc2f35d03f622272007f69d","subject":"Merge \"update constraint for python-novaclient to new release 18.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7d3312a31340af93dc2f35d03f622272007f69d"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-02-09 23:23:08.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-03-21 10:02:25.000000000","tz":-180},"subject":"Downgrade pyzmq to 19.0.0","message":"Downgrade pyzmq to 19.0.0\n\n``pyzmq 20.0.0`` has an issue in its packages, which makes it\nimpossible to install via ``pip``.\n\nTo reproduce the problem one can do the following (via install\nfrom sources):\n```\nwget https://files.pythonhosted.org/packages/15/81/009a2cd08bc5de6ed092c94508b45d6dbc6f6d0bbeddbe14d36b51f28216/pyzmq-20.0.0.tar.gz\n\ntar -xf pyzmq-20.0.0.tar.gz\n\ncd pyzmq-20.0.0.tar.gz\n\npip install .\n\n```\n\nThe error that will happen is the following:\n```\nProcessing /pyzmq-20.0.0\n  Preparing metadata (setup.py) ... error\n  error: subprocess-exited-with-error\n\n  × python setup.py egg_info did not run successfully.\n  │ exit code: 1\n  ╰─\u003e [43 lines of output]\n\n              It is possible a package already installed in your system\n              contains an version that is invalid according to PEP 440.\n              You can try `pip install --use-pep517` as a workaround for this problem,\n              or rely on a new virtual environment.\n\n              If the problem refers to a package that is not installed yet,\n              please contact that package\u0027s maintainers or distributors.\n\n      Traceback (most recent call last):\n        File \"\u003cstring\u003e\", line 2, in \u003cmodule\u003e\n        File \"\u003cpip-setuptools-caller\u003e\", line 34, in \u003cmodule\u003e\n        File \"/pyzmq-20.0.0/setup.py\", line 1378, in \u003cmodule\u003e\n          setup(**setup_args)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 107, in setup\n          _install_setup_requires(attrs)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 80, in _install_setup_requires\n          _fetch_build_eggs(dist)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 85, in _fetch_build_eggs\n          dist.fetch_build_eggs(dist.setup_requires)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/dist.py\", line 879, in fetch_build_eggs\n          resolved_dists \u003d pkg_resources.working_set.resolve(\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 812, in resolve\n          if not req_extras.markers_pass(req, extras):\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1016, in markers_pass\n          return not req.marker or any(extra_evals)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1013, in \u003cgenexpr\u003e\n          req.marker.evaluate({\u0027extra\u0027: extra})\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 245, in evaluate\n          return _evaluate_markers(self._markers, current_environment)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 151, in _evaluate_markers\n          groups[-1].append(_eval_op(lhs_value, op, rhs_value))\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 109, in _eval_op\n          return spec.contains(lhs, prereleases\u003dTrue)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 565, in contains\n          normalized_item \u003d _coerce_version(item)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 36, in _coerce_version\n          version \u003d Version(version)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/version.py\", line 197, in __init__\n          raise InvalidVersion(f\"Invalid version: \u0027{version}\u0027\")\n      pkg_resources.extern.packaging.version.InvalidVersion: Invalid version: \u0027cpython\u0027\n      [end of output]\n\n  note: This error originates from a subprocess, and is likely not a problem with pip.\nerror: metadata-generation-failed\n\n× Encountered error while generating package metadata.\n╰─\u003e See above for output.\n\nnote: This is an issue with the package mentioned above, not pip.\nhint: See above for details.\n\n```\n\nThe same error happens when installing via ``pip`` and downloading from the\nrepository with ``pip install pyzmq\u003d\u003d\u003d20.0.0``.\n\nThe error does not happen with either pyzmq\u003c20 or pyzmq\u003e20. This error\nbreaks Kolla build of containers, such as Neutron.\n\nOne of the libraries requires ``pyzmq\u003c\u003d20.0.0``; therefore, I am proposing the\nversion 19.0.0.\n\nIdeally we need to propagate this change to stable branches as well.\n\nChange-Id: Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/078c2e528e66aada59436cc53ed1b5781c761927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/078c2e528e66aada59436cc53ed1b5781c761927"}]},"branch":"refs/heads/master"},"b0c237319fd274a8efce179fa5a528f182809c0a":{"kind":"REWORK","_number":5,"created":"2023-03-30 20:31:44.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/43/873343/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/873343/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/873343/5"}}},"commit":{"parents":[{"commit":"be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8","subject":"Merge \"Add requirements-tox-py310-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-03-30 20:31:01.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-03-30 20:31:01.000000000","tz":-180},"subject":"Downgrade pyzmq to 19.0.0","message":"Downgrade pyzmq to 19.0.0\n\n``pyzmq 20.0.0`` has an issue in its packages, which makes it\nimpossible to install via ``pip``.\n\nTo reproduce the problem one can do the following (via install\nfrom sources):\n```\nwget https://files.pythonhosted.org/packages/15/81/009a2cd08bc5de6ed092c94508b45d6dbc6f6d0bbeddbe14d36b51f28216/pyzmq-20.0.0.tar.gz\n\ntar -xf pyzmq-20.0.0.tar.gz\n\ncd pyzmq-20.0.0.tar.gz\n\npip install .\n\n```\n\nThe error that will happen is the following:\n```\nProcessing /pyzmq-20.0.0\n  Preparing metadata (setup.py) ... error\n  error: subprocess-exited-with-error\n\n  × python setup.py egg_info did not run successfully.\n  │ exit code: 1\n  ╰─\u003e [43 lines of output]\n\n              It is possible a package already installed in your system\n              contains an version that is invalid according to PEP 440.\n              You can try `pip install --use-pep517` as a workaround for this problem,\n              or rely on a new virtual environment.\n\n              If the problem refers to a package that is not installed yet,\n              please contact that package\u0027s maintainers or distributors.\n\n      Traceback (most recent call last):\n        File \"\u003cstring\u003e\", line 2, in \u003cmodule\u003e\n        File \"\u003cpip-setuptools-caller\u003e\", line 34, in \u003cmodule\u003e\n        File \"/pyzmq-20.0.0/setup.py\", line 1378, in \u003cmodule\u003e\n          setup(**setup_args)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 107, in setup\n          _install_setup_requires(attrs)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 80, in _install_setup_requires\n          _fetch_build_eggs(dist)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 85, in _fetch_build_eggs\n          dist.fetch_build_eggs(dist.setup_requires)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/dist.py\", line 879, in fetch_build_eggs\n          resolved_dists \u003d pkg_resources.working_set.resolve(\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 812, in resolve\n          if not req_extras.markers_pass(req, extras):\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1016, in markers_pass\n          return not req.marker or any(extra_evals)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1013, in \u003cgenexpr\u003e\n          req.marker.evaluate({\u0027extra\u0027: extra})\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 245, in evaluate\n          return _evaluate_markers(self._markers, current_environment)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 151, in _evaluate_markers\n          groups[-1].append(_eval_op(lhs_value, op, rhs_value))\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 109, in _eval_op\n          return spec.contains(lhs, prereleases\u003dTrue)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 565, in contains\n          normalized_item \u003d _coerce_version(item)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 36, in _coerce_version\n          version \u003d Version(version)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/version.py\", line 197, in __init__\n          raise InvalidVersion(f\"Invalid version: \u0027{version}\u0027\")\n      pkg_resources.extern.packaging.version.InvalidVersion: Invalid version: \u0027cpython\u0027\n      [end of output]\n\n  note: This error originates from a subprocess, and is likely not a problem with pip.\nerror: metadata-generation-failed\n\n× Encountered error while generating package metadata.\n╰─\u003e See above for output.\n\nnote: This is an issue with the package mentioned above, not pip.\nhint: See above for details.\n\n```\n\nThe same error happens when installing via ``pip`` and downloading from the\nrepository with ``pip install pyzmq\u003d\u003d\u003d20.0.0``.\n\nThe error does not happen with either pyzmq\u003c20 or pyzmq\u003e20. This error\nbreaks Kolla build of containers, such as Neutron.\n\nOne of the libraries requires ``pyzmq\u003c\u003d20.0.0``; therefore, I am proposing the\nversion 19.0.0.\n\nIdeally we need to propagate this change to stable branches as well.\n\nChange-Id: Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0c237319fd274a8efce179fa5a528f182809c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0c237319fd274a8efce179fa5a528f182809c0a"}]},"branch":"refs/heads/master"},"2473053c350d5a9034a6ed21d7ce91391b6937fc":{"kind":"REWORK","_number":6,"created":"2023-03-30 20:32:44.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/43/873343/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/873343/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/873343/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/873343/6"}}},"commit":{"parents":[{"commit":"be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8","subject":"Merge \"Add requirements-tox-py310-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4f55d91c67a94223b05dd9d9bc1891c1c4d5a8"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-03-30 20:31:01.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-03-30 20:32:32.000000000","tz":-180},"subject":"Downgrade pyzmq to 19.0.0","message":"Downgrade pyzmq to 19.0.0\n\n``pyzmq 20.0.0`` has an issue in its packages, which makes it\nimpossible to install via ``pip``.\n\nTo reproduce the problem one can do the following (via install\nfrom sources):\n```\nwget https://files.pythonhosted.org/packages/15/81/009a2cd08bc5de6ed092c94508b45d6dbc6f6d0bbeddbe14d36b51f28216/pyzmq-20.0.0.tar.gz\n\ntar -xf pyzmq-20.0.0.tar.gz\n\ncd pyzmq-20.0.0.tar.gz\n\npip install .\n\n```\n\nThe error that will happen is the following:\n```\nProcessing /pyzmq-20.0.0\n  Preparing metadata (setup.py) ... error\n  error: subprocess-exited-with-error\n\n  × python setup.py egg_info did not run successfully.\n  │ exit code: 1\n  ╰─\u003e [43 lines of output]\n\n              It is possible a package already installed in your system\n              contains an version that is invalid according to PEP 440.\n              You can try `pip install --use-pep517` as a workaround for this problem,\n              or rely on a new virtual environment.\n\n              If the problem refers to a package that is not installed yet,\n              please contact that package\u0027s maintainers or distributors.\n\n      Traceback (most recent call last):\n        File \"\u003cstring\u003e\", line 2, in \u003cmodule\u003e\n        File \"\u003cpip-setuptools-caller\u003e\", line 34, in \u003cmodule\u003e\n        File \"/pyzmq-20.0.0/setup.py\", line 1378, in \u003cmodule\u003e\n          setup(**setup_args)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 107, in setup\n          _install_setup_requires(attrs)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 80, in _install_setup_requires\n          _fetch_build_eggs(dist)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/__init__.py\", line 85, in _fetch_build_eggs\n          dist.fetch_build_eggs(dist.setup_requires)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/setuptools/dist.py\", line 879, in fetch_build_eggs\n          resolved_dists \u003d pkg_resources.working_set.resolve(\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 812, in resolve\n          if not req_extras.markers_pass(req, extras):\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1016, in markers_pass\n          return not req.marker or any(extra_evals)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/__init__.py\", line 1013, in \u003cgenexpr\u003e\n          req.marker.evaluate({\u0027extra\u0027: extra})\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 245, in evaluate\n          return _evaluate_markers(self._markers, current_environment)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 151, in _evaluate_markers\n          groups[-1].append(_eval_op(lhs_value, op, rhs_value))\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/markers.py\", line 109, in _eval_op\n          return spec.contains(lhs, prereleases\u003dTrue)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 565, in contains\n          normalized_item \u003d _coerce_version(item)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/specifiers.py\", line 36, in _coerce_version\n          version \u003d Version(version)\n        File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pkg_resources/_vendor/packaging/version.py\", line 197, in __init__\n          raise InvalidVersion(f\"Invalid version: \u0027{version}\u0027\")\n      pkg_resources.extern.packaging.version.InvalidVersion: Invalid version: \u0027cpython\u0027\n      [end of output]\n\n  note: This error originates from a subprocess, and is likely not a problem with pip.\nerror: metadata-generation-failed\n\n× Encountered error while generating package metadata.\n╰─\u003e See above for output.\n\nnote: This is an issue with the package mentioned above, not pip.\nhint: See above for details.\n\n```\n\nThe same error happens when installing via ``pip`` and downloading from the\nrepository with ``pip install pyzmq\u003d\u003d\u003d20.0.0``.\n\nThe error does not happen with either pyzmq\u003c20 or pyzmq\u003e20. This error\nbreaks Kolla build of containers, such as Neutron.\n\nOne of the libraries requires ``pyzmq\u003c\u003d20.0.0``; therefore, I am proposing the\nversion 19.0.0.\n\nIdeally we need to propagate this change to stable branches as well.\n\nChange-Id: Icb5eaec46dd187e5b9ccdfc2dbdf9f9ca7ef8c79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2473053c350d5a9034a6ed21d7ce91391b6937fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2473053c350d5a9034a6ed21d7ce91391b6937fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
