)]}'
{"id":"openstack%2Frequirements~873384","triplet_id":"openstack%2Frequirements~master~I582a316720eb480b87148e4c9acfb25808404c85","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I582a316720eb480b87148e4c9acfb25808404c85","subject":"update constraint for glance_store to new release 4.3.0","status":"MERGED","created":"2023-02-10 16:07:52.000000000","updated":"2023-02-16 16:08:39.000000000","submitted":"2023-02-16 16:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873384-new-release","meta_rev_id":"7d280ba651255e9a4cfc1114157c6194f3b24e90","_number":873384,"virtual_id_number":873384,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 16:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 14:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 14:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:12:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-16 14:29:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3378225669fbffb29ea085e873deb9d5ce8dd412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:07:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca0d04a5690e2975357d8fa7cd26189f7e00e116","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:12:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bd03285b3684e83b25eff36e6062074\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/884f37f344d14293a69d7476ace2d494 : FAILURE in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3e31f34ab938334710f73ec6abb381977ed0d372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:12:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c0cca7515a34433886d67b26202d759\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc357e91bc6846909599eac55f9119ce : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091e4abc2b524e3ba9dbde74cddb5b9b : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24d4d7ecf6734c0eb725f5e076d17dba : FAILURE in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2d87eb628cb47378744f494883f1bc8 : FAILURE in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25fa0e18abce4362819f749dd78ea1c2 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8100cb06fef647639d7a591cec62081c : SUCCESS in 4m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/534a438b720c4abfa9fa7a5f7261a181 : FAILURE in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bf79bf37cbb543d987ae5a860002ef4b : FAILURE in 12m 18s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/dec69dfdf089490a8bd114acbcc1fb58 : SUCCESS in 7m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/70468766903e423da11461ef7e35b44a : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/59c9240357914105be0618d54756540d : SUCCESS in 6m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ce67a0ebbe9e4b408b709a513a3f1b64 : SUCCESS in 9m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a83db20e68304a9ba7e6722907068ffa : SUCCESS in 6m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/75be8a37a20f41ab8a70743c726a9fb5 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ef02f798258647dfb2afdf7fb8a824d9 : SUCCESS in 8m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f79ceedb4d55401b85a0a097545bcfd5 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ccf2c72a7c174d83a5bc3eada804a947 : SUCCESS in 7m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/36df72feac9c443b947b0b50882bc604 : SUCCESS in 9m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/30c538bb5d1749cbb99a0b34ff338109 : SUCCESS in 8m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c13385244db8434980f9fbe168ad8c21 : SUCCESS in 4m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/034b623ff3a74bc98a81e4170ae72071 : SUCCESS in 12m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/14aa7027364346649028fde5c4dfd3fa : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/02d12768ec2549cf92829a4c320cbfa2 : SUCCESS in 9m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/093fd027288948e880e8225a18ff3f2c : SUCCESS in 35m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9d24fa327e5d4296a3045308d5bcd6ad : SUCCESS in 13m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5008471e812f41a696ceb512c849ac71 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93ff945c2ec44801bca200431661df39 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3bc160b3cb5742df88ee484ede2a1301 : FAILURE in 23m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/23b7071e8a704d87b3138e2be6bbea00 : SUCCESS in 7m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3e328888ee2642b2b82d59e76077ec1b : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5620b617a5f4879aff51fcc586d3eb0 : SUCCESS in 6m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/663a280ccd38410cab1e20d08733ea9d : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/991ff378e0a54c41aa08dad4641d8d31 : SUCCESS in 56m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87881ad462b64c80851f3bacf7a72bbb : SUCCESS in 1h 55m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7e9e20a360ee7c1697965ec296a12c12afa2dfff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26afd008647dfe3018d8ed963f133e0fc57a053e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:00:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7bd8af535764f019f23dd9dcd3b5f04\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fec6fb52f6d54554a27850e7eceae458 : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3eed46f832b3487e93b9237cf13086e00fc87e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:40:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c13c19057e754a4eae5d8a8c053d1cd6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa3607d0bf04412b9a6a57bed7888e8 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3144743d58d4d0d99838490eb44eebb : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e143e25bf3e45fe88dc86f23c1caa41 : FAILURE in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71cb5d55f13b4e898f8e8c2600a33414 : FAILURE in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f81cf239685c4b45b1ea4035e9825177 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c398e4bd0f9b44d7aeeab2e11f869a31 : SUCCESS in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f42b639e6a12487abd6dc5428ad4bcdd : FAILURE in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb53b111dcfe4babb586abf334458a94 : FAILURE in 13m 30s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/17f3615bb92446baa07d5a0e68d1920c : SUCCESS in 7m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d77cb35c7ae8479c8e34a4236468723e : SUCCESS in 5m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/13012097d1304e6d85103f80ee30c84e : SUCCESS in 5m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5ab3ddc675fe4f6aaf7bae7d58219f59 : SUCCESS in 10m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d12e3372f8804eaf906de39285fb3498 : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4013ac0cb7974d3f8d55d5249cb21316 : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/335bec1bd4914f93a82fe2dab20415b3 : SUCCESS in 8m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a3b1b4f484074feb9bfe72f5b188ed10 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/184c3a4a16ba408f95722c0b1496d867 : SUCCESS in 30m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ca06e7a5016749c8955e56599f2cb480 : SUCCESS in 8m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0fa0aa4df01b4e7e84db5b5f117194b2 : SUCCESS in 9m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1a45d670ae2f4a64bf72ae4b52c33039 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4b5028da64944f28a66c01508a2183e5 : SUCCESS in 10m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/285a069785e6499b8741c15cab8c0b6d : SUCCESS in 4m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f9d3feefbfcc46658c8b471a6520e123 : SUCCESS in 10m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8c04837e8d9248d28120c79a980c5193 : SUCCESS in 25m 13s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/978c25ce1ca44cb8919ed34f3ba84e69 : SUCCESS in 12m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/69a55275b2224a39bc79b870fba5e570 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cec815306c564b7eae930b08bd6331d7 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df250a81c2c145f181f951bf876f7c58 : SUCCESS in 24m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d5414eebbe784239ad20834305438842 : SUCCESS in 5m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/820102d19b9548d0a5e7d98052b83b8f : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0660fb2694d49bba5837d2533a3803a : SUCCESS in 6m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/74962469eb48411592f69655a5f4f8ff : SUCCESS in 15m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69d72ace1d384f92909877c7c373c0a7 : SUCCESS in 55m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5edf92330222435e84a4298651ff248e : TIMED_OUT in 2h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0b74a5a7280a7697cef114dc51bfb014898f6658","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c9211c5589ba85d186440cc1c391cdd8f185124","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:06:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f080316bae8d49f292c49494b579863b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2418e79920f948e08ac6d5fa87b7a3f1 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9ccaee3a7999a54ca00c2731dcf0ad66f3d21b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:29:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2d2b2595ca84fa1a48935ba387d273a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515ea64b4fee455cb79ff5a7a6bccf84 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2dd1a2edf0e4d7db9219d54b2d3ddca : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c315ea3dcfa443f873b6d88fcecaabe : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12fb4ab2d76247c48f456fb8c57ac5f8 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a37d8d126d24163bd625705a4347a78 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5f67405672d459d98846f1717062433 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5e95aa9c66e24c97b5bc13e1b096d9db : FAILURE in 8m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/510b7bbdf2584dc09cae53fc5f0c08c8 : FAILURE in 11m 50s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f3c49242689d45ecb8d04139f270c24b : SUCCESS in 6m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/546b1b2be529483bbc4d6534d1334ff6 : SUCCESS in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6f3c3ac106494291b5e176c7071860d5 : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e67023f4918b4985bcafb7bef33fc108 : SUCCESS in 10m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/599581ff9efc4f57b6df766c4b5e89a3 : SUCCESS in 3m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5f7c34bbee494f079eadd018e012510b : SUCCESS in 4m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f8087436c5b34d3cb5c426bb804d4ffb : SUCCESS in 6m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6ca77ed0ba5a497ab5567308746a6dd8 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/162a9bb9a54048208e2849a0028b7af0 : SUCCESS in 6m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/aac53627eff449ad9957c802a79e648c : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/047456e547dc4dd4834ba62746c63f8d : SUCCESS in 12m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7f31db1a72a946a8a59a5f33789df0f1 : SUCCESS in 4m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e39811f6b75d4109adef0ec7cb6ee7cd : SUCCESS in 10m 23s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8d0b76925cda4d88a119050abc0654cf : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d6ec8672f79047a09be1de6548e3ffe4 : SUCCESS in 10m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9ccdb1e503674aa5980bfa4d21cee120 : SUCCESS in 25m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d59d70cf8fff444bbb659e133fb8707a : SUCCESS in 14m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/110529a3df1245deab812119d9a48a76 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/612f3e607377427b98a04a5a0e27fa69 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4299361d9d0049ab85072328be13cf5a : SUCCESS in 21m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d849ba5449b14e85ae3538fb4ab1c79e : SUCCESS in 5m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/095721c2f539449b8c28d4162b182350 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3829230e739642888cce376de15fe8ae : SUCCESS in 5m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/46adbb28be9541c68d99d1a3de4b0b3f : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fac1f7590e44d2783aea122659cfd46 : SUCCESS in 56m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/080c3de527774afebb369c01bbeb53aa : SUCCESS in 1h 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"400968931e88dc46f897d7edba2ad0ca52f177c6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de7922f217d3ee9d4911c20125fa5bb6ef7952ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:28:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82cf65ebfe6b47abb686d34a4968bee7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7fe6be6ac0b143f6aeda51f67bba2a42 : FAILURE in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0ffa0f67f388b1b5045cea3b7e56c05a3c6c8de0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 04:46:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04f74fafa7594a5eb044780c1ca51f9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62234c0089b4905a8bcd18caded4476 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b1f2da4246431b96cd0d13d991e8ea : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69a3bac3834b413090546b873550cb0a : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cddf6aa84ca14226aaa584a9dc7a5623 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bacc51831ae404fb53d52cc1e40b7a9 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7689b71445b349b7965b6fb25dd9d2d0 : SUCCESS in 2m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5a1025dde4149668dcdc51f5f8b52b3 : SUCCESS in 8m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/61748a04a3b44c12a637de728e011ae7 : SUCCESS in 13m 57s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/011fa14c9e424b0d9bef80c7e8178d45 : SUCCESS in 7m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/03d643571329405fac7023b6623cc2f9 : SUCCESS in 5m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d7d094ec6fa44585aed61da53f075156 : SUCCESS in 5m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/05cc74c11d324e7e908f013a8f1c5812 : SUCCESS in 10m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b94da60fa8004ab38066fad9651ef485 : SUCCESS in 3m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7d9c2be6fc4842e087131c4dd103f4d9 : SUCCESS in 5m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/72e4d57576f54e51be3fe32ccbdac234 : SUCCESS in 7m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/701a94f9af0743a3acdb161a2fdbab94 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60b9fc9e89e54b70b854615d42ab89af : SUCCESS in 6m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1a33ac30361d480ea8d0c945199c7cee : SUCCESS in 9m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/660e369aa1b04902a591a34c5a518964 : SUCCESS in 12m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6765d229bb884f2ea99a5d285bc92dc1 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3c0dfb580e674001b63cd2b8a943aa7a : SUCCESS in 8m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b323917450324349862ec7c1a366029b : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f1f49bcaa6a147a29f1959c9f8459c1a : SUCCESS in 11m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/39bb432253404571a82a6f2b8baf1099 : SUCCESS in 29m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ffe94c1129ec4264a453bb073a7ff4c0 : SUCCESS in 17m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/97cbc2450f8749d7822dc5eff1ea0b88 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d01463cb0df041fe8e5f73b20059f6dd : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1472bab542934ecaa662772f6805274d : SUCCESS in 24m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f8c83b797de54749b21dce9f1609d5f6 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f504bc2270c041aa90ac5f82c0ac3583 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f27ecff8eef439a8bd1d924adc5dda0 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/94e4182276bd47a38c5d062323662080 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a08657724af4bf7ab4e6f41ef57f5a3 : SUCCESS in 46m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da19938b831c4a329ba49eb92ccf322f : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cd5ac709a71f46c7cef5c5039eaf0e281d0eb92e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c52f11117f7c5552d580bb9c191e4c53bdfe0228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:30:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c1da3a0c380cf34412ea73d653defb16496811d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:07:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a4262c4100b48318a7c9895474414d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c77ff83d3844b0a913d615278ed592 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffdcfd5d2f694e7faddc9f934d5d4f79 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2714820b8e7f4b1ca12c39dc09aa7f36 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c570d94815664a4c88b78b407c4fb584 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35f40eca1636412e8989ffb8f28f473a : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4a4ca6b14c647aa9386b350af457099 : SUCCESS in 4m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99c6d83bafff41b7a830b38b33403890 : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2c17acbc692743a884ae961e0dcfa87e : SUCCESS in 14m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cfbf942b5ae24ed7b1d60db3a5c1d490 : SUCCESS in 10m 58s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f9b732ced504474ebdfaf652ffef3e01 : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/edc36cef46664da8acfedce06793f727 : SUCCESS in 6m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/46ebf4b5893a4f1d93a35f21f232a81a : SUCCESS in 7m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/47d59be8401a4d20a06caa885ad67b6c : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ead5f31449d242b9944fe45012dac558 : SUCCESS in 10m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9d420180c959415789d96bf825b62ce4 : SUCCESS in 12m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4aacb988dff14401a526d7cc7670a7d1 : SUCCESS in 10m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1f02f377c04145629e0ae4c21d6042b4 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/26f3bd313da04bd1a03886b838d92017 : SUCCESS in 10m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9da455cbee214258bd2ffeac8f0c8ca9 : SUCCESS in 4m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/440fd258093d400dbd0c648621aa357e : SUCCESS in 33m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/db397fc6f8d044a1a84488ba02bde9fc : SUCCESS in 12m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5cebd0b23ffe47f783056acecf69ac50 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/214c6706cacc47ed9582662659f2b51f : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ce9d1dab5fe4ba882d9e9dad583c96e : SUCCESS in 25m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c6bc3864cb8a4443994f0c223b355f07 : SUCCESS in 5m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9e98d797e27c44b19c9d05fea4da499e : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d59bc174180447f7b0377acb33d16325 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e887bca754814d23bad6a12b58f7e797 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01878dc970934fb88406447e7534312c : SUCCESS in 1h 04m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3af0cba7545642cc9133b1a545b295b5 : SUCCESS in 1h 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f52bd936cca5064775f2f820f67a2b7246987501","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:07:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d280ba651255e9a4cfc1114157c6194f3b24e90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:08:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a91877e34946e3910f6a0ecc338140\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0866a541b9b4e519c8cc9b2decdf8ad : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5d747c0c7ab409473b8b0d0c81075046ff21cdd","revisions":{"b5d747c0c7ab409473b8b0d0c81075046ff21cdd":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:07:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/873384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/873384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/873384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/873384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/873384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/873384/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:07:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:07:51.000000000","tz":0},"subject":"update constraint for glance_store to new release 4.3.0","message":"update constraint for glance_store to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Pranali Deore \u003cpdeore@redhat.com\u003e\nmeta: release:Change-Id: I239b4fad3cb9a143ea6d5bfb040c86bff74acc1a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I582a316720eb480b87148e4c9acfb25808404c85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5d747c0c7ab409473b8b0d0c81075046ff21cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5d747c0c7ab409473b8b0d0c81075046ff21cdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
