)]}'
{"id":"openstack%2Frequirements~873385","triplet_id":"openstack%2Frequirements~master~I4bd1b43fedb3b2e293fa8f09a50faa85182062b8","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-02-17 12:26:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bd1b43fedb3b2e293fa8f09a50faa85182062b8","subject":"update constraint for ironic-lib to new release 5.4.0","status":"MERGED","created":"2023-02-10 16:09:07.000000000","updated":"2023-02-17 12:27:23.000000000","submitted":"2023-02-17 12:26:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873385-new-release","meta_rev_id":"6f367390b9556cd3b6247075f2695c52363f7d0e","_number":873385,"virtual_id_number":873385,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 12:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 04:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-02-17 02:41:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2023-02-17 09:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 04:57:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:01:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 19:42:59.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2023-02-15 19:43:10.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-02-17 04:57:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-02-17 08:23:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-17 09:43:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"7b68639d8dbfb3b5ccdd45ecd34cb80575d22626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:09:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ede88c7f46ea6c2797127e52e6e150d81a7f4d56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:14:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5d371511e64fac8018f524da930d46\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4c5cd698377c4ba1ba8c483002df6bb2 : FAILURE in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"06453aba2cd60cac37fe25354a65cbfc60bcaa87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:01:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0e8e922f8db411e9ba0e3d43ffa8e35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b6427498421430599d85351f7b8ea86 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af49c031b98d472796b04cd746509b98 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/734bb0f151074d0ebd9a34099807e3cd : FAILURE in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46ff2d70c34540118c2c168cfb275701 : FAILURE in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9aec00ddcc4d4e2bbe7879df475165ac : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36ef1576e3c74d66bfd216836b94c42e : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9847b5837cfb43809800ac04c3e6d6ff : FAILURE in 7m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eaea1e592ec34765b1b8c7de7c1c9a76 : FAILURE in 11m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/36359a8b479f42e28f0dcde607cd0b9e : SUCCESS in 10m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/808157a3731b4eba970f2d8d6adb0180 : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e55272f347dd497983bb93352aa94268 : SUCCESS in 7m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ba5d056605744901994f40ae93b9efd5 : SUCCESS in 10m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a47336a1c72c4fb49949a6909685cdb8 : SUCCESS in 4m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8908651caeb349ef96a213a480f05753 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0a8c562eddbe450e86b35a16a20a38a8 : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b0f0a9269f75423cac8211076a0e4dac : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b4a0d30b10b46d0a3524bb3ef19ff72 : SUCCESS in 12m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f962ee05f2f4429da0d36422884adaba : SUCCESS in 8m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/59b983901012466da898d49c3b93799f : SUCCESS in 12m 39s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b59d8d877f6d4fbca9a9d72e9003ca67 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d5d52552285343ad8ffa83bb87eee8f3 : SUCCESS in 10m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fc23063764ff4f1eb7cc5c097c86c7f7 : SUCCESS in 5m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c7ddbdf116454ab8be368c408f1b5bea : SUCCESS in 9m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/26fa10800c724472b1c0dcc1e0850e2f : SUCCESS in 28m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d7d7bbee71e949639f517383823012c0 : SUCCESS in 11m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f6ce00d649bc4021af324d0b949edf98 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7d16c2a1e7594576a0d94e9b1e3d4c1c : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1aabb7937bc045c197ce20f9d2345660 : SUCCESS in 23m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0ac958416a894e24b904cca092eb54c6 : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a1539cd573f8443a9916330b3e0b37f5 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a5ca90c4fb14277b541060e556c07b7 : SUCCESS in 6m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b2d59a98929346b288495ae6a6f51418 : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/365b4af197c047d18d2a3fb0e586a300 : SUCCESS in 59m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d37fef2e209464a84e84d736fc48b24 : SUCCESS in 1h 53m 22s","accounts_in_message":[],"_revision_number":1},{"id":"be74ec759b2fe547e72a499acf4ddbcc64dbd18f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3dfef1ae9f6a92b67363f90817b1f3a315d56b68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:01:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96aefd4e5e5c4bae93b2a651c2f4588e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8f0620e647a1447bb6c6d16a6cf435b2 : FAILURE in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d34040e9e12510fb999d949e6e901c48e21361c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:01:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a114cc9cc284c3ebeef42c24142855c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10be707ad9dd4e7d997bbab6ce8b9e60 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a3878e39c849faacd2a1ea4253d357 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1ece2ed6668423d9b04fc08c2417bde : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7762a262017448bfad3093e0d428e3fd : FAILURE in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a34c978b228482ba61f8576b09a5330 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70b6c82817c5468d84f5a1fec60f1244 : SUCCESS in 4m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46d6a04c0a51471089f5df2aa8ff88a3 : FAILURE in 8m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fc804df3ff29459c86c4814d3f96832c : FAILURE in 12m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a0c85204d3504be9add611722b29c6d0 : SUCCESS in 10m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6571ad035da54e84bd5d375cb21298dd : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3def45f5ab574e988e877a33f5914506 : SUCCESS in 5m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1c81bd2368284b95a05c230654709a34 : SUCCESS in 11m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/809e99565c454a4499af377fa2aa4924 : SUCCESS in 6m 35s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5a369ddc7bde4ca68b1797ce55f41379 : SUCCESS in 5m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0cd8d01fc8824eed952cec3209c3fc92 : SUCCESS in 7m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/913f69149ef3464385d86dce4c563768 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c59f1d017f794d578e60f6b3526eb599 : SUCCESS in 10m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c5287a37e320483d899e2b8a7557f642 : SUCCESS in 10m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/567b3c693ba14fa5b1c09f43ca6bef6c : SUCCESS in 27m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2443cecf8310492387f155abdbfd20eb : SUCCESS in 4m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/974db047b3294995a0868eb0037c7bd4 : SUCCESS in 8m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/134e86914960447fb9cdfb475a1c238d : SUCCESS in 5m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/22b71fef71e14c7aaed011e9997b4656 : SUCCESS in 10m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/084b44d8a1c14c5488071bd25aa76d50 : SUCCESS in 23m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/88f5738ce8fd44f49676b197358673de : SUCCESS in 11m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1348c35cd3a24400986ff1f28548ff07 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/393713ab0abb4b04933ed5d3bda1393b : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca7eea16b5d848f3baaad13f01eb2cf0 : SUCCESS in 21m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3efe2f57e69d41d6953b60fa5f7fcc76 : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e29b30d8b8ff4f29a19116b705f5958d : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/251d50b8b93c49d39040c6e8a1ce690a : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/119ea0f35eb441e1bc5f0eba89bc88a3 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ff7638a4f9b40b3a6ea074806fb5f0e : SUCCESS in 50m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a37967056e4e45368dbe6e8cdaea27e3 : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6ee4a298e6111bb219740a2f774f3845a52f5551","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d22b3102398e63987c789278a803c0d5ccad0445","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:04:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e1a58d1dabe409dbaa6541e3a425802\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/63ac9e8af75a4913951b3f3cb441b750 : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a16856461b326e40f9e105d4dc577ea0bf25f2b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:08:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de04bf1043a34fd99efffc6a6fe03a01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e371e1f0e054e4492bc3eefe0753020 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52ea9703bfc49b2ae003954f4b92188 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ceed48b83aa34e24a393fe8762959cd7 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9fbaf1b421842f2b52079947931702b : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/450cc4d97d3b47fea17c6db68209e497 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c1491bef9ad4994a3bada696f61bffd : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/047ac0e963f94a08874a73cce42cfcc8 : FAILURE in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0b874079a8df46e599d7031671f2c305 : FAILURE in 14m 33s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0e4d339229a54749a866554a60b80499 : SUCCESS in 9m 06s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/169916647fed4938b722aa86d1a1ff51 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/189e5d80994c42c1abc746de531f920b : SUCCESS in 7m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/35a869c85f634b9db70702b822850b68 : SUCCESS in 11m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/47bbafc242c64e7f95487b15c07d5a8e : SUCCESS in 6m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8d56ac23a6fa405f99ea1790a5227dd1 : SUCCESS in 5m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/01f21985a457471e8c9261b786640d74 : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/562fe3a501f54065a349069593c6f9d2 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec6afdf5f3c74fb6a2d2a90ea47f5c8b : SUCCESS in 8m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e72e082a12a840a1a0d00744f253c296 : SUCCESS in 8m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/28ef2366085148b5a1f0f6f55bb14c66 : SUCCESS in 29m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2b64f00e8f144deda22afe4f4c89fd51 : SUCCESS in 3m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e56492fafe0c4fd6b305aea7f9ddb4ef : SUCCESS in 8m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b1be225f357c4032a27cbd2dbcbb1314 : SUCCESS in 3m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2d889b0244cf4625aed438d0c7f78dd3 : SUCCESS in 9m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6b14e45b26bb464a8b9df716e8e69ec6 : SUCCESS in 49m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bf45083d38c945659743f7b28cf91a8c : SUCCESS in 14m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1a20a0d39ab74581b1904517c881b20c : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49d894ff4a954d8ea57d5deaa6b5d26a : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00860002fcbc4e44ac35bc60e6e80b5b : SUCCESS in 22m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2ca85d96417b4479b63dd0679a61521e : SUCCESS in 5m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f872778db99f402aa5e9fe759f5683ea : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7ddde3a8b7b41b0b87708d16934dcfc : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7619af7a477c4addab085a7bcbf042f9 : SUCCESS in 9m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9e2739c17704e1dbbcbcc6bece89b06 : SUCCESS in 49m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b76a2b70fc241bc859c1f95c32fa47c : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"26d2c7ebb6840edfb4a631ace21c265e3292e304","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-02-15 19:42:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f2c4b39d77a11bfc98766ed17660676263773f3","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-02-15 19:43:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad8217a0a0a39d6ea7cf1d3e1581b0c9fc59ac18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:48:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4faee3ba88854f0b8c7a7b66853fbc7f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/346985f1b7364e9a8d3516887b757947 : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"13819db5096b314d4de1a835d8cf99313a850f8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 21:40:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be4b875cdf304836a03e10fe6b1da9e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471058c528234688afaffdf69f8c325f : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058c05365d4c4b3496a18148017874c5 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd1dd547d9a84af48f4344a32c6f0fd5 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d800628a1d49456fbd4cfa201fdf973e : SUCCESS in 3m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/352c645f89d7467ebb04fbf90224b910 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6afcae25fb7a4ac4bd46d27ab5d082bc : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/764c7dc794f84477b705d2c4f56f7034 : SUCCESS in 10m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7225b577b043498bbdaca7502ac2438f : SUCCESS in 9m 23s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/eb9c6016cf5645eda4ce17ca0cd4f760 : SUCCESS in 6m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e0c72b890c5e4cf8a3facaa3e0611a50 : SUCCESS in 5m 17s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/76ba8d72c3ce4d7eb41f0be0118e3d49 : SUCCESS in 8m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4c80107b469e47bebff4eaea62953834 : SUCCESS in 9m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f90e5ba664f64e5a8271b4aa8ffe41d7 : SUCCESS in 5m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0f20723971ef459c8e75996aae285239 : SUCCESS in 4m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/83b0f8c28a314036aa7b764174b6ba30 : SUCCESS in 6m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/72f3378c19b34c6db330db4a297f9db1 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6b997e3ac6e46d6adef3d846fbc2d62 : SUCCESS in 6m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/17c27051fd9240e596dffc666d96affc : SUCCESS in 11m 47s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c306ff1fb7434a0f80691c0a0f5eb599 : SUCCESS in 11m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d7316127cbbc457aaa8da94f41db6c7f : SUCCESS in 3m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/63b92d809c1646b28964a7a6f6f6e1c7 : FAILURE in 18m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d0fde925069d46eab0adf8eb36b1e13b : SUCCESS in 4m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/efa97f1e4a824755afd769d77876ac70 : SUCCESS in 11m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1e98ff1c085d4565bc7cefb0b7799978 : SUCCESS in 19m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5f79b19e2a5249dbbf9418b1e3550f62 : SUCCESS in 12m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d587474df3ce4f2d87101e0d7c0071b8 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3223343e69a04271a612c952a7010f2c : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/179b0ab7717b48aa82cfc000e36aa764 : SUCCESS in 23m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/edb7730993fd459681af91c217f96118 : SUCCESS in 4m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3e015103628a478ba1a64c632baa818c : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63fcc7a0bbe24f469d2aa3f20be639df : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2dd8f12cc4b84ef88b1b9b12bc59fbf6 : SUCCESS in 8m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69c93372b5e6424498602cc4e72fe0f8 : SUCCESS in 54m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/623744ed4f6d45ffab37b73fa063c100 : SUCCESS in 1h 50m 19s","accounts_in_message":[],"_revision_number":1},{"id":"41e247dda66cb70ea98ded4ee5cf2ec6ddcb3ed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:41:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"09cfc4f2b3d5ea078ac1d646dc672a408988ea1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:50:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e75da536a4f14436aeae074af431d26d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f525b13be758408295f95a91b4bd5b80 : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d4f3b3fc2fef930b6f5788007fd96017f9261650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:41:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2ab10f66bf45239c078ee94677c329\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dcaf2067e944df892ff8ac805cf38bc : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a829c9bedb42bf91ee3b1afd1ed24a : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a434be19f546bbb64a02ca01689d8f : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e85bd19080974cc3b4161f2ee9ff6227 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1367384c6af0480a8391875a81c65f30 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50e21903df1f4edca185a4b910f69fa4 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef9307f3c1ed4db5b336a94a845f3fd2 : SUCCESS in 8m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/800e597e169644989b8f6cffc3206946 : SUCCESS in 11m 42s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5af30d04c5154b1594014eec967c0a47 : SUCCESS in 7m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/78a9afbca91141c4baa48df29d0ce73e : SUCCESS in 4m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bd65d32cd27f4b22a5967c95595718a7 : SUCCESS in 6m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/29a9955d0f6a47fb83484b3fecd18d78 : SUCCESS in 11m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c4b9a2fa72cf4fccaad9a228f9b952a5 : SUCCESS in 4m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/419097407bfd4f36a1fbe8e326aa581c : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3f4747f44a3a40cc866e0e022e7de9e1 : SUCCESS in 7m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/06fdd05894594f8da2899323a9e06bfe : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/583c4bebe7914a498a41235de8fa8315 : SUCCESS in 9m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8b1f93b0e7294fa980c4b1a4e7b0e797 : SUCCESS in 8m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f720f898dde44d318408f8047fb5675c : SUCCESS in 10m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/919c189d1dbf4412a2bd90e5d664ad37 : SUCCESS in 4m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/df87cdbcb4164934906550360ebcc1dc : SUCCESS in 8m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/92552f529900427b80c3a59804bad766 : SUCCESS in 4m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e83c2d0e8c6f4090814a16895bd44ac1 : SUCCESS in 10m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/badd65c9b73a4969b37b3cff4149d287 : SUCCESS in 21m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a20c001c5a50416d91a0f858ef4cf77d : SUCCESS in 17m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c6607b7613bc48d7ac868ccfbfc1a5e4 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/370db439124f4ac59ec091b8236ae1fe : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf78ad2550564b27bd020d303fa4b2c9 : SUCCESS in 27m 40s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4b141941c5f84a44bde0e38853ac67c0 : SUCCESS in 5m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fdb94c296c50437ab9cc3508329b432c : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81aa03061474491893c389aed1cd2c05 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d38c4a1358f84a619491a7463a01983b : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2b421028c7d48b1831c43530ad4355a : SUCCESS in 1h 01m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d58e9994b93a4f0590b7c1248dc9ba40 : SUCCESS in 1h 34m 21s","accounts_in_message":[],"_revision_number":2},{"id":"46b32a249f5a5e63fe0b400396c294b4ec5fdcf7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 04:57:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"78e43edf289b44586ec61e2ec5bf68672f1770ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:58:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"06a70337ce1a1494af3e3646590622ca2b23a463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:25:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7adcd6e756ab443090538d2b460f2923\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cedc069785a94f2ea5a601824577761d : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b4c071c7074665989b160869e6a340 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cece7e64e874c6ca973c128ca31aaf6 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f5c441885c94d41ad99882d3f7dd9af : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23b9c4a714f648ff996f3a56d0e67b36 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ead3c80eb574ae1bc726679f3f955ec : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/582f29bd7e0c4fed99e543042cd29d35 : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ab741cb620a407e83b27ded4278c6a4 : SUCCESS in 12m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/61fe6c19ceef4abdb9ccb8e4d55d0171 : SUCCESS in 13m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bb3b6ffb9e884e83b3c3f88d79ee51da : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ed2d69febffa489e8da04f06f2d2a582 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/841ca5b5caeb457aa1c8ce08952d348f : SUCCESS in 9m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7e907ca4a1ca49d68e20568db55a7d64 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02bcc023458d4d82b4eea1e36448ab17 : SUCCESS in 8m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3bd95cdc404241be9da10a6f31cee9c2 : SUCCESS in 11m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/88fcfd81b126418088660d372eff2169 : SUCCESS in 10m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f44067e2a7e24e20adc6e05028574789 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a35b51c48896441f963c0ccf2b9cd6b5 : FAILURE in 18m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0269642779ed435fbcf0abf1171a4083 : SUCCESS in 4m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ac40c8c5ac81404488b8e40541a123ee : SUCCESS in 24m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/68bea6c369c145dcb1628a2d7bc5503a : SUCCESS in 24m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a9215959fcaf4532932b4e310ecea7fe : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/35d7439321804889954d7f6d50d6ed5b : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e475a78bc3541b398f8c5eb2fb38150 : SUCCESS in 20m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ef932e0ea96c448ab280af5e674085d7 : SUCCESS in 8m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ef8ad1ff39da43a39fb3e9d189d9ccef : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aacdec2c475e48a785d617ad9cfb36c6 : SUCCESS in 8m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c3be750653cc4c408c7007a79ac4530a : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ce0296b506946609957af35bc85c18e : SUCCESS in 1h 03m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2014976176e454087bbdadb45037a88 : SUCCESS in 1h 26m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e036f8c963c3ed84ea57e839acde75bb1a14a58f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 08:23:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b997d9a8d1a898c283da963e1b9d1b07a9d64a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:41:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7480ce48a348c1a7b5a9ab3e5c013f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0043c00c50e14eac952ba95194d04653 : FAILURE in 15m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6e99892a1f39cac30a0be0afd6e797ccc69be82c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:43:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43c6609e63ef0b9d5dc4f72e86e696185b05d71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:45:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7579da18d9f3466488d3d449ca57198f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c520877e0c04515aedc58e0db37e7dc : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d0dff49abdb490e898361d34eec4f47 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e39d36187b045d9a12e5e20ef3d8c5a : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/387a5c22ffd245febbf5209454e387d8 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee2a34b1dd24444bacc352c57dbf61fa : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddf91f0f03824902bfe061f4cbcc8142 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f6906e2a71ca4505839d002cc8e38925 : SUCCESS in 10m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eb496be03dda48feb85b0c58532225b2 : SUCCESS in 9m 00s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4e05f69ab4774bd1ba8b5bce4e8b2887 : SUCCESS in 9m 28s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f3c761739dcb4c6e99da1f870fcbad41 : SUCCESS in 4m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b8df29fa08f946ee80d6be30241345f0 : SUCCESS in 5m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c037de5da23f4a3da32b82d953c6f574 : SUCCESS in 11m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/17cf4ca72fd9498b915efac49a85c910 : SUCCESS in 4m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b84a77ae187a49ebb013d2bc9661014d : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4e397571f3ac469da1fabcc8bcd2fc99 : SUCCESS in 9m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9a56f56da655488681cf438a0d7e4e91 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9bbe2504dde24cb7b1d1ec4c4239755b : SUCCESS in 6m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c381e88adb744747af1a5b901ab880d5 : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6b119f3cbc14442fb555d061be166b85 : SUCCESS in 9m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f14c942073514e1aa510b407eb8b12e4 : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5cbb42d90d79463b805c364de55e17fd : SUCCESS in 12m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b72aca44dffb45b0a14551d79f2cfc12 : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5d87df131f1645bc9b57505caee33e9e : SUCCESS in 12m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/43248961ec85448985f9072a6d87b45d : SUCCESS in 29m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8c63448ec51c4253b7925bb2980df012 : SUCCESS in 11m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b6fe7618e2b84f91bb538dc9e5b88da5 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09e2c97b3fcc4c33bab615e3dda5bced : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a59f39b4b1674ac9b5482641dad77173 : SUCCESS in 21m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/55b26ea30c8e47ca9205395d1745671e : SUCCESS in 5m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a22867c8b58d48aa9c7445df5a13793c : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1418f929e30340f7a47850817fab34a4 : SUCCESS in 5m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2893454ec7a84b34aa6b3889d14c5e68 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe7b49602fe8495a9434432353af7420 : SUCCESS in 55m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e80f9252eb3416894931d9fa965a29d : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e36ff13d04499a723f741e4d2a9fbc832067cd12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:46:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db51b13eb6b646ef8668180cba97a8bb3a86070b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:01:17.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8748cf0be440c7b953219c4f666899\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d3f1cd2067384e22acfbfa86b0a24026 : FAILURE in 15m 08s","accounts_in_message":[],"_revision_number":2},{"id":"52d9a464f430b1301b92a3bb6127ac497956e250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:26:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a738f8b58ca4f9abbc3c63a5947cb9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a64819e7eda4b9e8501db23697b8bba : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6125f34fad04e04ba7951a45427ff15 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f662aaded0164d3e8df5b4055fe508ad : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb786be9b55240aa8e777fffadccbaf7 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66a73336efab40e1a431a3b8f9a9318b : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5fa20f79e054be5af9bfcb27b1aaa9b : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e1eb9f7f72bc42f3b73abae15e12c5c5 : SUCCESS in 9m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dbe1964d748f45b891530202a6d2ca52 : SUCCESS in 11m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fb3b4bd6f9d84ad6beb194498ef51983 : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4bf94652de2f48fb94cc9ed070a89de2 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d083764629e74c04b5f2ed6959fc6d04 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0c8bccf6c520471e914b0f643b7225a9 : SUCCESS in 8m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8c0ab853b7c64219bc0b8e4d08b9fee4 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/464fd5882fc24547a588f1231ed2e7c3 : SUCCESS in 11m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7c91f93fd7634db2b6fcf100eb709b9f : SUCCESS in 10m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b89232170f4849649ee6e79322219464 : SUCCESS in 20m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/db15a6b2215f4ce3bde03bb4e30b6ff4 : SUCCESS in 3m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c075c0dbac224662a2d594c0ec003434 : SUCCESS in 9m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/924cbb26990f42fa8065d0cdeee1db65 : SUCCESS in 4m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e3b7b4b806234a9aabbb5195cbc73acc : SUCCESS in 24m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2590957ac0444aaebda9827a093c4bc0 : SUCCESS in 12m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/17a07bba5ed8498ca63ac26bf966cb43 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f71cc3854f64a3e9e7aee0c453153ae : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e33b97bac8542fa81add5002180ec7a : SUCCESS in 23m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/976cb2ce100e43519299df7b03c919ad : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c0f4baed38764ce293b11325c0f88587 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a75469190e9340329127bb69381ae973 : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3ce652d94e1e4820800cf9edf9d939d5 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a90b4e093f7d4114a31a3f447682fd2c : SUCCESS in 54m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/513e821b9b4646a6883f77be782c43b0 : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":2},{"id":"64d7ca460ad247da599a512ec260658327283c25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:26:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6f367390b9556cd3b6247075f2695c52363f7d0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:27:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57cde3cbc82b45739e5a48e689ff5788\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0421a88d17374d7ab3aba37b38753b58 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"65000bc31a5a53975280d63b12442c4d25db92f2","revisions":{"e59dae9edf07cf4380c7c4f7921a709d5634ec60":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:09:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/873385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/873385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/873385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/873385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/873385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/873385/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:06.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 5.4.0","message":"update constraint for ironic-lib to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: If03219739a40473bf5bd3912759862b5ca6d7d71\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4bd1b43fedb3b2e293fa8f09a50faa85182062b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e59dae9edf07cf4380c7c4f7921a709d5634ec60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e59dae9edf07cf4380c7c4f7921a709d5634ec60"}]},"branch":"refs/heads/master"},"65000bc31a5a53975280d63b12442c4d25db92f2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 02:41:55.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/85/873385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/873385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/873385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/873385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/873385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/873385/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:06.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:41:50.000000000","tz":-360},"subject":"update constraint for ironic-lib to new release 5.4.0","message":"update constraint for ironic-lib to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: If03219739a40473bf5bd3912759862b5ca6d7d71\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4bd1b43fedb3b2e293fa8f09a50faa85182062b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65000bc31a5a53975280d63b12442c4d25db92f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65000bc31a5a53975280d63b12442c4d25db92f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
