)]}'
{"id":"openstack%2Frequirements~873386","triplet_id":"openstack%2Frequirements~master~Ia2df0e26b7e08bcb65904f9fa506601108894511","project":"openstack/requirements","branch":"master","topic":"cap-virtualenv-setuptools","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-17 02:08:07.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-14 09:09:17.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Ia2df0e26b7e08bcb65904f9fa506601108894511","subject":"update constraint for ovsdbapp to new release 2.2.1","status":"MERGED","created":"2023-02-10 16:09:34.000000000","updated":"2023-02-17 02:15:27.000000000","submitted":"2023-02-17 02:08:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873386-cap-virtualenv-setuptools","meta_rev_id":"38026ded8095fd9e0fed8998fae71b819244ba74","_number":873386,"virtual_id_number":873386,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 02:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-02-16 22:08:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-17 00:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-02-14 19:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-02-17 00:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 16:21:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-10 18:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:58:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 11:35:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-17 00:56:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5df9dec3ab4d82f0f350962aa0f0fa7d56148ca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:09:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"284357a30079d6aa349d19463513a6789d9a526e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-10 16:21:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"49b4536504db33c58384e26743274fe42004e1ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:13:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49cfd8ae0c174d0ba713d84d23c95964\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/265db949b73c4d24a943bef494890d38 : FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ab7edc85bc7fe38b187ebab0a455b216010f347d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:58:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a023bf717f94aebad0ed5106977cf68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe157da684164bfb93c546f7e78105cc : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c83b2e659743a990c62d2b9fd32359 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6858361329d245c2804b650bfe331766 : FAILURE in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65ba322d56594e84996fd11649687b90 : FAILURE in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7866a6d70e2040de889d1f40a0e7d1c0 : SUCCESS in 2m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6097fb67fee84ac2bfd73e0f569a0436 : SUCCESS in 2m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5036b0bd767b4cf5b27dbd5e42abfb3c : FAILURE in 9m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b417cb7b7a4e426395904bba5ce6c70e : FAILURE in 16m 02s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/40120fea6b60407f9ff4917466690ed8 : SUCCESS in 7m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a4a5aca2df6c43c78f5eec08e22c67f1 : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9923b63f0063474e9015882060a7aa39 : SUCCESS in 6m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0eee76217d6c4666961c4c1bfd530449 : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8578831df3bf435f9e0afed877bf0b54 : SUCCESS in 6m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6d2fa5d7fae64fb68724ef0427aee788 : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/885de28d596f40f8b9b3fc3cd04976dd : SUCCESS in 7m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/62f6d3189b1441f8bf31cbb9a585ed8d : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8df35970748a4b2b87835d246eb80bf1 : SUCCESS in 7m 35s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/45c8da6fd0ff4b189ea145a40558af8e : SUCCESS in 9m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/fcb6af80f8324626a0b06ca31fc6e59c : SUCCESS in 10m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1e34c61ff4cc4f0fbd430b02498593fa : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c9d322a4a53944edba5b3d52dc2e96e9 : SUCCESS in 9m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f2c9212dd9484eb08fe065e36038714f : SUCCESS in 4m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3140ff16115844c2bb94d604d0c3fdc5 : SUCCESS in 9m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4588c366da4446f8add3f862ce170e42 : SUCCESS in 35m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/19b4e2ee8ed448c39729424be7f8ade2 : SUCCESS in 15m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7dadab753610439e991a4315fde64ab6 : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47f378e665844cf3af149305d8d29a4c : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f28b81e4bf1949dc98012b92852ee572 : SUCCESS in 25m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a3f875e25b6e4c0ea3062aaf1ef5cf5b : SUCCESS in 5m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cdbc53f497444e248879d29a39b9bda2 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/695d640255494708a16d8f4f40ede715 : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/39979c62e0c2468a8dfd0eb04d3a78d5 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dd301e9e8b44607aeccf6b0af0d3de5 : SUCCESS in 1h 06m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcf7184c0b9148a2872886a73b98ea08 : SUCCESS in 1h 48m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8dd5d1e9fc0d8fc1a1b039a50768304818b9b97a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-13 15:37:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8bde0c242598d870dd4e139b26bc1dd66457a4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:52:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d58df5da4e44ffba5d57e22029bce1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/44c96662e6df4b50a1e19d1d0089526b : FAILURE in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9a2eb8e422ab5abd19758c9cb7e0cb56a66d6413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 17:58:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04e64edd45194199bde0f9df0cdb3e31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8825273d9761405789b07be5a79fcd15 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925986208cb642e886204039904d3d2e : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69c9627b452646bb93fb3e5386cffc45 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27dfe1df642e46ea89e4525f7a7bda95 : FAILURE in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc28c651ef544023a3494a708a858c83 : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f6cd423f5c7d442cb538036ece7f076c : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9440bf44d9db42da999d34bcef7db305 : FAILURE in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb80da8d0d214ce997a778c2d9696a4d : FAILURE in 12m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c50410d10c52419d9ea04df08f814d92 : SUCCESS in 6m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9f2a80cdc9654c4aa45cd0217bf30873 : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b628b6d37d104b309eb46ce7cb06849e : SUCCESS in 5m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b0f7d0d0221344edba7351a39588de37 : SUCCESS in 9m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8c1ba3698b2c489fb1762d787d0cfe11 : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fb0512267a064e5c87d33939b765dde7 : SUCCESS in 4m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/be78cc97a2e2486680aaff71f6bdc854 : SUCCESS in 7m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/afbd3425a0564f35b67a1db9944912ce : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d3e5b5fadb94cb0bd2073e4de696958 : SUCCESS in 8m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a538915b52614c2a8f4b968acafd616f : SUCCESS in 8m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8459489f01e24e3888747ac1e0ffbbc9 : SUCCESS in 9m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/de4596fbd9fe4ddab4c3cbbd8c6e956e : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d0b971d4ee534a7e8ec8715ed9ed4ce4 : SUCCESS in 8m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bfab522d89794f5497bf4b44fe9320c6 : SUCCESS in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/73e314cf57714d029755f70ce86450ee : SUCCESS in 8m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c06fa207da4d4ff78ff15c48812509ac : SUCCESS in 25m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4d9712934172430fa9921d8ba899d004 : SUCCESS in 11m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6625c9e887c34bc299ebe6e078232d5f : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afdef217de934f59b1cb1955e3562701 : SUCCESS in 2m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/856a5eeb2e9a4102a8a4b06fb3078031 : SUCCESS in 21m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aa49972e855546aca509b268ff7d8be2 : SUCCESS in 4m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c38d726f737942d2b8922c560f154b2b : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7716c6eca1254259ac908ef37a93dac7 : SUCCESS in 8m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0b101c141f6d49768c939a618fd71292 : SUCCESS in 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8ac2021fbcd414f9b16e3302d89c745 : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f4913f880c64e80a91ee71eea5baee4 : TIMED_OUT in 2h 03m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b10805b10730c659bd9d5dbbb754a8ba74e240e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-13 17:59:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4f02ef5e51ddf3d496e4b3a2fba60675b81189e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:05:01.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9929ffe296e143278bb18a318d9010f0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/91d06db9093142499cbea786b4889792 : FAILURE in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ac01582d95672e6300e5db359e4e338d9742bbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 19:33:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea3f85b3918b4ee4b2bd8b71b7c3a93b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f9fea2b76c43bf8cf6c4f172b6541d : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4532d787fc514927b73bb197234c2865 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbcf992c91e14feebf014f7855bb0649 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e62625ef33ff4823a46aa07b1ae78519 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/689dcb3870c2431d866d56b35c5102fb : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e692f786b57c4184b4c66b7e945e44a9 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cfd3efc6640d4108960211445e6bb7c7 : FAILURE in 7m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3749edd6b6ec4a038c8a3d88822945b1 : FAILURE in 11m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8e5588f7576140b08e44ede0e8191a8e : SUCCESS in 7m 06s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/67181a33a7f447599685e6bd81802ae9 : SUCCESS in 5m 07s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/58e21137b92947c1b9229ababad7f2e0 : SUCCESS in 5m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/37907fe72cb944548e771ea45d252bab : SUCCESS in 10m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3a1e485685024ca8a9e6255252ddbc2a : SUCCESS in 6m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8674baaabe3b4b2a859c81aeee7ef3ee : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fe72638c965249ac903db908e705f0cd : SUCCESS in 8m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b69c1fd1f6df4eaa8240fc817efe4f59 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/12805e43924f4f65967fa61c82f94699 : SUCCESS in 8m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/552301ad269143bdaafbc25a2bab68ff : SUCCESS in 9m 25s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2fbdd3be8d70435e91a0d9bf5f4282ea : SUCCESS in 11m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0330372660934d2ca03cc2b32710c026 : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a70ca1a825f4400cbb397a2f4b1fddc2 : SUCCESS in 9m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c1b124c15a224194b0804734fee32f70 : SUCCESS in 4m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/79e7ecc8f6674cc09c8a77227a8dc6c8 : SUCCESS in 10m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4e17ab026ae04136850210d2dcf195ee : SUCCESS in 25m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f66a18824a694ffcb4b72c73f3d7cf06 : SUCCESS in 12m 12s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7435e92945f24c77b7c69f1abbd82a81 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/566a9e2f9b0640f28d8fb4c36d495cb9 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97a373684085447fbe540685b8827d6c : SUCCESS in 22m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/807d070aa7904753aa27f7ec1837c458 : SUCCESS in 6m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/316e0a4e05f94e8ba5161d2e35d38e4d : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1ba4d4dbf224a6094d54c22fc423979 : SUCCESS in 8m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6e12bac77e874ffda88273097d2e41dd : SUCCESS in 8m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8e87bf307a44354b64e1c9003c8715a : SUCCESS in 42m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a228edbaacd4051a4b3ab87ccabb4a7 : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b21de568e3a74d8d935553a9dbb84b3cc82c7720","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-14 09:09:17.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"345c56a692f38fac95dba8af437c9f613262ef58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 09:16:37.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d84b9bfdece846f69df6c063de4940af\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c3682b79d004de0989dcc5c86826bca : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3edf5b79f4029854369a05152b5b962564f89844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-14 11:31:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"83130ade08b3c5a5375ac59de0f548a480f292bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 11:47:51.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac7c7647c29049ce82fc287464d246ed\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c558b5caa6984305a888a178a75d812e : FAILURE in 13m 32s","accounts_in_message":[],"_revision_number":3},{"id":"9802a81053077e299ff037674ce910f129d80219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:42:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/420f3eaabcf04857ae643629d85e0e64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646d356b91a74eb5a9904ed58a4c1061 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867732f1b43a4913b4814f74950099db : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1557d6b822fe4816bb55c2189c07c716 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b54b4bc770b3499e9c587ef5e7b84ed3 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db91f05a06c64bbdba2527db8576f61b : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d728cdf4cea345d0a8bef4a64a587b74 : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5da9833e8cf24b78a319db0775e80436 : FAILURE in 7m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9cff3f21c7ab42c19a2060d726d09721 : FAILURE in 13m 59s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/20cbc98a42e842998cbb296c380d11b9 : SUCCESS in 6m 33s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/529ecbc65008462e94ecd30cf3b86133 : SUCCESS in 4m 50s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3e85247433f44e6491408e6dfe6e99f4 : SUCCESS in 7m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5c675988e8d04ea9a63f1521ed8ec223 : SUCCESS in 10m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c9b94585947b4e789bba94499d447797 : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3119b87a697c40d1bb24f4ad4021a01b : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/faf0f253342b47f9ab62d3468687867d : SUCCESS in 9m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/76205f91fbf24080961b29d705e3b94b : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06a0f516e2da4a8b8d5441822c5377cd : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7f5a39e031f7442981f2aa7c9ae2fc9c : FAILURE in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2130341836b445228d0bab19167da298 : SUCCESS in 9m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eb87d3c264bb44f3829a618e5faf5741 : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b99a31c054ef4a24962774fdee617d5d : FAILURE in 22m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cc1425330d0247bda705bbfd669e257f : SUCCESS in 5m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d7786602a9764a22a41cc0dca9802ee9 : SUCCESS in 10m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/253777c50fbf4978a6649d3197e41a4f : SUCCESS in 28m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/055f7fc0aebf4ebc87aa5744ac9d28ff : SUCCESS in 12m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8082792880a749778f1e19f025443b8c : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd6a02be94354b7291ee9c11574ca4db : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88f63619ac8d409fad4eb1983eee4b70 : SUCCESS in 22m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2e3bf5ab89aa47949f3600b2185cf64c : SUCCESS in 6m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fe7054f769c340e3a8a8d3baa59ae30f : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7af7a089c1724e94b6976d2ca5cc43e5 : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/29ee1308df88493fa4313774520e587a : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aab3f0a0a0d94a47b5eb59e3a54f5f10 : SUCCESS in 1h 04m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74a4574c85204d409a94c3029f73fe5e : SUCCESS in 1h 29m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b67a7934fe1cdb0afc5fa01c84efd003756e3fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-14 16:54:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8259ac6c51e81f639732a8ed7aa606e32a552461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 17:14:43.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db7f248bf6cc41e6b1518f120e5918e9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8cc4fc60c5c648fbba35e3c742e7ad98 : FAILURE in 17m 26s","accounts_in_message":[],"_revision_number":4},{"id":"08e50763e1a4825b947c71f5d4e29ab8b5f77a06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-14 19:22:10.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"0e87a79631a7f625384d550114980451693a2d95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:27:14.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fdffb146c704170ad919d79342cd206\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/11a316bd88904e9796641de11174961e : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":5},{"id":"f57260a7f812ccf2761dd4b532ed46609c5a31a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-14 19:46:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"86006ee614fc9d7054cb9cd389f01a80ebd756d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:52:41.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02475a6d0e9448c1b8ffc7dd12a558af\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ca274e2e9cca46fdb95179b623789d56 : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":6},{"id":"609f916ae5291d16969a8577fed71dedb8e97604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 21:33:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e7bdfdbb72a4771a8ba666faeaefec2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73be0fbba4134db6b98a43b2fbce8d3c : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2239e29ec2d04ab09a4edf5507d132b7 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0557e2ec179a446aac3b66c06c3bf376 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d71d28674d447dfa7f4f275057d845d : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef1ef2eaea8e47d5b26bac8ad7ac7794 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/142bf9abb59c40b089a2ee0af1f3d6e6 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b2e4d0a8943342d0940f029d25b5fbaa : SUCCESS in 9m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8a1a842997eb48c79ea2bc9a8b54c92d : SUCCESS in 12m 22s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9bb533e876be4a46940869d300a394d1 : SUCCESS in 7m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1d04633f00914f62a2fcb3c2c9f41d6b : SUCCESS in 5m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3e8b9a0d8dab456ab61cedfca8f13e58 : SUCCESS in 5m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2766316d36d9441abb17293ba36942bb : SUCCESS in 11m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/07690d8865ff49c0909a2e0f1799d3df : SUCCESS in 5m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/157de35c88714a53a5b57de8eeaf3ea4 : SUCCESS in 5m 35s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e125ebd3b41a402f925e0a074cba717f : SUCCESS in 7m 16s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3b79e0c4436b408794c7ac858cad47d3 : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40b9b8e78b1d42a0a5cbc6d44d5cf222 : SUCCESS in 12m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/acbf45b70b32418c90585502f08b9901 : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/89f8639d8d8f4757914f2923a6540352 : SUCCESS in 12m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b7b0b86a25ce4b598737ee51dcfe92e9 : SUCCESS in 4m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f619d842b9d94544833871daaf1b5e4d : SUCCESS in 9m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/72ce0618192c4bdea7492ff33f9c3612 : SUCCESS in 5m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cba97a2da67a4fd29f0781e185796b76 : SUCCESS in 9m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1a004a9c778a4dd886f7fa4e319a4ab6 : SUCCESS in 27m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1ef6a3675da74613862d5cf2530705da : SUCCESS in 13m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7f63c100c96749e2bb1bfae7d987a649 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1dcb5c30c5b94ab49d741b61f32bac54 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e4d80c9c3d1e40d09fe4d0570985966c : SUCCESS in 25m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0cccd2b2a55240ddaf5f92333f6fd164 : SUCCESS in 6m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dfd5810649e04aad863135cca85647fb : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2cce84db702743bb95d405ab81cc171d : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1378b0801eee4d26bde8025ebe49cf3d : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bf746350d6a4cc8a9f8d1902344e2e8 : SUCCESS in 1h 01m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/243eac587b7b407fa98112c25d0ce315 : SUCCESS in 1h 43m 27s","accounts_in_message":[],"_revision_number":6},{"id":"f70578e6bca6b539168a446bbe4980a2389b180d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 11:35:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e05ce5bf2d0021dbd36fbe563592513f7b469b42","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 11:37:46.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"38ee024d0fdd4a56fc7f02b090574a14bf464c2c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-16 22:08:42.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"35dea5bd5ce64b28d37deaa755eec845df9aff3d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 00:56:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"757eb2e20504c3ed5e25678d585925ebe2299484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 00:57:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"578d49bb3e3dc29e21ef5ce5b0403a4e1b741da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:08:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4c5a1efee444125b70d2c348790bd91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411efa234d424dd6830e1c92e29f4324 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d53666aae6a4c05922b6675a7656f9e : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38f59693f3147ee9a1c7d6a9d494621 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcf8e1b2ff5e4f45aee9c40c4b403542 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0efc49e39b2a4c72a2a64f0c594c5644 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34e1ae97511442e09e9babecc1523818 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3a0fdb32b7704fcd97aa5ff7468bd1e4 : SUCCESS in 10m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/87b8a66f65ef4c37b683f6cacdfb1f4f : SUCCESS in 15m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/866a6d103dde49049794420257ad9c06 : SUCCESS in 9m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/16817dca43104d069dabc7b3aad7c729 : SUCCESS in 5m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d6a9c1abddb14d1fb78d239c91fd7a4f : SUCCESS in 6m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/841d33897fbb4ce2bed15a97c5ac2330 : SUCCESS in 8m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fd9f7aa9605a40568916cf566921e1c6 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/30ed6b8ad26343cdb7c121a40f536df5 : SUCCESS in 7m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/3736ae9241f0428da1047a7a9f29381d : SUCCESS in 10m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/99336a3fcddb496dbab9b2775046768c : SUCCESS in 10m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/51ef996890244308a5bc5c8ce769b3f5 : SUCCESS in 4m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0378ca7c6540438da43389c1d7a0138c : SUCCESS in 9m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/99d9b932a7424af6b57569071b140036 : SUCCESS in 4m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/88ce12991c1b474290968f482aaf82d3 : SUCCESS in 24m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2d9326ead7274cc38b08b87dbc60784a : SUCCESS in 12m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6bcb498e8701437380921f8a51ae435c : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4df2c78f662241f9b0a0b64bf0b9e049 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/278a35d224374d81a5afffa711be8000 : SUCCESS in 18m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7da461ce83ae4d59a66f7394cc669325 : SUCCESS in 5m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ebaa3e5249ff41ae943fff113ae5185c : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0df588cf4ac541589766dce7f988bbc5 : SUCCESS in 5m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a9e18799b80a485393da1bd673e6c1e5 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4efcf48c3594efb8db5b3012bd6364f : SUCCESS in 58m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6368a8969aec47449b1d3e4552d8dc54 : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":6},{"id":"755f8173ef27b3c40ff0cf2c55d072d3dcf27eab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:08:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"38026ded8095fd9e0fed8998fae71b819244ba74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:15:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef8347d56dab4f25be400cc5261d4f7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d6e95dd11344a7a47243deffaf582b : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"841a8e09008a5228979bf7cdf506eb6e14ed58f6","revisions":{"9f3343c67e4bc18e28f60ebfbe3d25c325a5d491":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:09:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/873386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/873386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/873386/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:31.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.2.1","message":"update constraint for ovsdbapp to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc58556f0a78c8859d4f6ceccb4ae32ce97dac82\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia2df0e26b7e08bcb65904f9fa506601108894511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f3343c67e4bc18e28f60ebfbe3d25c325a5d491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f3343c67e4bc18e28f60ebfbe3d25c325a5d491"}]},"branch":"refs/heads/master"},"0b60e49560bc4e0c41b1d56f4ef6894baf435cb0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-13 17:59:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/86/873386/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/873386/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/873386/2"}}},"commit":{"parents":[{"commit":"28248c2d01151b10c7dca8e373ca621228c57a86","subject":"Workaround for unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28248c2d01151b10c7dca8e373ca621228c57a86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:31.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-02-13 17:59:15.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.2.1","message":"update constraint for ovsdbapp to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc58556f0a78c8859d4f6ceccb4ae32ce97dac82\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia2df0e26b7e08bcb65904f9fa506601108894511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b60e49560bc4e0c41b1d56f4ef6894baf435cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b60e49560bc4e0c41b1d56f4ef6894baf435cb0"}]},"branch":"refs/heads/master","description":"Rebase"},"1648295a9ec5e3f5c82ffee79c1b909ed1aa97b3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-14 11:31:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/873386/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/873386/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/873386/3"}}},"commit":{"parents":[{"commit":"69bafd57c346c141790884758b3ba28acb151349","subject":"DNM: debug patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69bafd57c346c141790884758b3ba28acb151349"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:31.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-14 11:30:57.000000000","tz":60},"subject":"update constraint for ovsdbapp to new release 2.2.1","message":"update constraint for ovsdbapp to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc58556f0a78c8859d4f6ceccb4ae32ce97dac82\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia2df0e26b7e08bcb65904f9fa506601108894511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1648295a9ec5e3f5c82ffee79c1b909ed1aa97b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1648295a9ec5e3f5c82ffee79c1b909ed1aa97b3"}]},"branch":"refs/heads/master"},"1403a95fa972e925bdb065d5cf0b8fa9bb202347":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-14 16:54:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/873386/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/873386/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/873386/4"}}},"commit":{"parents":[{"commit":"697f969ddb285efae6588c6f7e2b4d857978e07a","subject":"Cap virtualenv (\u003c20.19.0) to unblock gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/697f969ddb285efae6588c6f7e2b4d857978e07a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:31.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-14 16:53:39.000000000","tz":60},"subject":"update constraint for ovsdbapp to new release 2.2.1","message":"update constraint for ovsdbapp to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc58556f0a78c8859d4f6ceccb4ae32ce97dac82\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia2df0e26b7e08bcb65904f9fa506601108894511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1403a95fa972e925bdb065d5cf0b8fa9bb202347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1403a95fa972e925bdb065d5cf0b8fa9bb202347"}]},"branch":"refs/heads/master"},"4de796d0d52cce9e3d6e38c3f23b35fd9e06903d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-14 19:22:10.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/873386/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/873386/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/873386/5"}}},"commit":{"parents":[{"commit":"645f08bbb6fccd8028938c72309f4b9bab8c7a2c","subject":"Cap virtualenv (\u003c20.19.0) to unblock gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/645f08bbb6fccd8028938c72309f4b9bab8c7a2c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:31.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-14 19:21:44.000000000","tz":60},"subject":"update constraint for ovsdbapp to new release 2.2.1","message":"update constraint for ovsdbapp to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc58556f0a78c8859d4f6ceccb4ae32ce97dac82\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia2df0e26b7e08bcb65904f9fa506601108894511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4de796d0d52cce9e3d6e38c3f23b35fd9e06903d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4de796d0d52cce9e3d6e38c3f23b35fd9e06903d"}]},"branch":"refs/heads/master"},"841a8e09008a5228979bf7cdf506eb6e14ed58f6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-02-14 19:46:45.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/873386/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/873386/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/873386/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/873386/6"}}},"commit":{"parents":[{"commit":"7c3cb297573b5dbec634e53829ec183ed7ba3068","subject":"Cap virtualenv (\u003c20.18.0) to unblock gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c3cb297573b5dbec634e53829ec183ed7ba3068"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:09:31.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-14 19:46:26.000000000","tz":60},"subject":"update constraint for ovsdbapp to new release 2.2.1","message":"update constraint for ovsdbapp to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc58556f0a78c8859d4f6ceccb4ae32ce97dac82\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia2df0e26b7e08bcb65904f9fa506601108894511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/841a8e09008a5228979bf7cdf506eb6e14ed58f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/841a8e09008a5228979bf7cdf506eb6e14ed58f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
