)]}'
{"id":"openstack%2Frequirements~873390","triplet_id":"openstack%2Frequirements~master~I70b0996de61b9cbdefaf9b33ea8b5604899e0e62","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-03-26 09:12:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I70b0996de61b9cbdefaf9b33ea8b5604899e0e62","subject":"update constraint for oslo.log to new release 5.1.0","status":"ABANDONED","created":"2023-02-10 16:30:42.000000000","updated":"2023-03-26 09:12:40.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"434420a0bfd0268a505560e7283d79b1da36873b","_number":873390,"virtual_id_number":873390,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-17 04:42:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2023-03-17 14:26:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:13:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:04:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-17 08:25:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-17 09:44:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"52f209142ce06bdbed8c1f0ef63a1a206cc4033b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:30:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84e42f450d3b0108436e48cfdb2ea163a94bead3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:13:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc049cd807254607bae4ce887a692563\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/699c95eb0d4846baa253665d5a5b43b3 : FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1e0bcdccdf9a609eb17c9935c38dd9dfef22908d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:04:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b521ce2161cb45eca8cb9b87524321f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46c9a8460bf4c35a5f87ccfeedb6a13 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ceb76eede14bf9972cbfc2724af2a2 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38c806380ced4a8da997104b153cd6ef : FAILURE in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff7d71f059d6486897d2661d0b672111 : FAILURE in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73db6bc1dd4c4b6c8000c1bb569129a2 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41e138eb8fe243dd93c52f99f9d60c3b : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/668a7b4a7ec24164803ae93882e15107 : FAILURE in 7m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3f8db190ceeb4ac0a9b186461cadc2c8 : FAILURE in 13m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/54abe0d6745d4d0d843133c66ca87cb8 : SUCCESS in 7m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/98e06e0ef1eb41cd837dc659bc28fdea : SUCCESS in 5m 46s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1a01d9feecc843c4a8699e475632f40e : SUCCESS in 5m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/755300357f1a43f6a39708bf3231ef77 : SUCCESS in 12m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d55ab867987f4b218e401cb983fce6c3 : SUCCESS in 4m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a0a0190894254973aa6099c629beda15 : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6daeb333fc5d470e94157dce0ff99195 : SUCCESS in 7m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0491a3b619e84300bdb4a62b0c9cdda8 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff75fc521a66461d9c1b5d930d60fe3d : SUCCESS in 7m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/39515c5e73c04783b36e4d209280faed : SUCCESS in 9m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/985bbef475cf499d8e806b514c01645a : SUCCESS in 11m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fd8e06382dd34994975880e79d6c28cb : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7373cf12dfae417aaa6a6d5276ae1716 : SUCCESS in 8m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0e66fb2b9df048ea94ef4bcfa4c06019 : SUCCESS in 4m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7a7c5ee3a0a949d885072d57f682bf90 : SUCCESS in 10m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/227b9971f88e4375b44de4081a00b63a : SUCCESS in 26m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8e5eafe0774c45e7b5afdd1b8ea62e8d : FAILURE in 15m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/92a2297c2652499396d980e61a6d565c : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca33e0d45bbd4edcb785bf911f5ce776 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c19ad744af64d1ea86e9f9eee645184 : SUCCESS in 22m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9abb6fd18d7841728e1e75fe3c7fdf2e : SUCCESS in 5m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/42295ea3cefe4eb189e833f776a10596 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e80fcaa9d434552a838f7646b73e4fd : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/92f91de498d94b5d985344c9c0da2f1b : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a180e1377d74a769030fe5238c8f3d6 : SUCCESS in 1h 03m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1dcff1edc95b4da799731b2700920f95 : SUCCESS in 1h 52m 55s","accounts_in_message":[],"_revision_number":1},{"id":"aae80dbe91e7c466ce53cfe025aca38118e58705","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3057110f9e407d7bec25afc04aef0d2440d6732a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:01:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3f320ddd394462a946522f06f67d550\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fdf79bdf8faa487d8800736456a87abc : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bf1c4d024e1f06feeb2e6c868ec609c41b029060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:39:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59b228cb67d643029a5dc090374b1abf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78424414c7749daab1e4ea8209a130b : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269d115376634ea6a09f61ea59cafc9e : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9220df00ab4f4e8c858607804171308a : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4d44c7dcd474baabde156577c491b88 : FAILURE in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/515f758027024850b6cad731c013a95c : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c845305e1eab4a8bb358e223f6b4ac10 : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdf0c71d86cb4c129a899501db33b26f : FAILURE in 10m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38f7f04349d24bcfbbcec431fa5fa403 : FAILURE in 14m 24s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1f9afdd67d5e4b35964dfcb9ecaefe31 : SUCCESS in 8m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a72c3d2b1be440418f28b82faa38972c : SUCCESS in 5m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/64883eb98ba047e3928437d43f4662a2 : SUCCESS in 5m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/628ca2fe3ebd4523989351d2074df9ed : SUCCESS in 18m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bfc5f8bc78b148beab0d642a5d548c40 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dcbdcfd36aaf44e392434bffadb0dc98 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b5ed26a540c5404492775cdfea0a00b4 : SUCCESS in 6m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/acbbf513dbf44173878e97bebc922a34 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a4f5805fbc246b2b6841f965736d0b3 : SUCCESS in 40m 43s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ccfb4789e2bf468e827b77ff2c6f9e2a : SUCCESS in 10m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/30fc7189dd024012a216fffbdf0e86db : SUCCESS in 18m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/68a377385e5d4ed49f4e2ca0c420dd55 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/98c088ee1b674defb1bfbf0bc1482287 : SUCCESS in 8m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ef337677e8344797aed458b45c127cf1 : SUCCESS in 4m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a68fe7f88ddf40febccfbda4fa753db1 : SUCCESS in 10m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9be7c1e87a974fa39f162d533362bc94 : SUCCESS in 52m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4f5cd06b716e476fa7fcb8380343eab1 : FAILURE in 13m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/16f41a968bad4fdbb4410a4039185f13 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/095b5b1ac4854305a85a68c59c16db96 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79633f8febb44ed6a2046bce3d691c0d : SUCCESS in 22m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9b37edc96cca48b282bef47baeb06cc5 : SUCCESS in 6m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/179f88ca5178482bb832a106cadf4d1d : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/571a461ecd644633b33dd269db271713 : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c0618f23b21d432a80cfbeb8807f5334 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fe2a263d1d54125b6dbc34969dd6de7 : SUCCESS in 53m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac706711e6f04f0099ac1f5892200725 : SUCCESS in 1h 40m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9c755855d90b9385e8cd1906925812de916a66c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"611a1b66253f3774985199b39af728e1408815e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:04:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5908a7e23e43868c0dfb1a31d0f49f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7b93056a54ac425cbca07ba62b34cd8b : FAILURE in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a806a744ca5d0968ff1d7b68c08f64767121747d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:26:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ffe59c0f0564c27969291d80a1088b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39435834066946d587a8524952a1ed3f : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8598fed4d1048d5bcaeef217d1f3388 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8597abc745574aa4af7eea8f46f40cf0 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85ad47b5e7bf4c53a2cc70555b6db22b : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d965234fa3fe4566829f566e1ed39400 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc60a19d9bd1416c8b6c217acaefef1a : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e713c6812a94d8d9e9d85e417dedc73 : FAILURE in 9m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c0e0bf5658c542eeb8df4d31ec0b1a7f : FAILURE in 11m 08s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ede5149b60e2499cad8539fd55bde35f : SUCCESS in 6m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e7d5327b46f645388a610d7eb9349404 : SUCCESS in 5m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6d371a5e1a384740b9eecafa1ff9f4a2 : SUCCESS in 6m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d1f0bbfbc5b0411a9ba949a3d52f9d8c : SUCCESS in 9m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/58373f19d9674d10a1543e5621f21824 : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e564e480e132472995f4d3b3fe26b58d : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e2b19529574648758a9a9609ed02b724 : SUCCESS in 7m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/17d5c2ff2d794a55824d74192a1bd672 : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c382d1e83e8f4867abdb69347bb3946b : SUCCESS in 11m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b4755aea2edb4195bce4076d854b243a : SUCCESS in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e185957f2bf340c7bb683291a8471993 : SUCCESS in 11m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/46a67bc98441405c8ffbcc9547efc763 : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8eebf39722124fa4ab297335f8d822f8 : SUCCESS in 8m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f37d63f1471d416fa673b4c854776a3f : SUCCESS in 6m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/855d6aa7a3574271a484c9e56ed5ab9d : SUCCESS in 9m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/90a1f681b30a49aa9ca72f3ce65f860a : SUCCESS in 25m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/05c036e90ef84e5885a91909da7dd829 : FAILURE in 16m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e808b923da564039ae36467b429403ad : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc5bc9a4a5dd46bd81d309b21ae09d99 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f054eee18b774a80bb50d050eef07f34 : SUCCESS in 25m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/eb7948d4822049cc92c747055d7a2de6 : SUCCESS in 7m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4976b257af2b449f880342e6504e5dc8 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c173038da5044007b2957b0db1103f5f : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/10d3df7ba9f74e20a07f42e80e0890a0 : SUCCESS in 25m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e0e742ccc774e198248a401adf7c1e7 : SUCCESS in 52m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dcba4fdf3f24c7ca7012189d896a660 : SUCCESS in 1h 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c5bf98d300b3429d1a8ec5864c3006ee5806c86d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a13ac1d8743eaa6c438bd39b14c25d4163fb08d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:28:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/588c59b60b15456aad53c1f2b3e9ae06\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ed0f5ada11d24b5bb937714040e18b93 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9c373d17963b807c4b3282defcaa432074a14f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 05:12:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/796a6c87ca624494984fd0a71b2c70e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a91bcf4c674220824a9caa1c54a303 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d61b456d8314164bba686d7716ef59b : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94b5be27980d41e388a970b0f89cc578 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27df08fa3a8548b89df4c4e5d6baab0c : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f717214b623245249955df7fedce9029 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/681323eb916c4da18c3787cee6a50bf6 : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/000c6b17c83a4df0b367c426fe8d9f5a : SUCCESS in 8m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b01f93ad1f774da5b6993792409fdbf7 : SUCCESS in 13m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c7cd8ea99d364919a390f5095b243372 : SUCCESS in 7m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4e49aaf0967946fda93aaa9583e384d1 : SUCCESS in 5m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/37e101b0a51843d5952e230dd4fed114 : SUCCESS in 7m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ecef0e503e9d42798cd505d81323652a : SUCCESS in 10m 55s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2b5e1f2bcb184c699f8536d081089a35 : SUCCESS in 5m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b754adc4460d48749696f4b3bfe26e93 : SUCCESS in 4m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/582b73ec192f42b2b9adc1efc056c0a5 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/77620b58e2ca4b3aa0ca31e7d2020ede : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b30fac5d21f34800ae8202644d054338 : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5fdb8512a21a40a18f79d27f0dc0b9e2 : SUCCESS in 10m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/25075aa751e6490a8260a07ce0f96285 : SUCCESS in 13m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4f34f19246c342d19cb6d7b622a21106 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3dd13368682e48f9be905a907862ba8c : SUCCESS in 8m 55s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ef1f0cefca434ce2b2b6081e089ade08 : SUCCESS in 7m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c59330ef184a4aeba5a3a5eecd835dd9 : SUCCESS in 9m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ecfc2a8f82444f2bbde17a2292d837c2 : SUCCESS in 25m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bde77bc6af2f4306aff515d663a316ff : FAILURE in 15m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0c905770cc0041c6828b30cbe6c1feb7 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48233affad4e4ab7a5dab9d1ab60e497 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/660c4d9c25734b538109a4bb063cafe9 : SUCCESS in 27m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/45eb3649bf934a6289d7e7d9c6879fc4 : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/032843f477b9491eb12782622413b116 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3ea748867fe456c98ab252bc7eb9af1 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/73129c5dd965489f92f3d87c667387a2 : SUCCESS in 16m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07178583ac5f4240bca077f04a70a816 : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f03c4c1e221342faaa099dfc1d807c97 : SUCCESS in 1h 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"95ee314f84f9059b4dd9fd3b0a5c72138a024a5e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 21:27:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"efb2fa18894d25625ec14992dd6d8a842ef46cb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:33:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3464d087223444d9b4851731cd49f3b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9b35d853ba834f098f4256b9e04f2c83 : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c8de3791aee4b92960f9707323d2e3ea2b42622d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:50:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e76c5dbd94ca4f0f838b6593845db440\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d45747cc9646058f6bf40e41984ab9 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42ba96d566644dbbf248f40e6330353 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfac9ad918dc4d7a8639870e03831df5 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc304978d16849e3ba93a2489e1f6a2f : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0f1941eecf148ea9bfe2e4f5b190f44 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d983b87fb929481993a91eac295142d5 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb92645cc3324ec1b1d4fc3b08be7d28 : SUCCESS in 11m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f03c1d4400ec4945b3e0b22f62879327 : SUCCESS in 11m 38s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/508a9134ef414af0ac2c2637457bf7dc : SUCCESS in 7m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1e0a3ff5520c4c20ad2f0041f31d7af7 : SUCCESS in 6m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1351c361ee364ee884375cf939143e6f : TIMED_OUT in 41m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/19bdaf9a957842d98810458283dc5834 : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1287dbde931a410d9e3c7f2738e3f07d : SUCCESS in 4m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/19237ac3429148cda520827bac5e39b8 : SUCCESS in 5m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5efa467596fe465c924ecb63bc024f38 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7a90485493ed4cf5919b241373a0f43a : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c71a720478174e3da0f2a7d1a194379a : SUCCESS in 8m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/39081d1a4fac42e9bdd002f6dc7f32dc : SUCCESS in 10m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b49c20dc99ca4efe83583f357ff2f69e : SUCCESS in 10m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/78fb7c1da8ba41f9a4fbdc3deb31cfe4 : SUCCESS in 3m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/00d4869499e04e9dac096a8741d5147e : SUCCESS in 10m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ad035a5c988a46018204aa3446ca3b24 : SUCCESS in 5m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d60679e2329e4ca9b9b23a5a3b96b2e8 : SUCCESS in 10m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7d42aad661234815b17e383e64cf7557 : SUCCESS in 24m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d11fc7fb69334a59921872640684946e : FAILURE in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/17dc648237ea4721a79b215b6b23a21b : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93ed512a2d84478c9d7588ab1ba7bfe9 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb461fa4a42a4fda8ec3e5fd59c5142c : SUCCESS in 24m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6a74d160a1944dd68ee13bd78b59d46d : SUCCESS in 7m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dbd8de45dc9c40089f7dd2355e967f51 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3d72031fd64432ea5b882cfb346dfe4 : SUCCESS in 7m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/adb73d8c65a446a88541f5d2d47e51cf : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/406c1ba3abfb4205afd27d80915a0fea : SUCCESS in 53m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b235ce29f5d4adbb7cf130c6d2a54e1 : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ce71b4850e9d29e4f01b819ca72ef31a9cc4e709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:42:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ed9fd7feb3d209c0995ba644bddfbf990ee8dc49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:58:01.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea5665554b749e2bb25ef018a9ff76e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1bbfd1c9a07b40caa2a5d5a39576e813 : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":2},{"id":"709b79f0ca293250aa1b24fbe9adb6b09792292d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:42:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e6b0c6907144299b6d5d383e89bcd0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ead9c61d54475a9604fcb49e6a84a4 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e01f215c9f49cc984a3bc82dfa9f9d : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7f1e4fe267848dbb87bb5d5bf122995 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc3d47e841ee48f49918d56aa10804d8 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be943eb32f1e4f2485041ee8465d2a3c : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7b90bd0986840aaa0b8924b3b74a0c0 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c4d43c44dbd648d49cc59375b9d767d6 : SUCCESS in 8m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f9b9c49b3c884c6992970e3ec4ad9bf8 : SUCCESS in 13m 10s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c4d6165871c54704a09aa2b902124ade : SUCCESS in 8m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/edf6e7a19f984ae6af063a0e5cecc4f1 : SUCCESS in 6m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3817fcbf879a4313a3e1f96019edb07e : SUCCESS in 5m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8e6ad86d7808479fada506fb32080175 : SUCCESS in 10m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8427ed20952a4e3db722a38fe6a82491 : SUCCESS in 4m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/52837c561fe7461fae10e0960ded22b1 : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/122e58ea3d244d948e3a2a3b96c95dd5 : SUCCESS in 7m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8c3b500464a34faa9abfb066a187cb38 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fed08a17b37543e19172a0c8f1ef7f68 : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/193500b5f4e8451aa644181ee987e6e9 : SUCCESS in 8m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/daea9e9489be4ef0a77aa797543cc0ed : SUCCESS in 17m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f82de5c2ff824b67b82b44c9f526d6d7 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/dbc923d06e064fb686cae5cb38ddecc1 : SUCCESS in 15m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/979d1250307f4cdf92d55f2cdd1fae93 : SUCCESS in 5m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/724257e835b240b18c5e7b6e7c40f9e5 : SUCCESS in 10m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2bc949bb196f413abaeba61630ca5808 : SUCCESS in 25m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/29ac120acbf1451b8bbf5769c872db40 : FAILURE in 14m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c86e18cbc080434386b9055c9c832e13 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a426b7e1eb3f4e3a8ede8500bad7061a : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/261fa68d29094d2bb269bf62e8f51556 : SUCCESS in 22m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/528d14255d9b415ca5b37b6eecfbead1 : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c82be6e09652487e87cdeda0a1bd5c03 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75212ecac0274b8f8644f1b1ba8b2d9f : SUCCESS in 6m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bc5e238358f84070b97ec53483af5d89 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35fece86f9d24d188e837512dd08f085 : SUCCESS in 1h 31m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ee8cba3acb947b98a21ade3f282a6bd : SUCCESS in 1h 32m 41s","accounts_in_message":[],"_revision_number":2},{"id":"85cc0e86bc9a912bfaa9ebc11b9a7e8776f1babf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 08:25:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"12416c70ef94655070de7888821ac926ef201093","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:41:35.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2654762e5a848e6bfa4c9a5d7b49a1f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cdd3042aa8f045d2bdbd3011da1955ec : FAILURE in 13m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e4ef7475ff9242d025cced0b8cbfa658cbd05c8a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:44:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d63ad9129ceca5d48b7f999b779331c30f353c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:11:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f58324ec31da4ac8be8aaf4d7ea22f0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d05065a36146f59355a96fb5a13b66 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67399fad3b2f44f1a97a6b7c5b2ae05c : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d01c832c8474521945177c316951fab : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9f61c673be84a90bee751922e4b88d9 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32dc99fc7a2e4608b17f83f9873d6141 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ec987a261074b77a4c5190372074601 : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5b865ae360842b5abd3690eccd073f2 : SUCCESS in 10m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/272035390b174d66ab5803dce4d2aaab : SUCCESS in 12m 03s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8ab0149f4b354e4aa73820086f0df55a : SUCCESS in 6m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/dcb5f76c747c4485baa07b3918c2103b : SUCCESS in 6m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/37b67504ec414bf6b6f303fdc5a60173 : SUCCESS in 6m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ed8f42301782494db5af5e5b6d3bc459 : SUCCESS in 13m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/32819c4b8ece4a449f792b42aa2a18ea : SUCCESS in 4m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/060face20510432984af396ccc1e0bc1 : SUCCESS in 6m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6a6b137ce64c4427ab0e33019753b06d : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/11b14f51890d4e9092ac3cf3d99ac188 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a07425213144830a54303586d940843 : SUCCESS in 14m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/99f9887101fd4983b08329386d21acf9 : SUCCESS in 8m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e42065e3ca874620b5d03016ab30a8cf : SUCCESS in 11m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/24ddb52b4749425dbdb79d8418ff5ecd : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/51faf148507c4c1ea9bcee9ac07a12a8 : SUCCESS in 11m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/477476d2f76d453da3ebdd17d5d31a88 : SUCCESS in 5m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/13632a854b5449cc93777e0d54d56068 : SUCCESS in 9m 41s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b7b7689f22bf40a6b92553a06b736002 : SUCCESS in 23m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/63225b77498b417b87987a1688088ad4 : FAILURE in 13m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ddfb020a593c41c3ad22cb22d8df36d3 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d34e83cd620243ed9a2c2a97b0ee8049 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/222aae0664a24f5a8ac490104977eac5 : SUCCESS in 21m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7db7b999385749b6aa8b2c6ef891b077 : SUCCESS in 5m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/de40a7afac9e4f489c2585277373c4a7 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e032616bf4846e39cf838a8133707d4 : SUCCESS in 5m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/671b7d15cb294b9881073a1d0ba8e4ce : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/753ba2dd29574431a8179cf710e9e1b8 : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5287b499dde847de9f1187360ac0a7b5 : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a4e96bd5b94a253684b986a6915c5c982edcbac1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-22 00:36:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e1861c114574f413887418fd66d8298194a170e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 00:53:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef1816b467cf4f2582fdab50f95efd4d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1e3eb978d35b46e8843a7c7f31ea0aad : FAILURE in 16m 22s","accounts_in_message":[],"_revision_number":2},{"id":"029cb315ad06255bb527933df74bc026b6ffec62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 02:17:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f874be02d6d4e068ae9f81f408c4769\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8107fe54c524c7095aff1dbdb469993 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae9400680c749b39da2c41abf17ee26 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c45ea2b04a44c70b739f5d9a62dcc2a : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ebb936550fa458eaf60cfc3e9476920 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c91317f0c5604c269734be4c241b048c : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c4896768149482e8accf8f4eb25538b : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46e9f90fd6ac4c4a8c99c1c6d3967817 : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/adf67adb45f5406badf2f199a4b1e726 : SUCCESS in 12m 26s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/55b2a2387e814901a82efba9cf36da54 : SUCCESS in 5m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0fbab27fb5c24512a1bd680be85818bd : SUCCESS in 7m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/068e82b7427b468588aa7160b45135ce : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1f9f436fe466462f969ac7405905eea9 : SUCCESS in 10m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/43f737b9e19949a1925b456436e7e15f : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/85b3e3a6a3934b8bb076bd1bd9511d88 : SUCCESS in 5m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/da08ae8871f54359b7d0d187e77da9c8 : SUCCESS in 7m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1c463442ccd44ef88739f27ff5663ddb : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e07c08f731f8419e8950227244b38f67 : SUCCESS in 7m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/31dc3b310f64400983a88dabeeed1e78 : SUCCESS in 9m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/231a6a3173e743dd9ead6a3dd9520b45 : SUCCESS in 11m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d91ae4be160b459d8a0b62d823345b1b : SUCCESS in 3m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f1434adac289443690edc51b014ddeaa : SUCCESS in 13m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/00173c1f190645f9897bca31d4ff0a6d : SUCCESS in 4m 39s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/40dcb58140e148fbb75e733d682bdc1e : SUCCESS in 11m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/766a2663bb7c4903a37d0387884f9042 : SUCCESS in 21m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3853d26f04ae4484af18fc5178f88442 : FAILURE in 16m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/48048d7820d042f68c577a6b0fdec2ea : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0856c3bdc354ca7842cd63f2428140d : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/273d13a3fbc2450da865856e772d99b8 : SUCCESS in 23m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b594b1d37daa472d85ebb86db598213c : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fc68dab5c9784944a5176fc090bd0069 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0f1e048ff0a44e4aedbfdb7309dbd2f : SUCCESS in 7m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/62cc0422e9b540a195501c72b62f0794 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/752a6af028bf45c788a4fff3385fafa4 : SUCCESS in 1h 03m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5c0dd8b70224746bbcb163906325295 : SUCCESS in 1h 35m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2968e6958042900ed1519ac93158a4125ee048e6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-06 15:48:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ba85cb94b165dfe403519cb78d1ebf6d833cbb30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 15:53:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c94f1ad36994f7d83b8fbe34a7a2b0c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d504aee49b3d4417a42eb136593bab8f : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"dcaad8665ed9899f8a52485f70ae42e732c6ae21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 16:54:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/596d1631f9764bf6aaaad1affaa0a468\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78fa7ea84ac648de8c181abbbecf60a9 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743c798db46a4a94be31b112cbf21435 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6d10d754dce4228837f8542e4eab011 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f50b5fc285a2418c980b6c0e196b1b75 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f914b09799814cf0bb8e2c67a0fff48b : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/112093e8c82746dabba2dcf80520b211 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78d09c259e1a4a2ba53d2ea18e26c59b : SUCCESS in 7m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/258160bc892f4ebca82a4b2a7d5cf4b4 : SUCCESS in 14m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/78647585577c491aaca9d7321be50edc : SUCCESS in 7m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/38d082df15914a75b2dd923a39f23acc : SUCCESS in 4m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/429a34334d74461c88c50cbda2944051 : SUCCESS in 4m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a84ea282cb10467c87a6fc02b15f2d00 : SUCCESS in 11m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/349d37c7fcdf41e291233a1384f93a79 : SUCCESS in 4m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5bc02f0798b24773aae4ab43d32077d7 : SUCCESS in 4m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/17d2f8a91247477d8b67cc9e826569fd : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6e45f38f3a214dc994e27a4860062b25 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f35e1a6b7df43c88af4ba9fadcf3113 : SUCCESS in 7m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/48dafa4f42e24b06904e051265e0867d : SUCCESS in 8m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6b3989f46ef74d3a901025f573b90c5a : SUCCESS in 9m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3d3326d726304c5d989f36f4f6a692a3 : SUCCESS in 3m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8f37eead9e6a4768b277d119b9438b6b : SUCCESS in 9m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/33b5498a689a4fa5ba0b604023e5a852 : SUCCESS in 4m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/eefffb7db9ca45f69b61dcd2093f9df2 : SUCCESS in 10m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8b8c9ad79cd74ae8bc54dbcd75d51bd5 : SUCCESS in 22m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/567197cd8735453e93918499255760b6 : FAILURE in 16m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/91b5c755003f46539a30684b935856eb : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8ebdd07cd3949c99dbb8cebc022b818 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00d5bee5033e4d38a1447c5e6703bb1c : SUCCESS in 23m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5eff8c93c42647ec8791ac42dd0b9f3a : SUCCESS in 4m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5adb790b76104bc88d2f425380a06b7f : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb1ad25a62714d21966afbe377af9255 : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/718ec35f8e0043db9e1efdac5e4c348b : SUCCESS in 9m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f669f86d0a944266863db7bd4405fd9d : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21868ef0f0cd4860928de53d55b39866 : FAILURE in 48m 26s","accounts_in_message":[],"_revision_number":2},{"id":"55ffe2134315772484bed2944ccab1ce8fa20621","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-03-10 15:30:24.000000000","message":"Patch Set 2:\n\nrecheck\n\nI expect nova to still fail","accounts_in_message":[],"_revision_number":2},{"id":"b396fe85556f5c94f754a98d5273cfd906587b3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 15:42:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97dad3009ce8426288ed9eda72d1bf2f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/98aa1a49404c4a5dbddfecedc6e8aa9a : FAILURE in 9m 34s","accounts_in_message":[],"_revision_number":2},{"id":"00610321de363e6586b09d73e348b387a51a0897","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 16:21:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"373012a6d6057bfb6e494e4badfd45b7307cee1b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 16:26:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"574bdeef4bc7402a602dd1d815af922a9fb0ebb6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 17:05:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"534af27a02deb2b6dfc0a0d179e26dc861f664b8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 17:10:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58713636a620d78abe79a38611bb76e8de3f4484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 17:46:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ab1a08cd18848df8b16ae44290c3015\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166e9fcde1d042e38cca380360182199 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e284cecc7e147b5b3441ba21dae3696 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/869a1c077376491394d4d55d5d5da902 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bef2b03d4c7648ed87e3962c6fb5a341 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5eb01e3efaa84a6cb46d58a60ee1ec82 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2f9f52112a74afe996c1066001e1bca : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a0755cf5efb4ffaa25c542a853efcec : SUCCESS in 7m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a9fed602901140b1ac9640a16b8d1fd4 : SUCCESS in 20m 49s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bc6942bfbcd64b1d98f9234c17691da7 : SUCCESS in 7m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/415d67fa0b39485a9a39e000ae2d3c58 : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e4fb57fba0764f48b7d8d041abc18f60 : SUCCESS in 9m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/81550b35f2e24f62a1f5c326195c9b00 : SUCCESS in 12m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8dd7eebf2e244ba082e74743326b1d8c : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/09da69a7347045abafda6ac92b4e88af : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a34e7808df9e4e37ae4f39436abd580f : SUCCESS in 8m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/35708fce2e704d7fb852bb63564351f8 : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/745a23c72dfd43b7b2350f312d9fb741 : SUCCESS in 13m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/78970edd1c8f4262a2789987a5582f3d : SUCCESS in 8m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ec517f3d90184d4fa380a7a8703d1365 : SUCCESS in 27m 27s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ba153085d6e2400f884d5eed7da594a1 : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8e067fd5eeb04433b834edd711949b1a : SUCCESS in 8m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5a516a1f9faf42b988cafcfc07de0cc7 : SUCCESS in 4m 34s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ae463ed22784450bb526e47b39938716 : SUCCESS in 9m 40s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3441c231193b432da154adbdd1061941 : SUCCESS in 32m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1319229a02be451bb2f86bdc754a8b7b : FAILURE in 19m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1cce84dca9b1483ab4d51fa9a7bd7d8e : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b40a2a771d63400aa958233fea9698ed : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a07793afa2934ccc9eeb932297bbdd8b : SUCCESS in 22m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7d727dfa50314bc9b658b7e4c2c52028 : SUCCESS in 5m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5af36a5ac3d6440c9e06a11df38e3419 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3bfc9097b394ecf98f21d739461031d : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3c668c5e5c5d49479c1397a30aee0fb4 : SUCCESS in 8m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f729283cb544bcaad8dad72385e558e : SUCCESS in 57m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eefa67da95634c3eaf28021c16df9160 : TIMED_OUT in 2h 05m 28s","accounts_in_message":[],"_revision_number":2},{"id":"bd2f3b6251621d4ee2d29d66334f30e5e8465e1b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-17 14:26:26.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"434420a0bfd0268a505560e7283d79b1da36873b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-26 09:12:40.000000000","message":"Abandoned\n\nsuperceded by 5.2.0 https://review.opendev.org/c/openstack/requirements/+/878130","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f613eb1ca2d817bf1949eefa1b0d03f89791487","revisions":{"fb8561ff7a903f873ac832fc582e1dbd37be92bc":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:30:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/873390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/873390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/873390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/873390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/873390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/873390/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:39.000000000","tz":0},"subject":"update constraint for oslo.log to new release 5.1.0","message":"update constraint for oslo.log to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3498a220518cc96c7e269db2e4f60d8c27de4986\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I70b0996de61b9cbdefaf9b33ea8b5604899e0e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb8561ff7a903f873ac832fc582e1dbd37be92bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb8561ff7a903f873ac832fc582e1dbd37be92bc"}]},"branch":"refs/heads/master"},"0f613eb1ca2d817bf1949eefa1b0d03f89791487":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 02:42:08.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/90/873390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/873390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/873390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/873390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/873390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/873390/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:39.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:42:04.000000000","tz":-360},"subject":"update constraint for oslo.log to new release 5.1.0","message":"update constraint for oslo.log to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3498a220518cc96c7e269db2e4f60d8c27de4986\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I70b0996de61b9cbdefaf9b33ea8b5604899e0e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f613eb1ca2d817bf1949eefa1b0d03f89791487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f613eb1ca2d817bf1949eefa1b0d03f89791487"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
