)]}'
{"id":"openstack%2Frequirements~873391","triplet_id":"openstack%2Frequirements~master~I8118b10e5c6a649a1311cda7c7b8013f0fa7122f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8118b10e5c6a649a1311cda7c7b8013f0fa7122f","subject":"update constraint for oslo.i18n to new release 6.0.0","status":"MERGED","created":"2023-02-10 16:30:58.000000000","updated":"2023-02-16 02:20:20.000000000","submitted":"2023-02-16 02:18:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873391-new-release","meta_rev_id":"941bae7b217ef708e30f344f903db7508654a3e3","_number":873391,"virtual_id_number":873391,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 02:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 00:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 00:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:13:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 20:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 18:09:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-16 00:33:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3de4fcfe7817a2638bac4617b4081f5fd6e13555","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:30:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa3f3c96c57399f8d1ce525d2e504488e7f3d7aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:13:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51073d716ef746ac95fd5c40b94abeba\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/89124940bc92421cae02cf96a74f1d48 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8aa4d630b6cbc6428c323c606b6ec58bc2ccafcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:19:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/552da15fd95a4f59a9ce47b846947df6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b27c4fb7ab44e294f75db19ecf7a9b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ed7320aa8d476ab5a8e1894af0a489 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f8e4a5e91a443ad9e725869d1d31c6b : FAILURE in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cbb606ec0ef4b1ebfad09fc04312455 : FAILURE in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2f944a2ada84673a28874195cb032e6 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c6e390289714aa6853d765b04c91f00 : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a2418e4f808b45d5998cec0d2bc7cd01 : FAILURE in 7m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/af772df8b5314f43b66716a6ad70141b : FAILURE in 13m 50s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7a858f10bb2f422e9573ba66d68f7258 : SUCCESS in 6m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/82fb08fa815d4036987ec24b526c3533 : SUCCESS in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e5232d658b8d46f89377855fbd5d7bab : SUCCESS in 7m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/514b193a655746a6afceff0e2ce6e016 : SUCCESS in 12m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/818f2f9e919c4eebb1b5869d96685886 : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c79f06a8ca784425b60c71936eca6706 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/22c27c8ec50740eb99f8fd13893ac224 : SUCCESS in 7m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/119c0d9c718748f2aebc402e1e0101fb : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd5fc28853064649aea51a461607f726 : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d737357375014893a15a0e5279a93ead : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/afab922237db433abacce11f5741f3be : SUCCESS in 9m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5311d78f75d442dfafc79de4e0403832 : SUCCESS in 4m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/001bf0cf06f64bf395dad9db4011d623 : SUCCESS in 10m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/48a5966d4d0a484ab90c2e352c9ece5d : SUCCESS in 6m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2cebc014b2c54833a464dc7d81f64434 : SUCCESS in 11m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/34d8592749e24408b30b8786d60d92e0 : SUCCESS in 31m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/967adf67dd894186ad213bf23e96d0e1 : SUCCESS in 14m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/104c0e19063848c19ec9ab0a84317f3a : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c47461c405348ef95dbec89ee7ebc96 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0be6a0ab96340359c15753634f5aeea : SUCCESS in 23m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e0904a6f4a364219a8b4ad3adbf61ea4 : SUCCESS in 5m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/300089dc1086440a89a944546143215b : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d860872a7bc4503853a87c6d39a4fb7 : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f5b5201e2fdf4fb289736dbcaf21aef3 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d34048759e594ad5afae643cca1506ac : SUCCESS in 1h 35m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce6c4f8d016f436fabf99a1983279e07 : SUCCESS in 1h 43m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bd76b2159d88154173ee1483c4c784da5cb63775","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b3c69ca8eabee88b98e01f5b49dbedbecab0d06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:02:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3934a4e2b5454e9193e2cfc1f93c2ba2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9a6b805cdcd741f486a40c1e2cdd3969 : FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"24bb3b98018f5429762dbc9be0a6a8532a788c3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:56:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62718116a19c4aa58a46b33fc2816133\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760940ee4dff4272a18c1c031cce36bd : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e38b9b8499b405ea052200c28a37154 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6703aeaa8a2480a96deaac7ec5c0b9c : FAILURE in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4918b129f63d4b7c940e30590f708630 : FAILURE in 3m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee0f1788e73744edac2ee5868d2fbc9f : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fb141bf74384786856428f5667b5a27 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9725c6514fca4f228f7c8159fd9ee429 : FAILURE in 12m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/67d88d3e9ea243f595285917b4154e6f : FAILURE in 14m 44s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f80035e9f0fa481dae923e8325a0c922 : SUCCESS in 7m 19s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/880565cacd2e4834a1dd20d995783775 : SUCCESS in 5m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/38224f93e1cb44cebf46568d63559fb9 : SUCCESS in 9m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/27db5a95ed574c2387417fcc54ab2e50 : SUCCESS in 13m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c0dbb23cf385470fa960249c33e3e7a4 : SUCCESS in 4m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5f430fe49552454abe3a77b366920e7b : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2e236c861dee461cb319b02b0cc2207a : SUCCESS in 7m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0250c808531945c2b2472642de152b53 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/648a5bc610284c6982924365742e95f1 : SUCCESS in 15m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4fce2b27eda44c0a8b4818460764dac7 : SUCCESS in 8m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a2aa4328d30247d1aa37b9623161ea1c : SUCCESS in 10m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a2776f524a86421a95b102db417d4635 : SUCCESS in 3m 49s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a0e5e19863d248c8a7b2db46ec1f814d : SUCCESS in 19m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/852dfd2bc856473f83296c41b83eb9d4 : SUCCESS in 7m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2917031600154264868130c03dda5012 : SUCCESS in 11m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/db0b556a6a964e0d8d9371ec051df176 : SUCCESS in 30m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/381297d8a2244061afd6ccdc6589e2f2 : SUCCESS in 13m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/54f86c5725c3488d9a8316c29a34a63c : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9349783529354535bf1547309cd9f1ef : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19487aa4600f4fa0821821cb25a21e0c : SUCCESS in 27m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/81ebae0783b74d75a8e63fe50b7b277e : SUCCESS in 4m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8552dfd6e9c44ae19eea27ecf403681a : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/700fc909945049f9bcfaba226b91c055 : SUCCESS in 9m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/52e38b828e214a45b8472f8a0c0e9c47 : TIMED_OUT in 40m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3888255473054683b779e49dd5ae3aa5 : SUCCESS in 53m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/478513c3c086496e9823c8f190b91ecc : SUCCESS in 1h 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a9cb7d04e91aa8d2e7bcdcc6d5dd52639d4eba03","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a0f7169161a13c1669b7bc2773a832af358e415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:05:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c57975466a5149cebc6ce7fc76f4b44e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3c69510e39bc45938c2ce792c6e21075 : FAILURE in 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"eb8da96df99babe86e0128ada1861a7d7b85494f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:44:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/852a19cc772a4616afcba82d480164a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390dc2f2de684a16b6f3851899c03572 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d5186ae463411582eef82f97bd03c0 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/418ed402468443f1b3190cad426a400c : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab1d6671cf874b14b30b10cbb997abd6 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e36f0bcfafd0422bb607dd1e80859c8c : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2206ce076b6f4b239952827bfe779eb0 : SUCCESS in 4m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fcc13712b494486eb0ea132182d17b26 : FAILURE in 8m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/39d5d96d18294dc3b7e7e8576bf94380 : FAILURE in 12m 33s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f28053a2a98d4c0d8e05dc6cd25b7c9c : SUCCESS in 6m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/813f8cad9cdd4f7ca84d9dd5c8b22c4e : SUCCESS in 5m 27s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/38acfb2ec272490d88ed738379755724 : SUCCESS in 5m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/65b9e3ede57443e9b3b12524e2564c92 : SUCCESS in 9m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/03c8c922b3c3464da6f5f78504780573 : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/21bc39b336454f15b1d8f4405be544ba : SUCCESS in 5m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2c21a24ed0424965a5804fa54912ddc9 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5e5a7d6f238f4100ab1d3259ba8f1107 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6fac9ea0b5f4f13a96419b23411d44c : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bece75a898614815a5df60d0b83e036c : SUCCESS in 8m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/74ed7835a59c4c47b76a54e903171bd3 : SUCCESS in 11m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/663001be898144fbba7d2eed5925e521 : SUCCESS in 4m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0edc9676bff6439b80f96945635ffce7 : SUCCESS in 8m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/75f0d9b74f754b5a9b58056742495359 : SUCCESS in 5m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e694f68a23b44dd2962b39be374ab798 : SUCCESS in 11m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/443d4d1a80774b85a8721fb9866f76a4 : SUCCESS in 24m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6ab0fa7938574f449f1a5ed080a7f44c : SUCCESS in 18m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3e132c2dba5140b39ca16a6975286e8d : SUCCESS in 6m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1a14bdb86b641cf906886e6e378b4b8 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08a2a55264d740418121b8cb4bec8062 : SUCCESS in 21m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/27e9277311ea4e7d89fbb992a9f88ba3 : SUCCESS in 5m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7098759496fe46299af9a6ed02afde2c : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/331270ef8c514d71bd7dd3ecff1f0928 : SUCCESS in 6m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a5eb9f926f2d4a208f5e62741aaba5e0 : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bde33dbddc640268f894ffc562f4401 : SUCCESS in 1h 01m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ec8f8beff08493795ad48168df37067 : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"af128e7df2b3940ecd46a4ed325a1f2da2d1ba70","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 18:09:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e321d73aafecae6d0741989c2d132c177ffb676a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:27:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5e61998e7a4755ac8b4fb38b2d037b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/baed3b6d58dc40319a03ff13b1c0f5c1 : FAILURE in 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"79b426ee3aadc21cf434c2dc0e720badd9900e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 20:23:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5bef14952c54355a40efca77fab0c8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163d979076f747c99ee0e81062f39e7f : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa877c0b6a94b48b0999f6f735b2b93 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cb4f651f47c474ebc2b8a24b50a5244 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1d76138b9f540e887c798d4a69de502 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c89100319a8848e0b2095a386d127a83 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c222cfb1887045edb258c08c9e70376d : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8cbd2825daed4086b266eb5fa94431b9 : SUCCESS in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c37f15aa0354c89a7e0810f7f75eb64 : SUCCESS in 14m 49s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0042a61370794c6e914047477d46b4a9 : SUCCESS in 7m 59s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e7ccbd490c1c46d896dd9b7e4bd2ab03 : SUCCESS in 5m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/952d8e38a4c640849a03238759eb624b : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6eab8aa97ec142c5b90f89eb54d68bcf : SUCCESS in 13m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0a6201725bc240d3a0830eacd26add5d : SUCCESS in 4m 04s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9fe2a43be2d942f49dd5acdfadcc6bd2 : SUCCESS in 4m 38s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9ce21ee25bfb43bdab48d082ac45cf43 : SUCCESS in 6m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d0d7029caf0643e6b0502bafdfa7de81 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/113e962d09204e5189951ab9cc42c5ee : SUCCESS in 7m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/62f0c67a188f4675b97f46ddb6ba65bd : SUCCESS in 8m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/38464173a74a4bb79e5eb7bb234d836a : SUCCESS in 10m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/97f162746b414f109dda2bda242a739d : SUCCESS in 4m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/aa0770f10a884a5ca8b36b7b665a8bb6 : SUCCESS in 12m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6de3d39e226a42caa4fc9fe52371fecf : SUCCESS in 4m 21s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c42e9c7eed85426da0269e3f5c460a60 : SUCCESS in 9m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ef74748ee2b943fbaae600178747df29 : SUCCESS in 22m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a73aafc0e0ad4938ba1c7fb409e26e46 : SUCCESS in 11m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/34bdc604be1546faa0ea62d982a62709 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd83032972b84a99a0ed09774e5e5734 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/087994865509452a8c48a0b5a0e2e1f0 : SUCCESS in 21m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1f067a547c2543519df66a58d583e553 : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/dd2bb3b8dd244527ab7a648802c94888 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc395babc443465286580b2e273ad8eb : SUCCESS in 9m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/981653f852744ec99cdf9e35a99c3da8 : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a212d5dd84c4164b1fddd8f00578db9 : SUCCESS in 56m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35d722033741468aa2e8c89419f4e2cf : SUCCESS in 1h 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c7a158517497e2c2c473aa1c08d04bcafe35575d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 00:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73dacd91b8cc65b9f721bcf21b29babc66396e0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 00:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8ce10498adaf65553e2f5be1d80d03d03ca1733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:18:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1801228b541d42f387018b9271d1eada\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877038a1161848e593ab1d478e0fd457 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cbb2179c784ceeaaa4100693a0daab : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac0aa8c0e4b4723b8ba0c05f021ec3a : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab9e2afd748d44b0bf85b95e1ed23109 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26e851af914945e7862cb5e5da74f8a9 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e90fffd924854c47a1268764a70061bb : SUCCESS in 4m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28d211784d7f45fe974661f8c88da82e : SUCCESS in 8m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/86bc3a2be782486e9444d661969c7a6e : SUCCESS in 13m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1e658b7f33744356a90d94c8fab7077c : SUCCESS in 10m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9fc8bddc21a847f8b719ea80315d7095 : SUCCESS in 7m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/371d608f468b430791b9123350e3b9f1 : SUCCESS in 4m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eae385cdccc1453db6a4ebce0817e172 : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/09fa3c1548854ca392bdefd9d93a5939 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/949855f9c1fb4181b96d265e2345b984 : SUCCESS in 8m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b6e0948e58ec47868d67cb8f6332281d : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/653ed9cbd09146a0b09b0ee870fd795d : SUCCESS in 10m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bc1dd6ef5a6246d198d43eecefb6a38c : SUCCESS in 4m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/44bb3f8eaa74416aa2fae4b10a17b4e0 : SUCCESS in 12m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d9e0a56c1f83411d8ea3b54ea17ed9b0 : SUCCESS in 5m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/baec7ab8ff92401bac475c86f1406809 : SUCCESS in 30m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/be8511c4df3d4a1eb42d1b686386527b : SUCCESS in 13m 23s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b78b11a3f67e4625a3ece51fd44110ec : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7c4739d7b8549829cce89061d2f5612 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d864b68fd6094db88efbc3332ea640f4 : SUCCESS in 26m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a439c2ca38de46a7928c139a86bd14eb : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1b0f19bec9b34a3db307eb4cc9cbde2c : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/703d06be4d984e9dbf38df6c0f1b7a1f : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4d92da4cf0dd4364ab89be1c4b3e0ffe : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/886f46e9afe542a6b09eadbba0ed5589 : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09dc3326c86343e4876ea323790e9fbf : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f395ac74afc9998a169b2e4e832a344e62c2653b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:18:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"941bae7b217ef708e30f344f903db7508654a3e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:20:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/428149825dbf4b2cac909327a348ced4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb60194933d4b1ab9039349c0a2d863 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0c0778a7980538c01f7847a5afabb8984c8ecda","revisions":{"f0c0778a7980538c01f7847a5afabb8984c8ecda":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:30:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/873391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/873391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/873391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/873391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/873391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/873391/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:57.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 6.0.0","message":"update constraint for oslo.i18n to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iaac8459c90c1adcfd30b4995ea339b47d3805314\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I8118b10e5c6a649a1311cda7c7b8013f0fa7122f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0c0778a7980538c01f7847a5afabb8984c8ecda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0c0778a7980538c01f7847a5afabb8984c8ecda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
