)]}'
{"id":"openstack%2Frequirements~873392","triplet_id":"openstack%2Frequirements~master~I2a53af1b9571496f237ceb7ac75ad9a014436197","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2a53af1b9571496f237ceb7ac75ad9a014436197","subject":"update constraint for oslo.metrics to new release 0.6.0","status":"MERGED","created":"2023-02-10 16:30:59.000000000","updated":"2023-02-17 07:05:55.000000000","submitted":"2023-02-17 07:04:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873392-new-release","meta_rev_id":"9984b333474812deb971ed24b3c95db0c5a6caee","_number":873392,"virtual_id_number":873392,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 07:04:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 05:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 05:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:59:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-17 05:20:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2ba8ecde1780099f7a0d5adcfa57a6496007726c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:30:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faaabe251c7e1d4dfd6b4f235b8996d6524a4335","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:13:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c3cd7b3e0446d58c01c46d4ff08409\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/17b9aab8fb8245faafdfcac4dcf0c99f : FAILURE in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"54719f63f5a4753e7adde669bc350815b75ef62f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:59:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57d5b7abd0a149b2ad09fc42891099cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38db244a0ffa408c9d591ec991f6b53d : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b01856ba6374581b9db1ca76b5e4fef : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5d08b295ee24fb1866217d51fea9518 : FAILURE in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6edabb45bdc4267a6de2669ca116bf5 : FAILURE in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe821f88dc5049f183376393340e1985 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be025c31125746de86147b6f68eeacb4 : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94fd2e751fdf47d6b3d8a975f69f2514 : FAILURE in 7m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e950797f2af84dcfb4c57d53688c8640 : FAILURE in 11m 42s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e2c05b9e7f5141399791ab6b30ffcb91 : SUCCESS in 7m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d47c6288f9f3480694230de7beed3d90 : SUCCESS in 5m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/77ab610bf3534e2f96b27303fa46fd6e : SUCCESS in 8m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/390c468b5abd470a865e312142738652 : SUCCESS in 10m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/784612fbf70a4b13b8551109dfc0498c : SUCCESS in 7m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f481424ab84240f0b3dd426d96f0eda3 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1be7d7e457ab410a8aecb1636867f21b : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/287a58b108124aaeb252fea08298d708 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6dd97dc7cc434a0889d984e92f38bc9b : SUCCESS in 7m 14s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/84f4ac2b7f91418ca8d477e9fb6ce87c : SUCCESS in 9m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e708bc6e2a3b41e8b98a8b156b2b22ea : SUCCESS in 10m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bc677a6eb0ca4e999a4568c0bedd1efc : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/37f4cae71d834328a261d30e6cb8cdf4 : SUCCESS in 10m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/da001899e2434e128b2ba0c8b695da3e : SUCCESS in 4m 01s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e3481deef82d421e9cd9150c09f3b924 : SUCCESS in 10m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8a4839ff9aaa431484c852ebe358906a : SUCCESS in 26m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2d8861e6896c4d77b10701e3902bb14f : SUCCESS in 11m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/08e528173d624a2aa7363be6c84af55d : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1345c4802f94c89b724ccb18012131e : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a32c6c090df4f61a43a6b88a467be6c : SUCCESS in 24m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9c1b0bf39f3c4c28b491430b92ccf606 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/73e670be24994f77b7e26aeb4df59948 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db36a232b24345ae9585ebc2f8ca117d : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2734750ea62d4d1f82be33ca3edc8640 : TIMED_OUT in 40m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/485d04c88bee4a6fb0936c5cd2dcb6df : SUCCESS in 54m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5614eeb7e52348da816d537a6c7f44b9 : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3eb19c0a5aa0e413c523c1172759e7d7866746f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b3364e88b1897d2c4a4dc507e75c37084697e888","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:01:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344545438272457ea3349833b6e8ce98\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e68ea43b7fd34e7eabf30a152b8454ba : FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f802a05d497915139ab185562e3cd4b1b0893b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:14:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b484a146dd642e7877d0d055562fcb8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe532fa588943e4b32ee5c113a9356f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365d6721fc834a8aac1f3f84e3359c53 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd823d73a82245758ed90542edc1582f : FAILURE in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f615c80297b4d358c786b3669350033 : FAILURE in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b3436f7b4744657b87985c6a59b2ea5 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/69fde7d60b5b41b8a36a415cfba094fd : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d57bd7af0f74fca92df183f38c90994 : FAILURE in 6m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bfc97615c5cd4ab98c53b9394d3723a2 : FAILURE in 12m 29s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3e53de699b7b4c638b1bc695fa0386f8 : SUCCESS in 7m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2f356caf83a84e14aad1abc951ed91db : SUCCESS in 5m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/32ea2f4fe30e4f2093cdc5a6a8d78acc : SUCCESS in 6m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8195304014f845229142bc35417fe3d0 : SUCCESS in 10m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/acd9c2e7f2814fb498dbcdfcc0e4899d : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d974c7a1438f43d1a9e955cd1faf9089 : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4a3c097946a84d3fa7b75ebbe895b405 : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f6bf14844d0443778c9756de9ecec65b : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e580f7c4af744e34ac8070047bff7be6 : SUCCESS in 13m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2e03afa83bcc4bbb9a9013ca1402df7f : SUCCESS in 9m 23s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f6a517246372494c94611ccb4134a2aa : SUCCESS in 11m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cba66ed893524c4bbab415ea661a3218 : SUCCESS in 3m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/98e2227d333e411eb9505a1ba9f45712 : SUCCESS in 8m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/352c0a63122e4abbb09c5a6f996294fa : SUCCESS in 4m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/76db79830dff4c81ab8176a8eaf73402 : SUCCESS in 12m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dc51640fbae143769a04b4f7c8962c04 : SUCCESS in 38m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f29a8c41bd7b4ef19d0627090baeed22 : SUCCESS in 12m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0c40db6de7da479e8a7bba091f18c143 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75263c39bfc940f4ba4ca45773d1806b : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45414b12ad394d2a8ee8cd3a9131321a : SUCCESS in 23m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/71d67f28d977478683dcdf0826b3808a : SUCCESS in 4m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9bc48325556c42f0b45f458f839492aa : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a049a74b297d4ef2862b867c6f1f10f7 : SUCCESS in 6m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1bf37405e5034926b55cf1534c0b764b : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/895ffd66849b4bc89ef5747d4308725a : SUCCESS in 56m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1588eafc0fb466cb88946436795251a : TIMED_OUT in 2h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1b4471c3df77be09d06d8710fdc7d577e3a9372c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0357e072662c36a832ad00a9e15aaf9431f5449b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:58:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91823f1041fa49b49d8b17dd2e10c1ef\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a2cbb0f57d7e49e2a319c1920757441a : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"22edfb0c6e85a6e529a89fc3b8949e42089abff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 08:07:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b66c0d5152445ffac3bd5ca57c70167\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d80ed0077c64b6dbb9382e18eb312c5 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd8318fb2ee41a2963762a2041c6a59 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da93691ba4514daa9bf8fc2b7c17da62 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87d54be717b546d09716d67be8668ab9 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dded983c3a9f4ecb83b2275f675fe9b1 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e9ce4424d0fb44218881a66e3f2a728c : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b8bbd2278a74074b885cb7e08707413 : FAILURE in 7m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/80aca33d5650409dbfb21f00268e2a80 : FAILURE in 11m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6fd087c9d14449069e9e3a4b5becc4ba : SUCCESS in 6m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/744df45805fb4b1db403b63393e5e1c1 : SUCCESS in 5m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f30134c215674099b50803b8c6181f5d : SUCCESS in 6m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/76b64729df71442c865cdd255ee27631 : SUCCESS in 10m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ab78c57a1ec847baba3bf33489fa4689 : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/13d7b9efc9584c139a5fa67669814d18 : SUCCESS in 4m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/41393be4a8bd487c80559c91a8473863 : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0c151c9860db4e16a44cade8f9416074 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a72c63819d5741d594cf9441913c8b15 : SUCCESS in 11m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f4122e53f90447078829de8a4173b5c8 : SUCCESS in 8m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1c65edacaab245f4adda9648390ffb2b : SUCCESS in 9m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/765c83adfe5d486883c685647b5ce497 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/df5b1a2796fc4484966ce5455767c658 : FAILURE in 16m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2ff2a81e4e7544fd9ee6aac6d711312a : SUCCESS in 3m 55s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fea286c739a54650a0ebb872eb7a193d : SUCCESS in 10m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ef62cc45dd454d1094be306b83758e30 : SUCCESS in 32m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b71ab8e510d84246b8280053ce2b4fe1 : SUCCESS in 12m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/31ace75b5a0644a0979cda9a7da051bd : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6a674775a5f477681babc89d475fcd5 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41312ef9583b4d3488978af20e5cb150 : SUCCESS in 21m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9c1bb338aec5462a8f9cf56ed0c43bea : SUCCESS in 6m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f6275cb6f5444a4c808d23fd0ba99cfd : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca7219a903404c98bd570a52efc3d001 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c088add6b6734a12a746c704a12781ce : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f82c2f620e614c3c880ead9fb41560e9 : SUCCESS in 43m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9637bf853884ac08d4ff3a41f996892 : SUCCESS in 1h 50m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6018a7f6414f2208eb0aefaa24e10347cdcdb550","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"070a1ffab0d6d2410942b8a47e7cec8f3b9996b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:28:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc13fb4043646c6adb07f299073fd38\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4eaac7e5640342c686859d35365d4b33 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ad5111d82c4b42b1820ef66b519a58b47cfc0b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:05:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/034da4d7b9b3438e8d04799dadb46bb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb01a57dbd54a0f81eb8e09c6dc6075 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db535fce93ee4aeeba39a2abf06077ba : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbfa2fff7795499a8b644edbc663260e : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/680d5cf3f7e54346ab2b8b7a9da9ff10 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdbb3be261a54014a9a9d4871d460736 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/167b9aabfa95428d95bf626748ff0b1d : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/41b23948cf26453e8faafa08444fef1e : SUCCESS in 8m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3da39418be1c4185ad93b33e61bdb67c : SUCCESS in 12m 06s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a51bf3a8d7da4ad59cf9f798b23fe75d : SUCCESS in 7m 55s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/92b647ba991d4b7dae294e11364be20b : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d9bfb0500c37455f89c87740732e3fd4 : SUCCESS in 6m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1d9ff07b11174892be95ff1c28eec277 : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f08cec8adecf4f9db72a1d8e4932f91f : SUCCESS in 4m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0f9b9a4762824994927c14bb14bd2ed8 : SUCCESS in 5m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2460cd0b340346be822759bb0d41945f : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e9b3b0007f7c4c3e85b3be45ab46585d : SUCCESS in 8m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b97eb5b9537d4f4b80b509c088482138 : SUCCESS in 6m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/22d83f36f607415aabc1be3845915ebe : SUCCESS in 8m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0e0f1ded411647ae9cc3df45549543ce : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c68ddd4cdd674ef28a234a311e17403f : SUCCESS in 4m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8394e131b7f5473180fa745d1a846f53 : SUCCESS in 10m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fa77b112e8694654841d409b53103dfc : SUCCESS in 5m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9625acb8eea941b7bbd3a10d169bdcc5 : SUCCESS in 9m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/80a19bcf36bf4e818fcde514c4c71997 : SUCCESS in 24m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9851298ea9b549218dd827e3b15c7d4d : SUCCESS in 12m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1b5c2bc8795a40c8950153b021655931 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a0f774aa45f4c5ca2ae30ce508b14ae : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58a7e4b109944f5b93513079e9611258 : SUCCESS in 21m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b973eff8f3194f7cba16940f41aeab38 : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2b546227c4ba43d6907b3955ab0b28ba : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c4cab9d3441484caa564fb880031ff5 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0ede841ddb7e4de6964aa2bbfb710fc8 : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be956ade1d224fb1aa62ef982b048c8a : POST_FAILURE in 53m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61fe90d4b3e44f74963de9d8c40f9ef4 : SUCCESS in 1h 37m 57s","accounts_in_message":[],"_revision_number":1},{"id":"87e7e53ac602d34e52a5c728951cd9fe28c7f800","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 21:27:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13fa13ec4d0d9842335b1cd30fba020015c94e7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:33:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b34ff23bc54d37b29a39ddb8df582e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/91173f53f3f94f07a075dc5c1645afcc : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a097b352346d76c114d255483690bf7409accacf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:58:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1497b27feb044984a9662a8eb40a840c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3653f8315437499f875d807fd4f0988c : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b20f8e8e0774e8eb33ac09e42629429 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f6ce228914644dfa2899b3cf0fe6bdc : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7646ce73c1be431eaeb8bdaefebc2eb3 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/079e29f114cb4135b6c695b14d2af864 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cb5c47a4a7e47dc8d0845f371454f8d : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ae337220c0549ef843439b5173ef5e5 : SUCCESS in 7m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/85712eb056cd4d068ad41f238a729255 : SUCCESS in 11m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/45806979ce944317a7d6e24b86f484bd : SUCCESS in 9m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b67b67fb9f644df199448e86934f55d5 : SUCCESS in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2e602f67eb7c43b2bb26b9c5ea6190e5 : SUCCESS in 6m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/031a863ecf9c4857a7e2e32e91626eb2 : SUCCESS in 10m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e58c6548f31345139182e145bb3bd8f0 : SUCCESS in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a4c5697c8dea48bf858c1abff71ce987 : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9706953c71fe433cad092e502319a91a : SUCCESS in 7m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6e970be611d5482f9e17e852dfb0109f : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16acb5595f48444ab3cd5a9537d507a9 : SUCCESS in 8m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/337c7932131543b989773897cec5594a : SUCCESS in 8m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2730d288fe884edb910e65e5eb4fb835 : SUCCESS in 11m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/911f8b2e595749c79a6140eb0f584a71 : SUCCESS in 3m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0a5d468543f149a3be1d5987757d248c : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bfbb7840672e4d039b4e6f057f86723c : SUCCESS in 4m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4763c24d800b4c80b9ec515e733b3386 : SUCCESS in 9m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a33325a363a847519b1aa86d1c64a2ba : SUCCESS in 23m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/780c0d5331d34d2498e7ecb01934967e : SUCCESS in 12m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0cb8ac83fd3a4fb384c3a9cbfd0d16c8 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab8d3b3ada63447983facda36ec83f55 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c035b4f877e745d582a0f3b928b6a661 : SUCCESS in 22m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ec806c0f4983475f935b0f45b1076355 : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9a13b3b49bbc4554908e18bda8db4a3a : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5929114d8ca14e8b81f98397905ad804 : SUCCESS in 5m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/589b3844de964b32a0146a051ca9524d : TIMED_OUT in 40m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2bc1e438d804f07b10bb6724d61f97c : SUCCESS in 58m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8f8efe44835480bba82563ed1d52727 : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":1},{"id":"771c764b2d66ae877c22cd4072e4fc1925815cb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:42:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e9bca96c97741e8f7f181846137f7f9ab5f4850b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:50:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be4c9ebd782b4ed499a6f1f03a3d1cd0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7cf734ddf70948578ef48327012bb48a : FAILURE in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"57c5abce43b2c931038cc789874f7fa453022071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 05:03:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9d3e7f2133466e9fcc9cb862d9500d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334edcdc86af4d76a254e8eca63e950c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1491c6835049fc9dbe580f01d48d3d : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f8b9adb7cd5415c839e5b157f865683 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b56ecd2c65664475913e0bfe7481373d : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b809ab1ea4f7407ea63ef8a82239cb7f : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3299ca74f7d44665a15a76ed61eb3826 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7754a6287fa14c24adb12d02169c2846 : SUCCESS in 10m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/205027bb1d6c4dd38a7b324180d2bf2a : SUCCESS in 13m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2ed8eb523d94497e8972a3297ed0d065 : SUCCESS in 9m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/53ce7610d1ad44e399cf7d5fab19141a : SUCCESS in 7m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3178241548d94a96942d950a0adf9257 : SUCCESS in 6m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ee61591bc52d432eb59a36dd9bb73c19 : SUCCESS in 10m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/611e2eec4ab94a38a77a2663759cebf8 : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/816781acbaf240b0b43cc3f67e0e5db9 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/018445893cd5455f8f94830eca92f522 : SUCCESS in 6m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3d14cd42236244a9accac1b9a3affa05 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e139d31eea348eeaf18ff156eefcbae : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e5755885cf0b4aca90aa6b0c31fd6196 : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/34f93e0688d34054a2a3280f83720b7e : SUCCESS in 11m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/236a271329f1412a88ab233f2d437607 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/72d4e544974f41099712c903ff07ac2f : SUCCESS in 10m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5581ee7036c4461d94c2aed54f54487e : SUCCESS in 5m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b1aa7658529c4a79bda1af86be4c8c3d : SUCCESS in 9m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b76fe2ce815e442bb169fc210ba7873e : SUCCESS in 29m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/67c71610217d4a44bd5eb61cc8f6f567 : SUCCESS in 14m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6fbc6b097b574127aed001127e11834f : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d14f773a8d76425fb95c737c8617a2bf : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce643818ff3d47cf942f936a22ab1c38 : SUCCESS in 22m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1491967d3c2c43cc8b9b8fc7936fe44e : SUCCESS in 4m 52s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9d5d534abdce44c2ab2df4327e9a9586 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50d4e87a60084dc78d9ffc5eda8171e4 : SUCCESS in 6m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9d6031f950fb4574bcde127c742773ae : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd7bc9957cee46d48bb9cb0702df79a9 : SUCCESS in 1h 13m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/103c5a5034ba45cfb8fdba55de032556 : SUCCESS in 1h 30m 02s","accounts_in_message":[],"_revision_number":2},{"id":"36eb38c477b6199f35385d99c7fdaa2cac4e014e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 05:20:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1565e55c1d02989c161a22d308a34c91f9fd5631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 05:21:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ea424c48edc8715c90efbecd2069362f9c71363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:04:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818bb57263234752a5981ed9d99b84a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f17295f41524a8f88f003a88f9b3f2f : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6236b15b01fa49de92f3616e55fdb372 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6fa8f1ff7574796bc5e1d26a55c9815 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b767f635cd234ce5a96c867383dbd517 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caab35d75d42402498d1f7f04bc56d1f : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8d2eed717c946dfa76719814b3b79c5 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35952f0a3fd24677bd04ad39de363d2d : SUCCESS in 8m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ca76abee64f4092b7b8aff449ff210d : SUCCESS in 12m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6afcfc33ccb54ca8b7d3523d7a1cad6c : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/304a0726c68e4d1ca6d5e5d663918639 : SUCCESS in 4m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/001b09b7a74649aabee4c578d24b3fbf : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/26b83c497b23471a98230ccc1f14f634 : SUCCESS in 7m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a5eed71ef11a403c9256489315de0ef1 : SUCCESS in 13m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0fef9fe010046d1b6e46d4e19ab9855 : SUCCESS in 6m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5f9437df93a049638e075ce579c538e0 : SUCCESS in 9m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3c724e779a6a4e9fad93b43ddbb27dfc : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5a8c010ae76540819d290504085f826a : SUCCESS in 4m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a4095bbe99524c6d91c980e4d81ffcfa : SUCCESS in 8m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/17a9b98c29134effb6aa1484eb600162 : SUCCESS in 5m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e56751e89d0d447e879d991c706a33d5 : SUCCESS in 23m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c2ffcbdfb7eb42d294d3c87a95185eba : SUCCESS in 10m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/aa43d62f42eb4b1e9dd4ca5aee54979d : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a3fcd4e8195d4cef809e16a5b0381656 : SUCCESS in 7m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbd164ec5748464f8ea1253bce00c7c5 : SUCCESS in 21m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/95d138cf6272443da464e163eb5b9f9d : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a2ced8f07f9f49e79034b899aec6f161 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6cf7680287049bea49fc7d666f5b9c9 : SUCCESS in 7m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f891e5c96b4a473eb85f66af04258ee4 : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a927f932f75a4089b6e972dc88d1d117 : SUCCESS in 1h 00m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31d6387b1c434bbb9d3e4b72b3cb6567 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":2},{"id":"518ff2d74ab647b6b1bba41b0b346d136c76fb10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:04:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9984b333474812deb971ed24b3c95db0c5a6caee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:05:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45e6046429d547f49ad035057dd0dccc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e583c3bfb2d24f12bbaf67e02cd2d1c8 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fecf32850b0a295c2d6356dbf32bfb7a2730decd","revisions":{"a7b865cf5b6975ab5ded83104c5ebc1244553aac":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:30:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/873392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/873392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/873392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/873392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/873392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/873392/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:57.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.6.0","message":"update constraint for oslo.metrics to new release 0.6.0\n\nmeta: version: 0.6.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id90151c5bcd2e431130579aa5d197be879a55bab\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2a53af1b9571496f237ceb7ac75ad9a014436197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7b865cf5b6975ab5ded83104c5ebc1244553aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7b865cf5b6975ab5ded83104c5ebc1244553aac"}]},"branch":"refs/heads/master"},"fecf32850b0a295c2d6356dbf32bfb7a2730decd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 02:42:17.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/92/873392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/873392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/873392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/873392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/873392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/873392/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:30:57.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:42:14.000000000","tz":-360},"subject":"update constraint for oslo.metrics to new release 0.6.0","message":"update constraint for oslo.metrics to new release 0.6.0\n\nmeta: version: 0.6.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id90151c5bcd2e431130579aa5d197be879a55bab\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2a53af1b9571496f237ceb7ac75ad9a014436197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fecf32850b0a295c2d6356dbf32bfb7a2730decd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fecf32850b0a295c2d6356dbf32bfb7a2730decd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
