)]}'
{"id":"openstack%2Frequirements~873393","triplet_id":"openstack%2Frequirements~master~I759e21944b67aeeb3009bef9a78e06a8ba81648b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I759e21944b67aeeb3009bef9a78e06a8ba81648b","subject":"update constraint for oslo.context to new release 5.1.0","status":"MERGED","created":"2023-02-10 16:31:43.000000000","updated":"2023-02-16 02:21:20.000000000","submitted":"2023-02-16 02:18:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873393-new-release","meta_rev_id":"c0613172054eaafbbc804ea69878787060db5887","_number":873393,"virtual_id_number":873393,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 02:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-16 00:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-16 00:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:43:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 18:09:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-16 00:33:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ab5e5c21c69d08a8a66b7add8fc7c70f007ee2ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:31:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2912435b1859d7795678598895901446f21ae08c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:14:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbf91cae10434962adc86ecd6594b03f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/54821e7a4f2649d28de7689eb3be61c3 : FAILURE in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"edc6c6f5f1fb8ee3f787ae2db723d6626f6dd225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:43:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dd1c34246ee41148fec9d7e4334c8f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1fa547d585407394dcc01193ad35d5 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75969b0d656e4b688e23bc316b972e4d : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7983867d1179488b94abdb35f72bc270 : FAILURE in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80b030a3da024d0bb6daa8c07214bce7 : FAILURE in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8907944121d48abb3efbd36af2f7e44 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71b17030d7fa4c0e919e88f2efd04598 : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dba3aece3b794b24b0ca388fe72d5adf : FAILURE in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/15686a52318b4eba9506289f32568c13 : FAILURE in 12m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/71e079729e044241933b68b4032b62cb : SUCCESS in 7m 49s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cb307c52c4fc4d9b96cc5e8cd4b62a48 : SUCCESS in 5m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/bcd892aa8c7648f69902a9bc75055139 : SUCCESS in 7m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a9d36530dd27413a97ea0ccfd0a5f80c : SUCCESS in 11m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a8b9e9a5676144dfac12022dacd17c2a : SUCCESS in 6m 36s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e91c02e8dcb746d98b1f4cf1aed86b7f : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/529431b0f504419e8af3001570e85113 : SUCCESS in 8m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fb64cfd85097425585e892e124865181 : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2c093f3afc545389ccafc2034a6b6ff : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/153c68cb6656431189e32e06f99d530d : SUCCESS in 9m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/67b2d96ec3914acd949d9deb33073c1c : SUCCESS in 10m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/083f7eae4a354a4b83a53ff001f433b2 : SUCCESS in 3m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/242e58139a074ee48066b7c36679b7fe : SUCCESS in 9m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c7103b68f3344172bcf72982f15e1232 : SUCCESS in 4m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/87a2c23bc9334839a79a5f54ebdf156e : SUCCESS in 13m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1e8532d7633e4dc69eab1e03af9c5d6e : SUCCESS in 25m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1b2443918a4e41b29f7cf0044a613e3a : SUCCESS in 16m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/77be6c0f95ef4a4fac23b3376a16090d : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8c13e661c3a4b2082a6ec26bdfbdc96 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46c60cc48d414500b65b32457a3d4bba : SUCCESS in 22m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2e87d77238f14b8db54447b042a275fa : SUCCESS in 5m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5c48eac09f4144998a6b370ce6be9475 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b5a464f2ce64b69b2236f161cd0509a : SUCCESS in 5m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6c67bbb3c1a2442c83ae2ef08f26b67e : SUCCESS in 18m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a3090efd45845a9b98ed432aef79fb4 : SUCCESS in 51m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5871c0d4ac74b408c541f09e0b9d25e : SUCCESS in 1h 34m 18s","accounts_in_message":[],"_revision_number":1},{"id":"99d8896253bbeef09d9c0d8e32f7bc7aeda8cdba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"231433bda920d970dabeae95f6856e98a7928123","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:02:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ad8cbade21048b09e23c984fa4cbb1f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0f916ec80d224af89a4a5dd7a71a78b3 : FAILURE in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a932dd434220815be4cbd55d1562766b7f2de251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:50:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f28e929a50184b469ef3792a31ce90b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f19c815056548b2ab201c9420830fd5 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd1d3f8fb2b49ec8310d5e852d45c01 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8384471343c4b7a8d185aa44d3237ff : FAILURE in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dba6e3216b114496a4a7143c0f824a35 : FAILURE in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db41f7ca9a2b49c8a25478e0bfb95873 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74dc0ea4dba54318903731aa31d249e1 : SUCCESS in 5m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42fefd17987f4c31904f0cfcd185156d : FAILURE in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/165d37e0511c4affb65b6b5444df3d2c : FAILURE in 11m 39s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/118233d2a4e64e8784f8b6fb789f1cc6 : SUCCESS in 6m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f91ab6ab9ad445de87aee9175b611157 : SUCCESS in 5m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/630eb3f334fe4521abc60aad4e0f0929 : SUCCESS in 5m 51s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/da2223255f0c44c09088af9a9dfbd6d6 : SUCCESS in 9m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b284796deb4847d89e504c8476ceaacb : SUCCESS in 4m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9811860907974bedb6113b871718fc80 : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/879df8a5751b4d27a38d03100d57017a : SUCCESS in 14m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9241a7353a6746bdabc051075c0b2ab3 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0d17bc96a454380b1008ca7a6c07dcb : SUCCESS in 20m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f12f6857945645e5a209600099685d9b : SUCCESS in 7m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ba87d5f6a28e41b8b55605614b4bb922 : SUCCESS in 11m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/022fdb68d655443d9a8a92e78c3e40f2 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/acc467510ee64003beeb1368893c670a : SUCCESS in 8m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/281b727f0e4d452a99edd0f66eb3f5ad : SUCCESS in 4m 37s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/51ba92ba32494f1884832f9a96402b7e : SUCCESS in 10m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5442924001a24a799ea6aa079ae2d959 : SUCCESS in 26m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a37c28e07e8740aa9a4fe5a9b6bdb126 : SUCCESS in 14m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9a5d751ed862415585c9dad1213751d7 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70c32040bc6b431abe79fc8cccc63fa7 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c47e4e3cb2bc4b89b100f9c76666ead5 : SUCCESS in 22m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e3764f3de1fa4fab9b541a0251def08f : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/90be3546772f4c35a9a0262c8d1d0cf2 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2aa6ae30d64e4937baf1cad93899c9d4 : SUCCESS in 6m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8c5d7415a6af42b3b3398da4845356f4 : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d65f79c60edf4708a5646de3a1661d41 : SUCCESS in 1h 01m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f197ea9559584669814dc4058a176aec : SUCCESS in 1h 47m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1ff11e04a4a0aa58826b5dac88fc5ccba3ef0bf8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f16742081a595d1e4754eeeee36147b2cad3a51c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:57:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea930433257480991c92da434028ac0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a3952ae3f29c45e79500950196f35d72 : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"68d5491f2745fbe0678a3a99317ee03afe191270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:46:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad5198b079764b63b10a0e6b5422a39a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28337f87eea24aebb45f0e2d28566e10 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c1a68a914b04ba08f494c5fedcd0e8c : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d92b421f5aeb4d0981ae5ece2f12a166 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d3e685c66a74ed09cf2a0a2a525d6d9 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89c9755fd25c4208a37f1236ca537632 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6c731177a4148d180d6bbfde9233507 : SUCCESS in 4m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0e14ccb7872493a870d944ff6b17ba7 : FAILURE in 7m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4094e130311740f08be007e183077667 : FAILURE in 11m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bbfb24d4ca194718aac63bdd52163ad9 : SUCCESS in 8m 00s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/901ba80fe22d4e6eb083d263fd308262 : SUCCESS in 5m 46s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/26ac984df7ae4508876d4fd4d4caf453 : SUCCESS in 7m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/aed3addd2cad483db266c38775b98208 : SUCCESS in 12m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/86e98f4439d54e14b499103791d2a474 : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/60ddd9a623014c158e8ad63590a25b0d : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4b3fa0bd950342dcb8c9d1e5751ae0db : SUCCESS in 7m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5324cee804ca4425ae98ed2e93638fcd : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e855002ba4f47dfbe3a195f5d8e74c4 : SUCCESS in 9m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e85c212b502c49d593b4c69e70802a7e : SUCCESS in 7m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1ae5fe80cdd94836bb3c52409f45915d : SUCCESS in 12m 51s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f60ae28d7b9c4b25b328c027cf22ac10 : SUCCESS in 4m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6b71f72d8aba4c9992891ecbf66787a0 : SUCCESS in 8m 49s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1d1eedc282ca468f859c0624a0abeea6 : SUCCESS in 5m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3aa815248cea4074b4c8783934123537 : SUCCESS in 11m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/df14aede72ed44da868f7c13dc468349 : SUCCESS in 26m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/064820aebb274fa4827faccc888715f2 : SUCCESS in 11m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2e62fd01a6924355bf16977711a6aa64 : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/879cd50b64d940bbbe063abed4c1d1fb : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea7c15b5c7394ba4a36d4cefac533f9b : SUCCESS in 22m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/5fef06c184b8497db678e93a128d27f3 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c24961a05b534a99a5ddfd8f21b8e7b8 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ae60446593f4921a2f7afa036ea1b5d : SUCCESS in 6m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e750d8e95dd441399baee048e81b334f : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baac2acb3934413790a7c69be1396462 : SUCCESS in 49m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cb5296d917f4d1a9f8f9f18b3661dff : SUCCESS in 1h 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"73c07f63069c7eeafe1cdec16446786204f989e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 18:09:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cef7d444c75678cdfa94c9def20f04a98c815947","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:16:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/621aba19089e44be88819f9223c62102\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d97fb009ebe45d1b252728eb70d5f53 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c47ec63f3d9b841ccd2579f121c48d6637dbad8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:34:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3651180a8174ad587cdd15255e4d486\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebb53e240ca45af9f8cdf96c75da46f : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345f631cbc9941da8d387a5db3392ca8 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0828c32df18b44568af2a84f3c96d49b : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a75e01d4b834400181185d63c3d73900 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/531984c4d8c844e9ad5721e9cf881c45 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c13b1c00d0c416dbdc167ba6b43e297 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb7239697b404480b15d70c5ebbf85e5 : SUCCESS in 13m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/62562c09c45345a2874c0b7490d72d78 : SUCCESS in 11m 38s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/92e3645026b2485a806a0611fca94597 : SUCCESS in 6m 47s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/adb7b517250449428708ea1937c1c2f5 : SUCCESS in 5m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4079f76bf08941f59a83cc5dc4f9eee8 : SUCCESS in 6m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d87e0143edd14d02ba7ab9cc4b45e616 : SUCCESS in 11m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5b89dd2a7c024011b57d56a19bdc8f5a : SUCCESS in 5m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5b60d651ef344ae48aa38240f758106d : SUCCESS in 6m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/23e4a9a6a22e4f6da6f083a460fb5004 : SUCCESS in 8m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d07737bcee22452fa2006d9156aeac8c : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e49cdccb80cc4ed2ad1455b3ca8b5cb5 : SUCCESS in 8m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e0f6514e2c414fdc8143d8fe1a417acb : SUCCESS in 9m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/54566e05c69940b09e2caa7227478f1a : SUCCESS in 18m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5d98faab3fc34bcdab1a754811776097 : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d5ba6a083d7446c6affa138bb72a526b : SUCCESS in 9m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/df273d8671a74ae894598f27ce4336b9 : SUCCESS in 5m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8d9511c7427a44e691832c6d7ae88058 : SUCCESS in 11m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cc203acd3fb844279010b1001ab684dc : SUCCESS in 26m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cc61ff6311304dd495cf55d6bfd37a41 : SUCCESS in 13m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/adca1d4dc6e342abb5354617251e5430 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5347fdcbc3fd4dd5b37f2b1c1737a396 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ae6d6dd927f41529218dacdd319de40 : SUCCESS in 25m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d8bc2492ba8a43499a115e5135d2b5df : SUCCESS in 6m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5311e36d0333408b8a5fe7d06518cd7d : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e83e79d4adb34f23b9ef464940cf25cb : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6e93adc53fe9493b86b1d92e186f7817 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cc58aeb18144d2d8ddc4fa88ce2e07b : SUCCESS in 46m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f12d66e9e3304d14a2ed392795700f27 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"61d8382648a68763c71dd07be8238c6e7ac0ba31","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 00:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebd71055c2a7537e20170c47af84483285604a03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 00:34:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e3073236f8b079c02850bcea18e89d30a61af89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:18:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/378210aae7474f30907f1969a3f98c01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6e884bb634418bb573da823e32f9d2 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5afd9d0617e483db9e240c6b6a937ad : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f8ed33a9ce940d9bb3dd922473307c9 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b553718fd49847bcb3291b22a1295b17 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37c80b9a52254e54966d255cdb0f8c28 : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81eca56d9a6c41999531ccdaa0c88317 : SUCCESS in 4m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2b60d6e6bd9a4b15afb80a8575d7b446 : SUCCESS in 10m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b63211cbd77a4be88dcc2dade4236272 : SUCCESS in 14m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/892cf9bf5b2243bbb82e1062da33fb0e : SUCCESS in 13m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/158f8e31450a49aea84379221cca264a : SUCCESS in 5m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4d251f50b4d84143879e6f23d1fe2990 : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f2633e82f44c488885dfbf236c347e27 : SUCCESS in 8m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8d04e4e487e8496abf62a55e51a6b114 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76dde8f3f964435d9be091f1e7357ed5 : SUCCESS in 9m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/31378217b83f40348a287c377dfc89f9 : SUCCESS in 10m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/13a0634786bc4cb491e86424aa8ff8a5 : SUCCESS in 12m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/18566e93ff1a4333aa21b59110df0226 : SUCCESS in 4m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6e39e9e717584404b46991b3348e2d2a : SUCCESS in 10m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/77833757990f487384194add5a6a6f6e : SUCCESS in 5m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/984cb949ba1947bb9fab88b73d128827 : SUCCESS in 28m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/16b15a0f1d6744f286bae6462d4a738e : SUCCESS in 14m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/32537cd9d4e84bb79f7065c623512d0b : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f07c979978c4aa997478df6f96bf72c : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a91e9780df346159a37dbb363c3504b : SUCCESS in 24m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/82e8356dc35f4270bc8fac99e011336c : SUCCESS in 6m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/58b3c85ff3bf44d0ad11309429a0160d : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f07e3795771d4e1f8aef73d579715f45 : SUCCESS in 10m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/97a45eb3477040a089df92896c57dd78 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c0dcb68bda14a33894563a39486b0c1 : SUCCESS in 45m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29855ade039e416c8166145fbcd54399 : SUCCESS in 1h 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"beff966f0d00331dd00e384f81894072c80d248d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:18:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0613172054eaafbbc804ea69878787060db5887","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:21:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a7bbd1c984042c3b5ddfcd2cf2a11c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f095a8221d3849279f30da263f80383c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e366ad3ad7e8c5e6233216ecbaa0173888eb494f","revisions":{"e366ad3ad7e8c5e6233216ecbaa0173888eb494f":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:31:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/873393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/873393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/873393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/873393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/873393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/873393/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:42.000000000","tz":0},"subject":"update constraint for oslo.context to new release 5.1.0","message":"update constraint for oslo.context to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I97cc41905fe9ba0ff00301a25202710804ae3bb5\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I759e21944b67aeeb3009bef9a78e06a8ba81648b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e366ad3ad7e8c5e6233216ecbaa0173888eb494f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e366ad3ad7e8c5e6233216ecbaa0173888eb494f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
