)]}'
{"id":"openstack%2Frequirements~873394","triplet_id":"openstack%2Frequirements~master~I06a50dddde4ba28faa7f77ce20846a92cfb2096f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I06a50dddde4ba28faa7f77ce20846a92cfb2096f","subject":"update constraint for oslo.service to new release 3.1.0","status":"MERGED","created":"2023-02-10 16:31:47.000000000","updated":"2023-02-17 06:41:56.000000000","submitted":"2023-02-17 06:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873394-new-release","meta_rev_id":"b7f9b0c7e8f11c86f701d272fc45d17c05aed61f","_number":873394,"virtual_id_number":873394,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 06:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 04:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 04:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:14:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:58:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 18:10:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-16 00:33:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"29106f85ce2ac1eb5392201e10602741fbcf10b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53a0585ab097ad7926b2c43904fe1a3aebbad60f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:14:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f580c9766e484a984f74c57285d86e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9d5665470e2644388d3eab706ed5cc51 : FAILURE in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f2258ce15823a55976b5d4e0acbbc2bc23766cbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:58:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cce7450c63342fe96d56756f794a66c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1f3799b7f84055b33c25fc4adb2af4 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ffced43be449768fdd334b3552862c : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bba5038294f04890947558d9d53b63b2 : FAILURE in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/737d78a6c73743238da951381c5b3752 : FAILURE in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14cd49edda6840ce85af1be54a5f0dd6 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7503310b150b45a7b0f56b24f996f26e : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ca9927f207144049058e97a7db67848 : FAILURE in 9m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1d97f4b60f8c452e869d01ca85a1682f : FAILURE in 11m 43s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/34df43e5a0a34883856efb1e7f9bc577 : SUCCESS in 7m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5e77991c531a44038e26d0744c36f3e3 : SUCCESS in 9m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7b44d93c1b2a47f09699769d3dcc315d : SUCCESS in 6m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6b3507245905466bbc7d6619f4296792 : SUCCESS in 9m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/625dd78e5d084845a4ffcc8944146b0c : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/118ca022de564b11a34196d812fb115f : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e73f60808e8a4f47bf28c4a89a767e32 : SUCCESS in 7m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bafb2358f28b48e8bdc73a1945d2697b : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e94d0fa9e5946f29ea026a6d4862ebe : SUCCESS in 7m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/09452969f2ae46288b10208c0e9ae918 : SUCCESS in 8m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/60a1268d0a4d49dd99a99a71eb7b9b75 : SUCCESS in 10m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/224ef6887802403eb90783c0b326a408 : SUCCESS in 3m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a4ec98bd9866443bbab0efa0e9658802 : SUCCESS in 8m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3fa2068a771745c78a9abf69404b926d : SUCCESS in 4m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ee11b1e60e8b40ddafaf0c5c797f1713 : SUCCESS in 9m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3914c0f5caa34c919a15db4475114a79 : SUCCESS in 25m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fc4eef5349184ed189faaa36c3b77198 : SUCCESS in 12m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/4e553949b7904afaadbe7cfce3dc4db7 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b186df2c2a454460933c331acb31a45f : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/537c61ac9a7045ffb47733865b1f92c4 : SUCCESS in 23m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2f78d3ef4f4d4066b68a9375442ad9d0 : SUCCESS in 8m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/51411ca365bb44449421f8bd402fe4c3 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a44b42adab1841f9a54dc120add261e3 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/39514c6740a5472a911cb3f8c6cd5edb : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa0ba198d0be4cec82227004955e6466 : SUCCESS in 55m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b49cfd51a9004fa19b504bda46bda568 : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2cd4bf1cf05324d008ec9f7e88389aade40ce8b5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2e88186bd9e102ee977de799e3add4b1a41f3b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:01:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4436ff99bd254a50adff58aaeb70f4ff\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c02227c47bbd4b2799daa79732d6e325 : FAILURE in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2e29e8ad8fae6cc3db780cb968f0e4557b3feb3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:45:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4040d8ebcc2f44edaf9ec89ecc7d62e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4484a461f64ad496e22166ac0d5d4b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738a42261e0b42f9971917e53bb4fbfd : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f23bad62005940a08a039dec32edae74 : FAILURE in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79bc25411d3e44fcb1da1a08ca154694 : FAILURE in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90da54b0e589480381af6c9a3a946145 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d89021e4f5e84ab5801adccf94f80916 : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/279c199184b047199c61e19a31e6c872 : FAILURE in 9m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1b1bb905b65f43a08cabf49ee98ce984 : FAILURE in 11m 12s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/37ce1a701e434c5185d967f4ee06217c : SUCCESS in 10m 44s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d471c291aa5647b1b3cd0840235315b6 : SUCCESS in 6m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f7e11bbf704140119ef14a5a0d79e838 : SUCCESS in 7m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f767218afd6f4d3b8a1c7e9b5a7ed3a5 : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ce851db71c344c3c9ea8a92e3588fdef : SUCCESS in 4m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b546346d4c314d02a448767e9a30af08 : SUCCESS in 5m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/650109eb432a4855bbdcd1d253465e0c : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a9fa5851acf947cc9c9332189c4b2643 : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a13e274bbe884fac8706d3eb6e85ac08 : SUCCESS in 14m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/10cacd7500864620917871d58f28dd3b : SUCCESS in 7m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/248172588b91457a96a5b75801e8a381 : SUCCESS in 10m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ec836e60df1e4674a5295a8fd0ae1231 : SUCCESS in 3m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5b6cdb65b35a44e490672be6583dade3 : SUCCESS in 23m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cace118575c742f4b37844829a522caf : SUCCESS in 4m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ee34697fe27a4d9aa254233a81730f5e : SUCCESS in 8m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0d6543cfa5124da7ae2aeb3f0a11e271 : SUCCESS in 24m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a94f04dba89a4fdaa9c6792654d22112 : SUCCESS in 11m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d430fdc5965640dcb4785f137f0fdcfc : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d917238356644cac9203d4b165e5a959 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/834aa77a07ae40c5b25ab993614d2738 : SUCCESS in 21m 18s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2da3eae1900b4928b2fef31d395c0d9e : SUCCESS in 6m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8c931b45089442baa216870996b60f74 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e6ec9b6f8694decb4c25985b74a80ea : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ca91bafa47a5481faa5b8036ae60c406 : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28735d72db514070bb799ccc5807ec01 : SUCCESS in 58m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e88a91c9dd0e4147bb2b9e8827b8794e : SUCCESS in 1h 30m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8992b021e00a7a2f1a9abb34b8cd15192a3f79d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f0fe0547891baa9999d50c420204e8f1e35063c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:57:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf9c83bd8eb4e21880ddf39e52f18c7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f51d9893dd824d7f823e8f35d19e17de : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1b03e5b8900cc28b57d5449eb53d0fe3ec52df80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:40:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e08670cc7e5400486b9df6d9769e4ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cafa8d64a436409d9886f8a270fe275b : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76ddea4224d4476ba7ffb1fb76b6e013 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d440c27e1b6d448c83ed69564524e978 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6463e39a7752451fb0eba9ac3e0ecee8 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52fffe5d72ac45e8987254c022f28de0 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21cbc80a7d2442049dbf56bd50f778c6 : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/96b611c0501b43eb857fa70f8de36d4e : FAILURE in 8m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0950ecbceb7243a2a1543d098504d47d : FAILURE in 13m 02s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2c7faf9f59c34ab790960eda9afd53d4 : SUCCESS in 6m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/79e55e0403db46b5b7fd01f18c391a43 : SUCCESS in 5m 10s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d890f06eed1540e3a871e5457243d692 : SUCCESS in 6m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/229358960cde41df9476967bbc914abd : SUCCESS in 13m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/994904e9f8ab441f96c9296abdd71e37 : SUCCESS in 5m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/47d91e26a24e4fe7a2090bee64213865 : SUCCESS in 5m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e756e4dab22d47bfa8b77482499bd745 : SUCCESS in 7m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c1a0d58553c64cc6a4d04e4c5056fd75 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3c3605c5f614dcf8afd079fe5176003 : SUCCESS in 11m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bd9dd914000c448e89176c6c43b6f7e1 : SUCCESS in 7m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/40a7c053f8624e5c9023211cb85a94cc : SUCCESS in 13m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d054b1fe4b904b1ea3b80e55f4ac7204 : SUCCESS in 4m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1feb87b80070460f9661495dd41dfcf1 : SUCCESS in 11m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d3c4587ae0bf47bdb84a5c97b856a9eb : SUCCESS in 5m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cd1bbb13452641e4ac56bf7114b97609 : SUCCESS in 10m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/af816833bbc046f393cb6267a143ec4d : SUCCESS in 25m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/735dbb7c4db149a099b358786fbf79bf : SUCCESS in 14m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/323791c77a9b451aa1b83b698f2f02ca : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ce1297ecf444c68a19129716fcc133f : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31122c9c174d434eba08765ab1a8f114 : SUCCESS in 23m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/928690670c4b421c92f83b1d0b3cf364 : SUCCESS in 8m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c9490f5524214d48bf6af5a4aa969f1c : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33f61ef340684e34ad58176f412f1384 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7d3da51886b34c2fbcad6b3015f76ff0 : SUCCESS in 16m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/975eb91be129495e913a08e7aa936eb4 : SUCCESS in 41m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/977fea7a04b24b63bca8568520c160d6 : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3a64868a83502e6153afac3261168468ec36c65d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 18:10:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af95903aeca5f94c5632c3103f6c2309a387223c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:17:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15086cc824124392992b95e611bf8814\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/73ff31eab63647afbe6b07ee18e3b013 : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bb0ee1eb3664f522d1d607b7e2145f1e2f2d284c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:55:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d97864bcf7e04daea43a7d81388e2582\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3161c76c778a4dd0ab6e90b753805e88 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f5b94a16f274ded8e4290a55a001042 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca8c29f00c314a6dbd40e9f3452c8127 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2587a26bda9a4f8f9adde0cba8fbe807 : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47b4a5224e7f43d08b350d4dc5dd19df : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/830eb190431a4e32bb534c7e91e63d05 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b173b3ab08ef40b6b771960e3226cf14 : SUCCESS in 9m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/74cd62f5ddd6407996f007f457d5bc17 : SUCCESS in 20m 36s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4d428e171c3d4060ae00c5838c9d2625 : SUCCESS in 6m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b46b7349120c48cc886911afd9384d86 : SUCCESS in 6m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/472a7afd0016437297a7ecef27cd93d4 : SUCCESS in 7m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bc0b84ab142342e28440b8ac58661ff8 : SUCCESS in 10m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f6b7c2ce49c14b78ad42d18826914b42 : SUCCESS in 6m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e2925210f16643908301bb511ad6db03 : SUCCESS in 6m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b82495564c864bccaedf28f13305daa5 : SUCCESS in 8m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/294c38bb8aad4f0c87777ce798cc045e : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/554694aec38640f288e9745214c86a73 : SUCCESS in 6m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/19575ecde0bd4e63b066269949b1b318 : SUCCESS in 9m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/15c416bb92a34e2a9322a0dcf808d156 : SUCCESS in 9m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4a3e5d6356a548eb92954daaf09205b3 : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9cebfea519c34028be3fd62cd0bdb1a7 : SUCCESS in 11m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8b894bca38dc4d8f8b9b1a1709ac08fd : SUCCESS in 5m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c5c979edfde246df98592d92d8f83e66 : SUCCESS in 10m 02s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/128e16e5490348c4a1591b91c4365129 : SUCCESS in 22m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c9817d02de674a29840e8815a0e5ef93 : SUCCESS in 15m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d5524e02b51e41549c7377723305bb0e : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea1e3511fe154efc889ae1aa06089b79 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/135078234e16478aa0afe821396c0e32 : SUCCESS in 23m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8d0a553b493f4e00aeb96f1b6386ab15 : SUCCESS in 4m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cec79a0d40a34f65afe462934f873442 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a41d5c7bd16e44028985ebdc8b38cad1 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f40ea77a330243abadd0e01c3498aa8a : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a4bd69d36514782805403dc94532b50 : SUCCESS in 51m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/796f50e96c1a4f1f993a9a1d08439990 : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8ae5dde27ce09898e3c3072e5674065858f7f7a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 00:33:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfc2494c1d8635a2443746503893dfa1373c6aff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 00:34:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cd81a592f3cf04c10d7e10b4f61925c85d89b99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:19:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3ed96d737784ed2b1b1361604640b54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ec275d4aa84a5781fe1489e2ef668a : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c41ae5822c41a29cbae2795160146a : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a593a1a79d246429d9312624f957114 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/037eb1d277e34d8b87e611175c61794f : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49e96dbe8abd4d7da999920014821c98 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aabe0423c4d448f1b4fdbd4a91e813d6 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/134e9bbb252e4481be6d6327607ba4d1 : SUCCESS in 8m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6ed99daa88ed4986a7aff830ba154626 : SUCCESS in 17m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/10663c0891124d0cb92ac5f0523cd911 : SUCCESS in 12m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/33febe274a194d8fb636ec3e0f87cac8 : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/803d636f2cac4da6b33423afa87122fd : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/469fbd0186804e9f847f70a98f6c2bf8 : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c043ade117434021b401cdd9fd4fe486 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/780c97fdeb614c45bfafa82087f68087 : SUCCESS in 10m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2595d7f9137f4b2594642178a44ddf3c : SUCCESS in 10m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3111744ab05441688a95da45d550cde6 : SUCCESS in 13m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/54e666bdf56c41928570f92a7611d309 : SUCCESS in 4m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/94ca19b8051e4e86ba405b8dd805c097 : SUCCESS in 10m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ee02312d8f034d71a12c1752419e4458 : SUCCESS in 5m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ae1fa7248b48449eb35a0c2160f3dd96 : SUCCESS in 25m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8fbd4060794743ca82589187d082bd9f : SUCCESS in 15m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0cf8957bd3db4ac8b0b8b48e60100834 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecd5069d8aa543e5aec5d1cf1bd8b7a9 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59f1bd41ca9d432085433de7265750f0 : SUCCESS in 21m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/06b6b32686aa4e0d9dffef61f402b166 : SUCCESS in 6m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/81bec1dd45f348409e0dbaeedca48a15 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0df8b53544194b3d876686ec9e3215c1 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bdc1a260dd994152b3e7887e04ae68d7 : SUCCESS in 16m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e62b22f479c243edb3b7d324b21ef5bb : FAILURE in 31m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5dc5f3d375a41b68f21d92536a693e5 : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9d24faa02ae768cdfbcf5c3e6b81c6ea6f6479c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:42:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cdc7ee2c69131c5ecaa8543c0880132cfe14eba6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:58:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac05784ade324f55bb6fae7677d86c30\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c8fb6462c2e849548f1dc48485d418b1 : FAILURE in 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"986b2e31f5db22603d6ad121e1e9cff8fb38c7c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:54:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e370fafd4434fc9a5c4ddbd27014d6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af5f5b4c0b3495f92c88ac59065e47a : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9141f58284d46f29b57964214fb6409 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7620a6c0a584cc88b7e8c71789363d3 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a28ce61512de4f93bc09c6f5d17acf61 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0bd5fc7659d14866a272385581cd021c : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff8e95f31491401cb64ab498210b612f : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b87d8ba941d04ce8bf2eb5059d47e973 : SUCCESS in 8m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b1aee32ffb0b4e6286aefa74a32bcb75 : SUCCESS in 11m 56s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2381a8e81dc247b3887e190c139f27df : SUCCESS in 8m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/839cb7bb88084d789db1bdcb4edb3099 : SUCCESS in 5m 05s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/057d4c2c7217437e8ead668084a0871c : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2a5eca3d4cc345488c5e58dfde344b53 : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dd8d021cf15a48e79d69fdd749003326 : SUCCESS in 3m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5c3af2c5c1a349e181af6c0c2bd57d9f : SUCCESS in 4m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/03f343b436794d22a8bba9f22c8c829a : SUCCESS in 9m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3f089cd70be540fa977603f60ef7e43d : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9389b7c72cd74990ab9f5d2dc38ef661 : SUCCESS in 8m 06s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ff0729997a68440eb757046cc99753ba : SUCCESS in 8m 57s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/65586fe508384979ad6384f3d446fb43 : SUCCESS in 9m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ad3748549d89423d876d0b9bbb93a2d0 : SUCCESS in 4m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ce79e8542ade49a3b80d2472c9808be7 : SUCCESS in 10m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/51c8ada0ed7d4b9ea9a3f5f2fa45c90d : SUCCESS in 5m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/254a1f48dace4f7eb8e5487ca2c423be : SUCCESS in 9m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/957176d40cff43c2af2f2e41c41663a0 : SUCCESS in 26m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9ca870de10ff4ea286d136aadb5e31f4 : SUCCESS in 13m 00s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cd5e667291f04bae9846632f4ba03f56 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/990d217039004bf4a24413f40557f6e5 : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24481c2ef8734d3785d513afc4f9e5af : SUCCESS in 22m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bf02caf9f84243fea1653432b7d86f82 : SUCCESS in 5m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2244c53330a54e87b7ca9ae75d45cc4f : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc6263b8489640e8afd5b687eebf94b0 : SUCCESS in 5m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b6d980fbacf34b199cb7184e091fda11 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52092030b48941dba4aa20c1f97d4955 : SUCCESS in 47m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9301a13e5934ac392e87a79e9458ba0 : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3d3b34b3b246540328d5b3646479f65b772d220f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 04:57:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0dddf1eee93198dea4c31f95ea86b54e113976e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:58:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"897dedb9894de35c01487fa7750c4cdb6fc9a2de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:41:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60917af75687429eba013fefbb7fc640\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd9b3269b9b4bd388fb52635bed71af : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df132b2da1a94f3bbf93bc39ac5047b0 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd67ba5f75e14554ba8bff26e489c9c7 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe4e31f780cc4d61a19f8ec81f6fe729 : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49b4ac14b4c746ceb56203368725cd72 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f6067b54b9041bd82d43346e001ff33 : SUCCESS in 4m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8324407f2b85458581d925e700443ad5 : SUCCESS in 8m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/378a51c2ed0e48d797da7f0d92c428f5 : SUCCESS in 12m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/15d5e62492c04a4dbeba89c9a3a809b1 : SUCCESS in 11m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2c80851167214c2492eb28dd7e008255 : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f01329dc8e8d42de9e3b15777c64d2f4 : SUCCESS in 5m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/35be8fcefc72410a82a84c519687c6b9 : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3bdb81bf001241e294a5bd1c62bc2c81 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6c974640ef441caa08e75128a1d240c : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f269e9f0507741d7a1b86d32e5723349 : SUCCESS in 8m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/57211f3cafac4d5db0141dbaa4a3597e : SUCCESS in 9m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ee7299c13d914f708caa00aa43637590 : SUCCESS in 4m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6c3b36685c3b4c17a56b3b887bb797b6 : SUCCESS in 9m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/018280884e014a22b091ef2230af6776 : SUCCESS in 4m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bc5058c279d642ce92a938413edc5f2d : SUCCESS in 21m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/faa9eece7f18477dbc5ea807f8409a57 : SUCCESS in 12m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/92c489dfef9f45108b83c63f1709e7dd : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f7c0f84fd1d443d9f7bcef2d18073e5 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03846d9e0d064463909d8c73bffb9ea1 : SUCCESS in 23m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a0764ddde1264ab4ad9a13d20e4ff8a1 : SUCCESS in 4m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/97296738710d467e96a1f01120882df8 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82776c1ab6104a6d95022df12cd25841 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6a4a037f32a44b20a02450edf850f5a1 : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3730a08251b4479184a23337a3f3a889 : SUCCESS in 45m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6e137fabcd949f1ab22ce041eb253e8 : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4d76347d4a1432a39e0ad896980c8d2c448331b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:41:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b7f9b0c7e8f11c86f701d272fc45d17c05aed61f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:41:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb2bc3aff014b9d80773c2445d8e2b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d6bd8cfd7d49709a788a25b049fe40 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2daad2f1326fbdb4dc008126c38f6a469860961","revisions":{"98c5aac3d1983a949f55daf50da13219c2a86674":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:31:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/873394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/873394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/873394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/873394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/873394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/873394/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:46.000000000","tz":0},"subject":"update constraint for oslo.service to new release 3.1.0","message":"update constraint for oslo.service to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id540e8e2c36da2fa89cd7edd1115ecdb3c752064\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I06a50dddde4ba28faa7f77ce20846a92cfb2096f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98c5aac3d1983a949f55daf50da13219c2a86674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98c5aac3d1983a949f55daf50da13219c2a86674"}]},"branch":"refs/heads/master"},"b2daad2f1326fbdb4dc008126c38f6a469860961":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 02:42:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/94/873394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/873394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/873394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/873394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/873394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/873394/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:46.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:42:42.000000000","tz":-360},"subject":"update constraint for oslo.service to new release 3.1.0","message":"update constraint for oslo.service to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id540e8e2c36da2fa89cd7edd1115ecdb3c752064\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I06a50dddde4ba28faa7f77ce20846a92cfb2096f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2daad2f1326fbdb4dc008126c38f6a469860961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2daad2f1326fbdb4dc008126c38f6a469860961"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
