)]}'
{"id":"openstack%2Frequirements~873395","triplet_id":"openstack%2Frequirements~master~If95a5f0514384707944a8c34c2eb205271dc7d1a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If95a5f0514384707944a8c34c2eb205271dc7d1a","subject":"update constraint for oslo.serialization to new release 5.1.0","status":"MERGED","created":"2023-02-10 16:31:48.000000000","updated":"2023-02-17 07:05:14.000000000","submitted":"2023-02-17 07:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873395-new-release","meta_rev_id":"24ecb93fb2bc3703dafadbd80ccb20beaf2f80f2","_number":873395,"virtual_id_number":873395,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 07:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-02-17 05:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-02-17 05:20:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 18:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:54:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-02-15 18:09:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-16 14:29:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c88c8353e43f03fe18f48244a2589e543be09b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-02-10 16:31:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54e5361921045d7f59250d609b9ed3ddebc0c332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:13:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d09f8b35ba484ce885d9497a3e652fd3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1891d074bd914243b7df46c02a752de7 : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"216cf3a001aa1731597b3c6c12f4433dc10474ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 19:44:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1810d78a3ea429c81d14dd48e8e5ebf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3bbcba23565409d919705cc0bc09260 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8648dacf1d34506a66874dfac392c8a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b36230903724cb5b34dd9ebbf54f3ac : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6dacf8f3eb543b18a8b5bc210b90408 : FAILURE in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86156adabfcc487d987fe5344b8ecfa2 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdcff64f5a854d04beca8cfb826f3900 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4c9ad5bb150c48b885b64f6b21cea33c : FAILURE in 7m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d1612e894f5a45a68be6e7a94a7dce8c : FAILURE in 11m 47s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/350de9388dcf48e892dc657b5cb2dedb : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/90ff24532c7f4c619b8902677baa826c : SUCCESS in 6m 04s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c7bc20c8fcac451cb4709f1d415ddaa5 : SUCCESS in 6m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5844cec283b248b6b7e082b93a329868 : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f4e27827713949ee933a1e940872ef8b : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/76972a7cc4fc4b0aa364486e9cd171c8 : SUCCESS in 7m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3bcec96cda3445b38028366318888752 : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/df090cc45a7d4f5aac44b7b7cbfe9fed : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73a1d4a3fcb545e0b75f57f37b72735b : SUCCESS in 7m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d23b1dffeba7456a95f82cc67d56a068 : SUCCESS in 9m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d9b8712728a34efe9dc887a0eaf40397 : SUCCESS in 11m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/177ac04a491142468fc57ed448fa39a6 : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2f16c4aee2ce4dd8b53f581a136a957d : SUCCESS in 9m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/24cf112e944b49c289a4252ddd1487c2 : SUCCESS in 4m 57s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/533ad98ae0e5463b9280533b688c1339 : SUCCESS in 9m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/32822b3193b4485ebd5c9b0a92f9d73c : SUCCESS in 28m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d012bae367b7452db18eaa156e7feef3 : SUCCESS in 12m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/25ea187a21864469aa14f44b8f3e6c20 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7bd316f2560d468d9e1ea095aa6292cf : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c087ec8aaa445968513e4bda15c1f27 : SUCCESS in 22m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/05bc26a6cbfa43dda6ef214760ebd610 : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/acfdbf000a6c4a45a0affa2242713004 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f68507dbf20452a8b63adc017606999 : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fb13572ac61b4503a431e9523a867b13 : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c53e3a04017c4d38bbe5a24516cd9d92 : SUCCESS in 56m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/005bd3124b724181b5e762073464dabc : SUCCESS in 1h 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b3ebf5e034426b0ad64eb424da65d7be74ea030e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-14 15:54:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c82707817b8b4521da0d4b9587fcc17060a19a6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:01:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa6df4c13eea42a7a77f2680a410ac7a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/52f785dec2e14e7fb82177f06ae0598f : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"178af47093921b9596856931e57324a758cec910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:02:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aec4f1d2a31646bc8a04278e9c2cd584\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da5a5044d974a0189042ff466c40707 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f8134fe8314f29a4492fbb53f2d474 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f63763939454fa1876d96638faaa529 : FAILURE in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01b3a563da2b4e9fa572791c923814f5 : FAILURE in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f66bf86ed20e4b20b5d6facffd471796 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/439031d6044545b686ee3f09e7ea4148 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/145c3ecd2f284f20b46a4b36349c7f49 : FAILURE in 8m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2eda22b682a4a8bac06cafe990fb279 : FAILURE in 12m 29s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a6480fe82d3c468d8d30d582e3e99dd6 : SUCCESS in 7m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fa670771b2534f328402d5dd0a41a3fc : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6e8e38be63d8468d9d17108d569ddd38 : SUCCESS in 5m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ab4997d0ddbb4df68b48e94c7766d18e : SUCCESS in 10m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e0fc60cf7cf44a429b3374b7ab3214dc : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c8cee657845946dba2e191226da9c6ba : SUCCESS in 5m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dc624c1f1abe44d3bdf5226fad956a81 : SUCCESS in 8m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/cc168daec45143508bbe250ed62e4c11 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97aa8aef91fe45d4b5aa480fc9388d99 : SUCCESS in 15m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cac201b01fc146dfb6d83cdf972099b9 : SUCCESS in 9m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0ba6ef9110fc4911a1c02ef8889c3945 : SUCCESS in 15m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fe2bdbb772a842d3b2d144aef31c7fe8 : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5d1340631dfe4d099da46c5e634f22c1 : SUCCESS in 21m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7497a8cc7d994849a5c2e5007cdfb23b : SUCCESS in 5m 05s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e90fada264f34f08a8454c2056754d1f : SUCCESS in 9m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/33dcf342518a480ab92e168b7c691547 : SUCCESS in 28m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d92e5d40df544485ae1f2079930fbf79 : SUCCESS in 23m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/10b9e830a1054f249a829650833a2d9a : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/432486bec5694273acbc876c4b6a97a7 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef319bdc64cf4ef684e7a112b7e35938 : SUCCESS in 23m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8c0d11a19c6d4227bd6b4d884d8594b5 : SUCCESS in 5m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6e6fc8c3eab14278b1352ed59b839e5b : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/740c09a3bc494f62a3343b24e2a571d2 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/648bae4a25f84938affabcd7dddc218d : SUCCESS in 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a5aee65d08a4000a8c47ea035d6acfe : SUCCESS in 56m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef0bcdd641ba4c9d918fe8d21277adbf : SUCCESS in 1h 46m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c58f09fde6ee4708b649311315781ee8be14494d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-15 05:50:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e13fe4288bceaa99512e1f510ba93ca18be535b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:57:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f89a9d63a8de41d6bdd682e2e92ab836\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d1d686cd91c42e498d4e0ddb0bc55fe : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cc4fdafc3ad2c06c4e51831595593fb9e1c6a351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:49:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f18f4ee71a4242daabe3dbd9bf26ab8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9653baf8354490b55b63d45bf8affe : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87fe877b2b24404be5efe272fdcf92b : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be1e373ebd08445594a91ee06368eb2c : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ef86abf8e8b49bd9944a5d7f827086e : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04ad65ea9cef4b758f5b554492922b7e : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed6aa14bb48849a693eef93feb44f79e : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d3500174af24c818984adfb91a31799 : FAILURE in 9m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb1b137d90c447fcbb44a79fe550e5dc : FAILURE in 10m 30s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7df5f5d2f3be4f908024c439206f6cb1 : SUCCESS in 7m 14s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/02ec2de403a14d80b318374c21b203da : SUCCESS in 6m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ab100f9b3a66466f8b456b356ef39631 : SUCCESS in 5m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/754de777a434430c92da8d56ba27a5ca : SUCCESS in 9m 53s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dfe97d6233f64e9785e8196c367f8ef6 : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/047a7672814d418eadb1bb33bef58fd8 : SUCCESS in 6m 35s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a0a4420515404f00becca21f6f3e3676 : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/09d39305081a4d61b54315b3a3f5a957 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f57ca9db5e54156b3ce364260fc17d4 : SUCCESS in 7m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/50366eb6caa145a59f8c7cceb3e8953a : SUCCESS in 9m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/712b402faaec425f8df2af15a25c8c37 : SUCCESS in 11m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e872a1a7f1354e3294f0ad79e89da709 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7231f2edc1e04f56ac0596598826c871 : SUCCESS in 10m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cd5f04137da042dfb4aeebd67495927a : SUCCESS in 4m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/faefb86a236141a380f616afb0eeefe5 : SUCCESS in 9m 44s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/68d0921a9bb6480e8f1c6047dee5c80c : SUCCESS in 26m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0b2d8eb4864146399a0ac97b61ba900e : SUCCESS in 17m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/70a6a489fcc94beabda66c6839630250 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/daca9e2548e04710b72a64b5fb51c637 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f456468b6e4a4f3ea946d57c6c657918 : SUCCESS in 24m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/22516d82e4d744479307de828be5a5eb : SUCCESS in 6m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/97cec30275634ebcb318f26a12d84ea4 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a3eb903d5b848c788c488cfbdf40416 : SUCCESS in 9m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/448ce976614a46ad84bd76f6abc7b6ac : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a33eb70204de42f29fbf6eaf1e0149f9 : SUCCESS in 46m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca0505e9036440fea16641232b58f9e5 : SUCCESS in 1h 42m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f035ae4ac55d2c6247148d6d3465bd7549a62d42","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 18:09:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bac8e962cdc3bf49a09ca6cf1b61be5bdfb8e8a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:17:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87a1c3e2e6cb43f8a87980202f047c83\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2d3a71412f0e428288d7a7c5378b9ef2 : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5fd19c1b3418a5df7a1d5379fddc1cd16fb08c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:46:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9782667c46c45a5a10ed15b2d6bddf3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da3758168134b4eaf2ea6921c5b3ee3 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46762e208d89405a9899109ba21f8485 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf82db2fb48e4a79a23b910e6f666551 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e16e2ed6708d40638fc4537c739f7217 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e87a7aad1aa04d7bbc1271eb4b9c85e4 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91f58db4a09f4ad3bf4436f87adccc4c : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/43c163cef09840d4984f8b552dfa4b94 : SUCCESS in 9m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/48ae2fea0c584768806d4385118acc04 : SUCCESS in 14m 52s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/06b50441ebb34461bbc66afec3794554 : SUCCESS in 7m 00s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6513609333be4f78ad9bcd9f86532077 : SUCCESS in 5m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e51e1244f43244a69a8e83c19e870db4 : SUCCESS in 6m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a7915efe90b04bc184500bc384a774e5 : SUCCESS in 10m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f9a1930825bc44d2bbbb86d82987bd66 : SUCCESS in 4m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fbcb423d1a194b639d43ee449cc45e03 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/81f44eb710f040298846d7f43175985c : SUCCESS in 7m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/482d7e6bb0424217a02b1ae41702192a : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b3f8e47bbb3499582cdba261edae0ac : SUCCESS in 11m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dda8e07dcab9418f9aeec4de3226adbb : SUCCESS in 8m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/abde15d5186d4f9ba053288d6f8f8749 : SUCCESS in 10m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4df6108253984818ae37cbd44f335e75 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a5ef8bbec5724ddea34af80a0348b961 : SUCCESS in 12m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/587a34beb2ff4e649ffbe9c60235b3ff : SUCCESS in 4m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8755b4f326e84cfdb3a4b9840e0470a8 : SUCCESS in 10m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/97e6706fa1df46708cff6407e7f68551 : SUCCESS in 26m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8d19338e90eb49979bda477ba20e0f67 : SUCCESS in 13m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3a7f2b38a9284eb3809aec093509cf1b : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56d81a26f1b04bb197646016d32940e9 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf5de1e9a3f74cf1bbf71753bb93435d : SUCCESS in 21m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/73e40896fc3c49feb5202c8481726bb7 : SUCCESS in 6m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e00971bbfb3b4ff1ba67d4ffa5212ea4 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b40ce25e157e4a46bc46d5f5690d6b61 : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a3970a4f5cdd4d3684b42d1010f64ff9 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47e5ecaee80a401285ce0fd11477b5d7 : SUCCESS in 1h 00m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/616d7b93a1cb4a26b406cad1274f6619 : FAILURE in 1h 20m 15s","accounts_in_message":[],"_revision_number":1},{"id":"80a16c44a1fcbe1bd5046ff136820096d702f53e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 02:20:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5e242997bda7ae50d1ed16d37fa29b35b288016","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 02:30:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac2450679d142c38ce150cee66d9c14\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/78215046aa3342ec946d6151512f5d2c : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"617b5d49727607138fb625da62a7d0e509fb1e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 05:33:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9eabd031ba2458e8ed392b7fef7eeac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e0722b169e4b7f89ffd7f7cd395af0 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed49be0f5804f39a479f28aff098b0d : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a11f393ecf94a819077e0a7512ece97 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a9755f041904a5d9a5a8e172425a62b : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/087eb3edabe14e46badba41ed398b543 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/987ce2928fea491084bea6ac8c874bad : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b37fde1b6b3442bb19e420bb6061d3a : SUCCESS in 7m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/08ae773f6eda42c8937af456e7d30ccf : SUCCESS in 12m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8a6ab040082c46fea283e9b59d910013 : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/23851d24bc0a42f19bae3bf63c17ba80 : SUCCESS in 6m 41s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/913dc5ccf2804f3e92ecc64554465888 : SUCCESS in 5m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/613d6fb8dff64aca8f8c5ba2c259e964 : SUCCESS in 12m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bdcaff06da39489fa53ccf318af5da15 : SUCCESS in 4m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6636ba4a530e437ca5e3c79e34467c0f : SUCCESS in 5m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/903c873a28c04565b9a31fcc9df93473 : SUCCESS in 7m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/22971b26e87c497eb12a589320dd2888 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3a09059c719425295f0484ad6c97d5f : SUCCESS in 8m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0945f68205284544854d634e9e3ba006 : SUCCESS in 12m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5d87afa15c714abd9261c53822ad1cf7 : SUCCESS in 11m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/aaef76b04c0f4ad79a3988adb7c25baa : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d6b6764a195040e9bb81a4d1583cb032 : SUCCESS in 12m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/44a026bdaef943cb9996db0a6afb9873 : SUCCESS in 4m 15s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9f79b146d35641da9651daa48b1f75cd : SUCCESS in 10m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/684ccff41a074883943111d9fe997c4e : SUCCESS in 24m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/173424c24f0d44cb88cc71abbbbe4e1e : SUCCESS in 15m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/49d7e86206814d458cd59ab90bfe0896 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b24b2b3dc678479bbee3f5a7d02cce00 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66ac60f3a6164e69a2f0c75319c20bca : SUCCESS in 22m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3fcb2b49a23d43b3ab757b2bd81a0be4 : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/97dfa431f76d48cba3d7864e86744afe : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da3d8143f22040f281db89ca9169c615 : SUCCESS in 6m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/95ecf4c4ef2744d3afae7e031cf80f47 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9658989f67a745479e26e3fa4abd7bb3 : SUCCESS in 1h 04m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7acc38a61f094920bb3b16e2a47cb195 : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"48654e4999f1f0ec6d4484043dbae4602defd6ba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-16 14:29:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"429f9608df204a84654603d83717409919e868b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:30:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa480dad4942bb5de50b6b24a23bcccc55b79cba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:43:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca0e55b755a2419eb3c911f410ff9626\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4999a90c69854f59ac1bd078a0a83139 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f9a82311bc405990f72f7de6c37cdc : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de8cdbb749e144b5b2a6c772574db92f : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6dbd0cf3778940458754f93a6ee8fd0d : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5647b65b2cb46d59dc09e816bf5f921 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ead5b5d9d9b54fca8c3c64a37b1aeefb : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/948d36247a594ba4af39b7db02483a1b : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ea0514307a004ffaa481502d4aa8ee21 : SUCCESS in 11m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4856003aea6a493ab6b414ac3fb0eb1d : SUCCESS in 13m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/282a92ddce644819ad44f9a2940e4fa3 : SUCCESS in 4m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d63a0d2f46ce45a38121c6e0b3207107 : SUCCESS in 8m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/576e2db853aa4302ac09e6fc4bb794a0 : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/efd8cdc7df5a44e4a67fdddeb8e1ad5d : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bea63c519d164123977e30173761ceb8 : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/683ac9c419fb4ad1b3f0ab65e38d9039 : SUCCESS in 9m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/74fa0059b5bb43c8a2de334d5c043c55 : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/48320895aa874e12bf4d8e7be1af95fa : SUCCESS in 4m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2c734b578f9c4b3b81aabcc43264b006 : SUCCESS in 9m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7ce07ffedbcd44efae018617daacc0ef : SUCCESS in 5m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8da849898947422cad450a4e786f820b : SUCCESS in 23m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/384367ed32c74cd2b3ec8e91b0487797 : SUCCESS in 14m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/36b4fb96bdff4eaeb3d6bf15716236b6 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d39e01dfa234c71b3465b8b64ca7652 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5bcf6ffd0394b2a8ca7b529edeeb06b : SUCCESS in 31m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4228d230a0de4f69a872d8ca28280e66 : SUCCESS in 6m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5ea5f05cec51435eb6ee8d1a76195ad4 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/678144821ed24d15bb00558d1a541cf6 : SUCCESS in 7m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/15eaadf60f0b4a8a82509fb7db87111b : SUCCESS in 17m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e47143abcb048118104f2169704c189 : SUCCESS in 1h 07m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0620c2b6a3034bb5a726c1346baf2a50 : TIMED_OUT in 2h 06m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0f8428e6949d4ae77ab9f8d6b0b2aa34d62b3d07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 02:42:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cb7460abd99d29b4b792c5cee78f508ec3d3d7a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:50:59.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac96782c780349c591ac5609a5e90f6a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dc33f0f7f2ff47578e764989705277b9 : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5f8d4c61056c722b73339a7ccb0fa4549c8db415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 05:13:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5db87235bd84b8ba2d06b4c012dbdc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adae407512814d79802409815124dbc3 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d915a0d5f30b4cbd9147c37470d20ee2 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a79c47ded8a8474bbc54f10dd8814305 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88f86efbbe314273b2e154d58ec1f620 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4158b7374914480ac0840f09c4aee4a : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41f836538fa34590a4932dfe14ed5d2b : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/712f2ff2c3ab4513810ddc776a243023 : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b84460ac6c384266846312bdb693a36c : SUCCESS in 11m 14s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/007009e5f6cd4cce85fcac6ca70526b5 : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/31a5406b242f4f61aa36974db02817ba : SUCCESS in 6m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/18896cbdfd594361a350efe188db1efb : SUCCESS in 4m 56s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/18a60a3b59ee49aba2d5bf9cde1c29ad : SUCCESS in 10m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/caa027b7118741c682102b9c58f72f60 : SUCCESS in 4m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a4de86b482ea45fea954ad9095d5b04c : SUCCESS in 5m 23s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bcabd26276624b3f9db0b373747d4e7e : SUCCESS in 7m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a27cf5209a8d434aa99325e29174bda1 : SUCCESS in 16m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cb55782a727473d85c6e6fff60f11fc : SUCCESS in 8m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0f43b9275c664cf3938e59b1290ce2ed : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d5c6c449965f44d0a2c4c42192f50c92 : SUCCESS in 11m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/58240e9792324cb8b95b89152097a378 : SUCCESS in 4m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8315f77118604f3b9e622d812a98a545 : SUCCESS in 10m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/82e64bb268af49f780c0c6c65396daf6 : SUCCESS in 4m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7b4df11b58e54e5cba900f6988897877 : SUCCESS in 10m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/98931852640e4d879fae822c1145b8ae : SUCCESS in 26m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7b4320b5866e40538091d4364b034154 : SUCCESS in 14m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fc2d2e3ff7f94ab6bbd3811b8d87de16 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c3be33a79efd4186957e35f43c1e1373 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09161cb7b66a4d9d8b7f6ad3ad1f6a85 : SUCCESS in 21m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/181ec92249114849b5560d9a0e39033a : SUCCESS in 4m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e6a90891203f42d58e8199d1c1f7c4d3 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d81ce16a99df49129151013709f48e68 : SUCCESS in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f4781290713e4c018670c0c7a812cd93 : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3a07d68132140699faed30cf26fa624 : SUCCESS in 1h 03m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ce3135155d64bb2b3553f0de2ba9287 : SUCCESS in 1h 38m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e463b7ba1b7a70623c8320804dce6021727a2be1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-02-17 05:20:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c7b0345d764a21142e43c12ec316ebf0b092fc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 05:20:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b890cee906681f307ba9b1454da5a39df54c7204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:04:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f52eeea2e704068917186f5d45745c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bf4d4855424d7ab06cb0849ad6e2fc : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/298340f9438d4b729178e20da94b32d8 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb887839e9594e51aa6a3af61ab85c87 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51e4f39744cf48cc963e05c576b475da : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a2b7d0f82694b1d836c6b17331d2294 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e83b29d588ee4b4aa0f144b879ab31d7 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aca066a00ed7452e8e68d48093529d45 : SUCCESS in 11m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/997916a3c596425e8ea4e03fb4d77d23 : SUCCESS in 16m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f4e2795af86c47288765a6b127c41fa4 : SUCCESS in 18m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/eeb2363f115340db81509d00bfb5cfc5 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/45344f750e8d4080ae96f878ee91fd62 : SUCCESS in 6m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0c372284fbe749a59625461e140f00fc : SUCCESS in 10m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/57ecca15067147f99c61b4b53d9e9d59 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcc97dc3e24046c5ae18efe1b7b22f3d : SUCCESS in 8m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ad86fd61c3424a7e928c3923be904595 : SUCCESS in 8m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c4043a1b89514e478e33a8b869705ef8 : SUCCESS in 18m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a9f53d0bb0484f8c956a11c69f7b20a1 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/58492f19cd064274a55873e54027d3a9 : SUCCESS in 8m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/43bd40d0ad9d4ac7aca6918d04276a55 : SUCCESS in 5m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/75079e0f57614d46a518a709baee43f0 : SUCCESS in 26m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3c9718b5d10f469b9757a3b669e02ccf : SUCCESS in 17m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d3fe449f447a4e89bc8af411cfbbd2f7 : SUCCESS in 7m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2c240f8946547eba991fab0bb59d81a : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e2f342985044850a68e7ff4a59f612c : SUCCESS in 21m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e7475f05a8804956b37cd9cd28454672 : SUCCESS in 8m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6dea2f74d0cd41bb9be57f6a198b8858 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91121d193a9f4890a63aae2de849623e : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e63fff9784a94d738d917831280c34d7 : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c39ba9a8ab434b3eb0b5709d7e23f412 : SUCCESS in 1h 04m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebed82757242422e8db58db50513b0b3 : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e468ab4dca5ce1bcef0c6405210852bdc3b98974","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:04:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24ecb93fb2bc3703dafadbd80ccb20beaf2f80f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:05:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6386fe9571be47398fc0d8f986ca30cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e1a90391c84b8b802a130d317d8f40 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c9f9d80f5a55766c600be1ba393e82226420a3e5","revisions":{"2062552bc8a078d88ed96314c7d523556a1b330b":{"kind":"REWORK","_number":1,"created":"2023-02-10 16:31:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/873395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/873395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/873395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/873395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/873395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/873395/1"}}},"commit":{"parents":[{"commit":"0b5f5d0273975dfa808e7f921dff14ed2887b9f6","subject":"update constraint for oslo.messaging to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b5f5d0273975dfa808e7f921dff14ed2887b9f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:48.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 5.1.0","message":"update constraint for oslo.serialization to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I63ddd008c182b3a5e65be4556eb0b86c973c6a02\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: If95a5f0514384707944a8c34c2eb205271dc7d1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2062552bc8a078d88ed96314c7d523556a1b330b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2062552bc8a078d88ed96314c7d523556a1b330b"}]},"branch":"refs/heads/master"},"c9f9d80f5a55766c600be1ba393e82226420a3e5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 02:42:59.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/95/873395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/873395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/873395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/873395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/873395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/873395/2"}}},"commit":{"parents":[{"commit":"2de915b29f8bf4e124025f910f37b4aa29023549","subject":"Merge \"update constraint for cliff to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de915b29f8bf4e124025f910f37b4aa29023549"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-02-10 16:31:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2023-02-17 02:42:55.000000000","tz":-360},"subject":"update constraint for oslo.serialization to new release 5.1.0","message":"update constraint for oslo.serialization to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I63ddd008c182b3a5e65be4556eb0b86c973c6a02\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: If95a5f0514384707944a8c34c2eb205271dc7d1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9f9d80f5a55766c600be1ba393e82226420a3e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9f9d80f5a55766c600be1ba393e82226420a3e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
